)]}'
{"id":"openstack%2Fkolla~928311","triplet_id":"openstack%2Fkolla~master~I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af","project":"openstack/kolla","branch":"master","topic":"bug/2079775","attention_set":{},"removed_from_attention_set":{"6994":{"account":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"last_update":"2025-04-01 01:31:51.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-31 23:50:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-13 07:30:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af","subject":"Add libnetfilter-log to l3 agent container","status":"MERGED","created":"2024-09-06 02:56:32.000000000","updated":"2025-04-01 01:33:51.000000000","submitted":"2025-04-01 01:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928311-bug/2079775","meta_rev_id":"51a0a968e34e810cce9311a378a71f4c6b45084e","_number":928311,"virtual_id_number":928311,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 01:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-31 23:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-20 06:31:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-03-31 23:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-03-31 23:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 04:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 07:49:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-06 11:22:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"de0b47d7a9c2a1c27a428fccf1955d8e868a09df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-06 02:56:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6d21c9ecf914026551ee07daad58ed5428ede73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 04:34:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbbc188175d743c1a69c31d0c3e18e54\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f40c2310a0e546078c2a1977607f067a : FAILURE in 28m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f6acd4036d8b40419f76bff65249ea07 : FAILURE in 19m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/835ed7ee2d6c462e93e28dc169b5c35d : SUCCESS in 18m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b0be645ca2754db58136197a53c9db81 : SUCCESS in 24m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/98f684d9552b4b3b932f946ab19a31d3 : SUCCESS in 1h 03m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c3228e7aa1c5461ea7d19e218c34e38d : SUCCESS in 1h 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513172aed02547a6bde1f1c65e5519f5 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d958d779f4b54cae86efd44f177c6e64 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c1808b74ca048cd80f42949567c16c2 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46ebd9e9ebdf4e69a23c0388c45876e3 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7bf00aa470348d397f5b3c2ffa3c4d8 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da94b0721abb40aa8ce706bdc478cd42 : SUCCESS in 5m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ad6281d33ba14677bdc578e47cd300cb : FAILURE in 31m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/651e8703bfef4f2c8cdec51efc10dff0 : FAILURE in 27m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5600a3cc59642fdb8b3ab3421a7868d : FAILURE in 22m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe5fef0b99f441cfb2a4a64ee008f2a2 : FAILURE in 22m 16s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/130d8f6aea374ee7b13be07c6f687139 : FAILURE in 26m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c86e41a618a24033b8207a2963f5d8f4 : FAILURE in 23m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/980f5a85df97401fa44e66942ecfa8d6 : SUCCESS in 18m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2b0938010c8b4f7dab38d2e8a41435b7 : SUCCESS in 57m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6be831ba566740a79ba7dee7443c1653 : SUCCESS in 44m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6d2cd0d531814249ade54d98fb0e53cc : SUCCESS in 1h 03m 40s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4a5cb454df4f41a5b376d0c4ae7a77da : SUCCESS in 50m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b1407da6bbfd4d43a383ca38856c1092 : SUCCESS in 1h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c8bbfcb002a545ea0cb53e88589763d27638a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 05:41:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d5d5127348426d9fb7f21e3aef8f85\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/44cdd439440841a9abeb4be09d8a2b91 : SUCCESS in 2h 42m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9c89811d77cb41118c6b08a3c79cdeb4 : SUCCESS in 1h 51m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fda5048235743bfb499934f5a573b30 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f2935631bfa44b586ee346c499f9482 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d352ca84441570571e61c147fcf2f71d122defd2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-06 07:49:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7104c1eda92782df7343f91f1b38d7c456bdb94e","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-08 01:43:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3c5cf29b8c07cdb1671f045309dce4b3f5bb310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-09 04:19:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c16f55d392b9fd935f3ad6ebd8f46d7d4ece559b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 04:27:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa0e1390e136475847c6a46c206a31feb4c3d164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 05:47:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17301cc1b5cb4b188a02bf76fd70dd91\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4d63ebfd1b544bed8585b6c4fda63250 : SUCCESS in 23m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f7a2fcfd39f64fa6ad04eaca42590640 : SUCCESS in 52m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25a36aaabd854a90ab718db704c19a0f : SUCCESS in 15m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/45bb51a16bba4de3988d4ce8feeada54 : SUCCESS in 23m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/caad84f4d81b4362b7e3b6118fe8b7c8 : SUCCESS in 48m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0b92bbdc6889465184e4318dc0375088 : SUCCESS in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a33240d2dbfa4ffb86b2a68f2de0bfb5 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29dba4978566499e92ee028c18849818 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bd5c6700dbf47c7bbf6318e9c2bd57b : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae5419ff03b242f9885ff65c470a5c87 : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23a43847a61a4308a98bd39f76370795 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c9e9d44fa749298e58bca3bbcf9fe6 : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/026b3e62d3a04331b44b21630ce36995 : SUCCESS in 27m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2c85eba0505541b1a74032c4a5198049 : SUCCESS in 38m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ca5850e61a884071931a5c666c57db3f : SUCCESS in 57m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/722a3f7363134e1381ef3233c8d51794 : SUCCESS in 1h 13m 35s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/840ac07bd7c74e82abf9512c2913f14c : SUCCESS in 1h 03m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/42dd6a8926e446e3b41d7b17a099ddab : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/47d762833f6043ba84ed8c9278f21dee : SUCCESS in 20m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b23031471c544564bf54ae54c537e607 : SUCCESS in 45m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e594c03d9bd4fdcb03a6e6b4fdd3ebd : SUCCESS in 42m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/90295cf6f8f54aa494f1c84ef25f28d1 : SUCCESS in 59m 54s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8c5771aee2dc440c91c64c5d2f5c0c0f : SUCCESS in 46m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/508a130322904204b092247046426134 : SUCCESS in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bce66dce61ceea1e615b1f5848db8637b477df05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:01:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c075bec0e1c24eddb074c5ec2df95030\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d31c13aca0f54131b1bc9dbc47249cff : SUCCESS in 2h 40m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bee0d44253784309be9e929ce9a6afa1 : SUCCESS in 1h 26m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca420551c238446e9d012fb27e3ad347 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64ffb44b7a6f448580813ad366be5bc8 : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64f255a1a2b4e4675659b49bd029ec4d4b6c8db2","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-11 04:53:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2cf692922a8c766f04533d95f9bedf3ab884c93a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-11 05:02:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a37acc0d5e6e93f8e79d96319d71cd2ac7b93ec3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-11 05:04:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"616cee5b8ff50c741dbe58ff0b8099e0d6a08aaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-12 04:44:53.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7e7891731e12811cd8a7d61567e4d468d779ff98","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-12 06:00:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d22501f6fa94e85f271f6c1a9417b8592f67a659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 06:33:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18f74da9febf44a7b64dd040e3c5e6ab\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4b1de11d701746a7b5055c8bb5c27394 : SUCCESS in 18m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b8b7631ecbfd4407acf1d0dd2a9d16c0 : SUCCESS in 53m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/69cf074c270048fd8772d337cd525863 : SUCCESS in 12m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4c4b26ac590743c88c8c779ae2d445bc : SUCCESS in 28m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22227e4ba3874074a68117285fa1014b : SUCCESS in 1h 00m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/59468a6eec0140ad903c996a9fc0c27d : SUCCESS in 1h 23m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7afda2f8a994cb2aa297ddba91f48f5 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aea155dc1fe44ceb16f3a0079bc18b9 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f17ca8b219d49c9b766fcae821f77fc : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/881b6bb9149c44b785169d7b0fb19908 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eef8e233e3d4508a5d65761083b5585 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9703c7a215a47679a77c87cb8ecee0f : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64da3df436a4488fa15853018538f018 : SUCCESS in 24m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9d9fbf7dedde4209be696badfe71baef : SUCCESS in 33m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/93b3805c2c684233a3ae4f8c17b5240a : SUCCESS in 1h 00m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3898fad705d7488fbcbb57f905e7c406 : SUCCESS in 1h 37m 12s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bb8c61eabb0c4de6ba04c358398d9707 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a4907fa66ae84fdda201453761013651 : FAILURE in 56m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7a51e5bcf6384372b66ab50b83b45200 : SUCCESS in 16m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36c2b4a694184be086b22cf92eee9a00 : SUCCESS in 59m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cc773e423a44013a25fb405209d3a8e : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/74fa9bbbcecc4a67907bc3b061f9e559 : FAILURE in 23m 32s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/620f4663a70747d089df6aa88e2bd9aa : SUCCESS in 45m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/76f2fbae888740929ee15b6d3270e373 : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d19d3d23b8f5cb56c60b5b9eb8d32f38e04b18ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 07:50:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdcbd8057ec4297a83da72122fe19cb\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8cd21a6344fe4710b7d599b4534874ae : TIMED_OUT in 3h 04m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f05bbfe518b488a9bcf853910b0ff98 : FAILURE in 1h 41m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5c8c5b68b24493184854ffdb9c3a950 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b8d2815f5674e28b391b6b4d0e9a4b1 : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ee8d8d8e68d925f37739e425254aba3c216ab68","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-12 22:33:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f61f552af26f2d9f0928737a6e9dbe7ea2b423d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 00:06:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc928a089bd044f19e4bb204f723aa31\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fb3d18edc4db425a92b10c5f0a2c6ab7 : SUCCESS in 25m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8fb57eec65cd44ca90510296cf11900b : SUCCESS in 59m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5ca9cd044fa445dfaa3a3fc9ae48fe13 : SUCCESS in 21m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5f6ff1723e6347c598630a2a6d787ecd : SUCCESS in 26m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/109b5683897d4aa695fc607b81def59b : SUCCESS in 59m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c75f09020b974b90bbb94e9e0bcef440 : SUCCESS in 1h 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723c3f9bf24c463d8f34f4b5fe27a394 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af1270d6e48440159a22b220dd008532 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/937b69a6c9e54ff69edec239d0aa928d : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71cf9b0c858242768c740ec0b45eea6f : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/829cb284b26e4879985d91ffa704abf5 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0df903c07da4134b46fc396308e0ebc : SUCCESS in 4m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8e33d40abce46eeae35ef033146debe : SUCCESS in 26m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df465dd01d934dd9bdcc962e8e284ce6 : SUCCESS in 38m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b0e7b05ba134c4e82e0fa1c266be94e : SUCCESS in 1h 04m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2a10d044a8cb495eab789983d77dc4d8 : SUCCESS in 1h 26m 02s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0538478470c044d0ba30c14f516e35b0 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c263309807e44ac4bc6d48275a737130 : SUCCESS in 1h 24m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99b1562f10df4e85a8fd29b968175c6e : SUCCESS in 24m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6d23ee4869f349f2ae8cc703213ab695 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26b2dcc170104d8ba1aaad6756673356 : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8268e5d4edf046488833eb4cbb22a8ce : SUCCESS in 1h 09m 58s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/19445df652bb444ea801ad71ec47822d : SUCCESS in 53m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3775294bda864d0f9060229d98123740 : SUCCESS in 1h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43f1e7d8597fd3c31cea2ed128fcc9741c649379","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 00:56:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3af1ca18cd45f7b0b91f69a3280ef7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ddb2b5f3b7d2418589d3a90c7cd5d769 : SUCCESS in 2h 21m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8167049c78d144b1ac899ffa17965568 : SUCCESS in 1h 51m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1aad0c7cb40c495789076c30270d025c : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dee2c9707dd94a0fad348e4de79ef68a : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3242bf14e22bc43a7b191c2afc8ca9255f18e4ea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 07:30:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd6c36f5b7fc2ba6be72b329e3feb0742a0302ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-16 04:33:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a0c3b1f00864269281a6a038d8d751e6296bb0ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 06:24:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1c31e596458429f946627d0d12a6fb5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3f86fcbee42e4d488e536a67caa3d5f3 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2cddbf0bed4b4469b1674cf9d9c5f06b : FAILURE in 7m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b257df8a96d94066a2b484a63b49a9ca : SUCCESS in 22m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/20f712bb769d4d868d92440e6ad5f132 : SUCCESS in 25m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fc2faccbfc4748608dd02414d658a746 : SUCCESS in 46m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/54a6b176a7894d99874de8d55f3fe517 : SUCCESS in 1h 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f2220f1c434a22b79cdba8f719d15c : FAILURE in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8453a8df83c745418a72bc090c780d2e : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1763ae5029cb40c6ab2627153b3985cd : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5ecc069f2e743c38daf940cbcc9ed45 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40ae969816ab4e26ad248f7839c9d03a : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927344eeca7f42c1ab546e14fb1cf8a4 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cfbe9e3afd44289bf96670d9a8eeac4 : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9fa2831d8ce04386b4ee1d78d58c02df : SUCCESS in 25m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0008021294d24cee92aa2a457d20f3b2 : SUCCESS in 32m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b60cbb597984119af96007cf90ffd82 : FAILURE in 7m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4ea93df914864f0381a3c0c8998a2a10 : SUCCESS in 1h 29m 55s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/57110dd4f10743b49ea3fdfd24cf0a3a : FAILURE in 7m 33s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d3ceb159b7734ad8a00da54b4e6ec7db : SUCCESS in 1h 15m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/43803c207dab4e0087544748221c61ef : SUCCESS in 17m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a6b75229039745d0a0a32ed864f88649 : SUCCESS in 40m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a4911c0eedcd42eaa4c1dd87e2e623c8 : FAILURE in 4m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7066ab22639549e6977a3fb25f920a60 : FAILURE in 53m 13s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a6773b9b6dbc4af38cab6fa6da4f99c3 : FAILURE in 4m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/976cf25d7d78489ab49a3999561f4d75 : FAILURE in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85b2afb9d40e46f91c717cd817ec03aaaf086605","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 07:38:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4eea1acedff44a9b9f359f81347a3fb\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32907b3f9e2e46cf94d7c9bfcf0b8645 : TIMED_OUT in 3h 02m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03ac523fad4349a0a5a735f1abc92dc3 : SUCCESS in 1h 39m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/493d79532c9b438a804ca4a0a4aefefe : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/274b2c7c82c644a3af7e636d7180f6f3 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"32b167f8e740d4404024be260b7193b7f1e438ea","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-19 02:55:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87e8a01fab47de80f96d1d0b4b83c9b9fec6d6da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-19 02:58:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"65d0460bb2e08c4269072c8c28f358783e66fe5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 04:34:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dcefcc4c22d4acaa7dca8740271b183\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/24b6f062875344c9b45beeeb85342104 : SUCCESS in 27m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/efbc2d54a4764b35ac33edfff08ada04 : SUCCESS in 53m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6939e9ee8f994f8ea4f9cce13f537384 : SUCCESS in 18m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4e38e34e27ae47bb86aabddbd2fb5b9b : SUCCESS in 24m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2a0cf27768b345628676eecd091ecfd3 : SUCCESS in 53m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dbe73e97e73e4ab1a17b712ca8613f40 : SUCCESS in 1h 23m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11096a4cd75b4c1c84170ab1d5ed1873 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46d69d04bc234c11994abb97ad92e84e : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/094fe3707a3d4dfca2978015d24d651c : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d55d63f4492048d2871a80f718021e9e : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cee5509c2bd435a830959f4e3b212ec : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb6eb36ec474a91bd07a24a2bc936d5 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/257c8f57bc7c4901a8d4a4db1c375894 : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/960e4b1cbbfe42f7aebe860b949ac475 : SUCCESS in 30m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/567e6b46d0284a478c893458e7a7ef47 : SUCCESS in 33m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4702f630b23b45eca614b28d6014487f : FAILURE in 20m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f66b41c3c3d24b3b8caeedbca8582cbd : SUCCESS in 1h 17m 39s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c5e0f51e0d9f458b816d402af5cf2279 : SUCCESS in 59m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/be58947a39974e5193cc3936da5f7e1e : SUCCESS in 1h 24m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f30adb36dd5b41bd83e3aac90240d7b8 : SUCCESS in 20m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d46b3f055e4a48ad98f7f0f84d2304b5 : SUCCESS in 54m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6fb860d7dd17496d9b3fb09700f4b3ec : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7145d9e181b54481b0280316ef039789 : SUCCESS in 1h 17m 11s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/540adc5704e24f32a35129a5f7adf3d5 : SUCCESS in 53m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/525ee685e13f4bfc8336f176b62c781e : SUCCESS in 1h 00m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8030a465a9b2c74e7bdc0dc19f0565cebce831ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 07:25:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3296eb3a86407793df06a222d57fe3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d23a61746444322b475f0e2d9dd9f64 : TIMED_OUT in 3h 03m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba643c6849c74bd0ae4fb21ede2ac1cb : TIMED_OUT in 3h 02m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca8b67dd269b48769643116ba9fe25b5 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed74cbc351b6494b9d0cc7bb3e0bebd4 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2eef32a80987adf4d9e0d2eee4dca352cc4c39e8","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2024-09-20 00:12:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f60ae0234618555ee3b438281d9a6cc32715f860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 01:52:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a24eb73d619e41cc963482a6913488a5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/55e59ac31b6148dc8da385803a1e9120 : SUCCESS in 22m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/07ce683e9f8d4ff2943787374c6cf9da : SUCCESS in 49m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba71989bab2345748cab6a2f68640600 : SUCCESS in 21m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dde3654c0e604ec692d956e6b9d1ebca : SUCCESS in 30m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47da89c16c4d4e35a6afc39dac7ca851 : SUCCESS in 1h 02m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f1486acfe2174db3a149434b02ad35bc : SUCCESS in 1h 09m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/495669397e03448c903011e705c86299 : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca4f53b122d04d5cab98afa2097dfe66 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5dd5c39cbd3499d98ad68e40f0b7f67 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66c923a77ba445ddaeb0ee4fecd12f84 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d82d3a039204d3cb048a6ffaf516499 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90316f257e2544a3a5a488e2278d36b4 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe314ef16bb8403cbace45bb2213623a : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b240ae960d454c088fd1312648a6ba8f : SUCCESS in 34m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3a249aec27584cfcb536a76eaa753990 : SUCCESS in 32m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b3c02c33160042b8a1a2984e0bbbee03 : SUCCESS in 56m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/98fcd40784074c89823ff8f72c36bd45 : SUCCESS in 1h 28m 02s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/31447747dfb64fb4b5efc05fbb63a5d8 : SUCCESS in 1h 08m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4076cd0f662241e5bf68c3b3173330be : SUCCESS in 1h 09m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98db31428d7b406b8d0ee3b33b4d7853 : SUCCESS in 19m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b0f79ae557e345b2a7382d1a1b23504e : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7900b4047dd5445382e3c277d5f08cc7 : SUCCESS in 43m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7288247534ae42d3be1fe375aa6ba911 : SUCCESS in 1h 13m 16s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/569266a72ad1416c811afdf1c1548e72 : SUCCESS in 57m 10s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bfcaa32a8fc74348ba995a9abca1eb0f : SUCCESS in 54m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aa2bc939ba82b108a057bfd07675e41914918eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 03:17:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7424a4a6d2d4459aa97f73bf4b18a26f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/94ade87a16ac48c68f4f869a3f531d93 : TIMED_OUT in 3h 03m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c40691c379c1443e87efd89e8d7a79f1 : FAILURE in 2h 12m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ffcca43313444ee85d7a7cc6294ae1d : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c15bb5b0214478b91baf00acbec04c1 : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c10ccae3c9e30212957816ca50d1871a690921e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-20 06:31:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"46b837f4c3cef2d1fd49b39f2130124e17dab2eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-31 23:50:29.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"094c264a3526401291388362c70e07e762f50f13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 23:51:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fab2f8cff2959f875ea0c325ea865634f01feca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:31:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2512cf591ac14cb8ae89e332d7639ddc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec229065315a4303bb67b824658c1571 : SUCCESS in 19m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2f48d2657a0430ab503c7b309c07a39 : SUCCESS in 27m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cfec9cc068af4e14be889688396541c5 : SUCCESS in 52m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fe0d4aafe1f4dc7a44bcb8c7b0d06cd : SUCCESS in 1h 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008da7efa4044024a48c1a3443d5d6ca : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d0952dd60fd47b98f702459df424477 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a46a74a1d8c4f1789ec3fd9086f6236 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23823726c1d2421daf358a4a7795295c : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a861cbd06b34fe2921181ea577d4499 : SUCCESS in 2m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d78053e0c87641c6976955c07a1e852d : SUCCESS in 2m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/42979b563cf3473896388e393de710cd : SUCCESS in 23m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d80f4a62a1944303aa669d4590f83156 : SUCCESS in 1h 08m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5489b12e051845fdbcb127aa6212ac72 : SUCCESS in 55m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e87b70951655402c900423791a8aeca3 : SUCCESS in 1h 17m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce139915027a4dca9b212580c84c8299 : SUCCESS in 19m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a19860e9234049248354ff7c33274edf : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1f3d8c6e63f14775a338ad9b68135f0d : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/57954dc75071479bbca2f2807112e92d : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":5},{"id":"107ccc170aeb540b52cbb8e916f9f174f6102419","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:31:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"51a0a968e34e810cce9311a378a71f4c6b45084e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:33:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b6b798c761d4a64844f00bdcf387aea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab98b258a7e4521af309a608632732b : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367774572e674047a09613bc8ada7d5b : SUCCESS in 36s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d389964ce262f354bf95296003cb32a85a9f363e","revisions":{"fe20322c48fd83ac1bb874a9ef06c8000a2debda":{"kind":"REWORK","_number":1,"created":"2024-09-06 02:56:32.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/11/928311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/928311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/928311/1"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:47:11.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:56:19.000000000","tz":600},"subject":"Add libnetfilter-log to l3 agent container","message":"Add libnetfilter-log to l3 agent container\n\nneutron-fwaas requires libnetfilter-log[1] when fwaas_v2_log is\nenabled on l3 agent.\n\nCloses-Bug: #2079775\nChange-Id: I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe20322c48fd83ac1bb874a9ef06c8000a2debda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe20322c48fd83ac1bb874a9ef06c8000a2debda"}]},"branch":"refs/heads/master"},"1654d0e4a8274f78917f316610a3a9387669d83e":{"kind":"REWORK","_number":2,"created":"2024-09-09 04:19:51.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/11/928311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/928311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/928311/2"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:47:11.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-09 04:18:52.000000000","tz":600},"subject":"Add libnetfilter-log to l3 agent container","message":"Add libnetfilter-log to l3 agent container\n\nneutron-fwaas requires libnetfilter-log1 when fwaas_v2_log is\nenabled on l3 agent.\n\nAs this package isn\u0027t easily available via rpm, this is only\nadded for deb platforms.\n\nCloses-Bug: #2079775\nChange-Id: I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1654d0e4a8274f78917f316610a3a9387669d83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1654d0e4a8274f78917f316610a3a9387669d83e"}]},"branch":"refs/heads/master"},"c61b8b28d42582b061dad700f6f5f30ec8d5add3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-12 04:44:53.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/11/928311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/928311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/928311/3"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:47:11.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-12 04:41:20.000000000","tz":600},"subject":"Add libnetfilter-log to l3 agent container","message":"Add libnetfilter-log to l3 agent container\n\nneutron-fwaas requires libnetfilter-log1 when fwaas_v2_log is\nenabled on l3 agent.\n\nAs this package isn\u0027t easily available via rpm, this is only\nadded for deb platforms.\n\nPartial-Bug: #2079775\nChange-Id: I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61b8b28d42582b061dad700f6f5f30ec8d5add3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61b8b28d42582b061dad700f6f5f30ec8d5add3"}]},"branch":"refs/heads/master"},"f80fc8d027fbf444794b10637cd748614b6196dc":{"kind":"REWORK","_number":4,"created":"2024-09-16 04:33:19.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/11/928311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/928311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/928311/4"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:47:11.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-16 04:32:56.000000000","tz":600},"subject":"Add libnetfilter-log to l3 agent container","message":"Add libnetfilter-log to l3 agent container\n\nneutron-fwaas requires libnetfilter-log1 when fwaas_v2_log is\nenabled on l3 agent.\n\nAs this package isn\u0027t easily available via rpm, this is only\nadded for deb platforms.\n\nPartial-Bug: #2079775\nChange-Id: I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f80fc8d027fbf444794b10637cd748614b6196dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f80fc8d027fbf444794b10637cd748614b6196dc"}]},"branch":"refs/heads/master"},"d389964ce262f354bf95296003cb32a85a9f363e":{"kind":"REWORK","_number":5,"created":"2024-09-19 02:58:43.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/11/928311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/928311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/928311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/928311/5"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-06 02:47:11.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2024-09-19 02:58:23.000000000","tz":600},"subject":"Add libnetfilter-log to l3 agent container","message":"Add libnetfilter-log to l3 agent container\n\nneutron-fwaas requires libnetfilter-log1 when fwaas_v2_log is\nenabled on l3 agent.\n\nAs this package isn\u0027t easily available via rpm, this is only\nadded for deb platforms.\n\nPartial-Bug: #2079775\nChange-Id: I22d05d1a1c6ab92ed5507d6c90b8c2281abfc7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d389964ce262f354bf95296003cb32a85a9f363e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d389964ce262f354bf95296003cb32a85a9f363e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
