)]}'
{"id":"openstack%2Fkolla~928598","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I0de9928b2657ab5f31580325e866bff28f2b5510","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-19 10:44:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I0de9928b2657ab5f31580325e866bff28f2b5510","subject":"Pin OpenSearch Dashboards to 2.15","status":"MERGED","created":"2024-09-09 10:28:23.000000000","updated":"2024-09-27 09:29:41.000000000","submitted":"2024-09-19 13:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928598","cherry_pick_of_change":926528,"cherry_pick_of_patch_set":3,"meta_rev_id":"aa4c7b287eee8292b7264170f4fa1942dba13cdf","_number":928598,"virtual_id_number":928598,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-19 13:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-12 14:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-09 11:48:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-09-10 09:06:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-12 14:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 11:48:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-09 12:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-12 14:07:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"da3addfd8f2ee4e38a6c32d91e15b251dc7bac24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-09-09 10:28:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dfb30a7a11a30d6a5f8cfe325b162952f168a2e4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-09 11:48:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5204f090c6b19707723da1f44ba9795652b8efd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 12:28:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e775bbca42ad45dbad08324611988dfb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6ed24f64bc394660afde35bcfb5e6025 : SUCCESS in 20m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e07945f0e95f464a8f037b1c7ab0af8d : SUCCESS in 55m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7dfef85ab9b44a8381f8cfc3817e43a3 : SUCCESS in 24m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ed5f1685a945fbb973ee70caf005de : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/274c5464029e44f6ab454146a6453406 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1100ab8f0dae4b838ca74069dc097938 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef0a28aa4244ef0884ca381ebf802c2 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7114b4e19b5c4c08b35d40158f1b6300 : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4c0bcb091e7c49568b7c61262f79711a : SUCCESS in 24m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/16fe6a060b6f47f5912bf4938e0edfa8 : SUCCESS in 57m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/225b907c4c13419fbe8aab681c04da74 : SUCCESS in 1h 26m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/092fda5a8bf548bfa544483e2b4511f7 : SUCCESS in 17m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6533d499b14f4aaea50c519eb4353916 : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2658b8387fc94e6591e5e867953633d5 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2810383f93b163740b88061825050e3cd29eb3da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:13:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e56a207931344648cdfc2650227feaf\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e73b31b52674fd7acdd01dfda9fdaaa : TIMED_OUT in 3h 03m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0396e62f293e44c88ae2e242ed3c203a : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f1b09bda9884a579f84f68b9a07e89b : TIMED_OUT in 3h 04m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c7327057fbd647bba4b4df6a3d78fb46 : FAILURE in 2h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90c5917607b2282e65a091f2568fa0098cfc7ee2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-09-10 09:06:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15b3c0bbaad3f809965d73cc6285c6acd2d8b604","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-12 14:07:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20d059b678fe1d37eed879d55495c839ed2c3d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 14:07:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c696d1f80780cb62706a81061d2a98ff53a4ca2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 16:27:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82c67922df3f47bfa76a6692a2d5a890\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33f1e6c0be644ef9bbd8f371660d3810 : SUCCESS in 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ebe4fb9bbb1492492449ac3d28a510c : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc330715541b442aa25b83e47721dcc3 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/459efa867b0040fb98e235558e60db92 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fdf3bdfb34649ddbeb40c838c545165 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2dfa0056eccf46a8971d2c5ced341179 : SUCCESS in 29m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/71a2c07f3d1a4b08939b8f9a6ef128d3 : FAILURE in 25m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a5fe4639dc3d4a90839155efb167b99d : FAILURE in 35m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/376c94b2c1b844aeb008ef760f5dff6c : SUCCESS in 18m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d9e37827f87b4941b294e9ffb7e43735 : FAILURE in 24m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d8987b2f43f54d349614b648951d1170 : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1da8312aa83fc1c081ee96bfe8510e3d77b80e11","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-09-19 10:44:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afa8ad0f142f470311d1159f59684051a8ced090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 12:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5679dd6a6c54be797e6261e94ad5beb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/001a68da6c0640e490e14a9d9eb455b3 : SUCCESS in 23m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1c2bccb6ea534b9a93879e9031d10cb1 : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8e5215aa96a2405e8b11ad8a57e4fa65 : SUCCESS in 26m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a0078dcc004bd6b014c1fd72fea719 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/870417155c3e4ac985b4948825ffd7ce : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83fa8585fc7f48679e546917484ebda4 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e8421ecaa44f4aa6e0b4aa5760daf4 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f45f33913b134953b4c5ad949832bca9 : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/76649a5e2aaf417ba8162be43f80afdf : SUCCESS in 26m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/274963aff0f94c989ef360f4ecf6f485 : SUCCESS in 1h 05m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fa61436f769a4921837f012db268f193 : SUCCESS in 1h 26m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff70ce2283cf450ea43d837ae8c17ebf : SUCCESS in 21m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8993e41ac30b495f89142e04151f8790 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/404b356159614fe297ca386e7f6d7803 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0b72c0f7d3aaa2057caa544a3c727bbbfbeb0564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 12:22:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b48a6107d51f7aef7e52b6a45e1354e4b3c365c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:38:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a474fa1a1cd445b83dfebd1753656b9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/78d5f30d5ce74cf8b3c30de6c9336f5d : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb5fec0a6337423784c910257c886a76 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba83c07f62c41f68b8d177c2d6c1d2c : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6b27c422484bc5a200b72c0965fda1 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31722a20ebcb4479aae2bb99119c38c6 : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea0de03e14eb40aaa616370a9ad4037d : SUCCESS in 25m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/562aad21fdf64958984928bb1ba0983e : SUCCESS in 1h 05m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ddcf89881d3d4cfdbd02f1468e16c07b : SUCCESS in 50m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8b464e16ffaf4e998c2393753eed2dfa : SUCCESS in 21m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a48d27fdab4a4fbbb015d88b3d8a6b9b : SUCCESS in 56m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f45ef53359284443baae059e05d75071 : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"875ac68c9407019837f152e141399a1c02fd82cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:38:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6cb25c19fcb67830f1ad80f2b787b4b3355a1a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:39:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed501ce5a1c54197b1112d5cc5954d5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af036ce50e0746b28a20b27db980a604 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80a13cbc05e44e598c2369a413725b7 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff62db66205e9209fe194618308295cd92f8e51b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:53:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73682b62fefe4da189f254a78a2f2374\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d0784b8fd9c845a9a174b239235333e4 : TIMED_OUT in 3h 03m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/659d1e9aaa4247788cb716fedb94ef54 : SUCCESS in 4m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d7916fd5706c4165813e17956b98f6cb : TIMED_OUT in 3h 05m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e2fc71118b5640298c19b71848d56004 : FAILURE in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa4c7b287eee8292b7264170f4fa1942dba13cdf","tag":"autogenerated:gerrit:revert","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 09:29:41.000000000","message":"Created a revert of this change as I7dae36f53fa59d9d50c0dcbc4a1a07837972667a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d73fe8e3dad5cffe1691ae22ac4a224a4282a7f","revisions":{"7d73fe8e3dad5cffe1691ae22ac4a224a4282a7f":{"kind":"REWORK","_number":1,"created":"2024-09-09 10:28:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/928598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/928598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/928598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/928598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/928598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/928598/1"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-19 10:29:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-09-09 10:28:23.000000000","tz":0},"subject":"Pin OpenSearch Dashboards to 2.15","message":"Pin OpenSearch Dashboards to 2.15\n\nOpenSearch Dashboards 2.16.0 discover page hangs after initial\ndeployment due to an upstream bug [1].\n\nThis change fixes the issue by pinning OpenSearch Dashboards to 2.15.\n\nNote that it seemed safer not to pin OpenSearch to fix this issue, to\navoid breaking deployments that have already deployed 2.16.\n\n[1] https://github.com/opensearch-project/OpenSearch-Dashboards/issues/7626\n\nCloses-Bug: #2077319\nChange-Id: I0de9928b2657ab5f31580325e866bff28f2b5510\n(cherry picked from commit 3c49991a86cb26dd0447fa748d1df9df5aed92bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d73fe8e3dad5cffe1691ae22ac4a224a4282a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d73fe8e3dad5cffe1691ae22ac4a224a4282a7f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
