)]}'
{"id":"openstack%2Fkolla~929301","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I2cce3dd2540c878878ff0aae80ee5a7e0f455da7","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-15 12:37:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I2cce3dd2540c878878ff0aae80ee5a7e0f455da7","subject":"rabbitmq: Switch to Cloudsmith mirrors","status":"MERGED","created":"2024-09-13 15:29:38.000000000","updated":"2024-09-15 17:10:08.000000000","submitted":"2024-09-15 17:09:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929301","cherry_pick_of_change":928928,"cherry_pick_of_patch_set":5,"meta_rev_id":"956277723d9f270419ebb628caa769e0f1d62865","_number":929301,"virtual_id_number":929301,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-15 17:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-15 14:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-13 23:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-09-13 15:35:21.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-15 14:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-13 15:29:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2024-09-13 15:35:21.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-13 17:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-13 23:14:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6c46563676c50cdf63876a4db4782b273e985f20","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:29:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* doc/source/contributor/versions.rst","accounts_in_message":[],"_revision_number":1},{"id":"9fb9b8d1b7b9b5f7c208583de3611ba2173162bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:30:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cf706236a786e9dca441cba1cae0dfa661f2e7b2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:31:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ca1c03259460c8a3952226c3d11b9f2bfb68c976","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-13 15:35:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1129c46a19bd62ebafbafb79e4fdc18ff7f21414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 17:44:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db8bb3ce8cba41f2a6d1481bd9d18a92\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bc2f02537fe44430925cc435ea8ff72b : SUCCESS in 26m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9a4e8db26eda4f09a9fdf74fa9d084d7 : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec80c5f2b23d4e5fa46cdb9d169df1ab : SUCCESS in 39m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7c31ee33aa07446bbabf2e8baf34c2c8 : SUCCESS in 35m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/477eb95f175148139d1b68bb923fe2c1 : SUCCESS in 55m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e22a770c474647dc932b0aec4439ef9b : FAILURE in 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b607a2aa0b4fcaabd0fdf43dca11ca : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c87ad0a93db4c8aaefe281d69365f32 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6062e80ebedb485aa9a6880de9904025 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1f07be603024761aa1f31886492c83f : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cb80256f73543fd9b53aad59ee41f43 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33be4edef8da41fb888f7c92e7ece95a : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00e45256e1f47ae9bfa08079fd23c72 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa01d1612064592a125480896633791 : SUCCESS in 6m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/75addd71c2004216aebf81bb3e376079 : SUCCESS in 30m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/74f5934061a84ca58f65c36f16d68fcf : SUCCESS in 38m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cd92a7a0af4a45d5ad0677ca6625578d : SUCCESS in 1h 01m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e50ae9613e8e4cc99e02a077fa3773f1 : FAILURE in 30m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c3371347f8724bfea88dabd947afdb14 : FAILURE in 1h 24m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/46a9437d34b94a09b3d2798e506443a9 : FAILURE in 51m 23s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8a1244203b74457fb3c330a5e7e24311 : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5ac024b93c1a4236934b9a572b0b7f8a : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2cc58f18e66c4bd78e594d837c6f3014 : SUCCESS in 29m 05s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f00426f9615b442b89bd8ce4a8a357e8 : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7d822eb71633478192c234da1b7f23f2 : SUCCESS in 1h 19m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f3235ebf4a704554b54f5b1ad6eab00d : SUCCESS in 1h 31m 39s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fa8ddba7a22b4ea996b0f013bc6ebb2d : SUCCESS in 39m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6833f86e7f1343f693e00ce5520d4bad : SUCCESS in 34m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c166de7e48284ad2872d976703a882ca : SUCCESS in 1h 06m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/47c52a75111d4eab8705cb327440927e : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fd8b9af7c88b4b559adcc94ab1649ed3 : FAILURE in 27m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d929a1cc0aaa4772862f23103ef73b1d : SUCCESS in 1h 49m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8088d61ebd734077a66f0e5dc78167ba : FAILURE in 32m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a6dfe13667ae46c886e597acd63c7587 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/25d4f7773430444d94f4da3b404e21a0 : FAILURE in 1h 00m 43s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/644fa844f35b4d9a8fb53ed07d7131ce : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/019ff9ee5c67453bb14f2ad8c6b1a11d : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d1b804a3d6d24325b86330e8cbba51a6 : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/70134af11fd34daa83fe77fd6daf84ff : SUCCESS in 1h 09m 17s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5a1058c5ee2d44f9a161a6ca513b613b : SUCCESS in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2021258ac5787cebe0582c5749cbf20190c8e68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182122f03d5e4eac9b1d4cbd1fefcb92\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fc7f5f6f6fd1454b96b93f99480c3cb9 : TIMED_OUT in 3h 02m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ccfaf3e4c7474d0cad50ce81e1574ad8 : SUCCESS in 2h 39m 18s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f84c3540d7bb42249b5034fddc0caa7e : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4db182f20c2a4c5db76a47b4b53506c2 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3df36cab15be420598fb0d831fbb5d40 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abd43d06b5ba47288754d4b10885673c : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ef255d5338c5bd10c79160b516ce239428a7f9f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 23:14:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aee670a208bb8ae4e47b5812f742d5cf515e2597","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 12:37:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3a1aa889da0127cbfe36ea6b6de615fdbca5f22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 14:31:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e1e5de45667a1dd246f4cdb0ca421973f037624","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:05:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ed7a2b4e5c426d8b41fb4216480354\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e627fd2730c46c89b63efc12f93162e : SUCCESS in 2h 26m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ac9664faea44f6ea682e0570fc34e59 : SUCCESS in 1h 35m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94ff58e8376e49a38637955e9aa3e46e : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0db2f0ab9e9413b979fa654ca63cc69 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c11de87979d244189d752085cb5aba8f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c81eb5f673034f0a8f9b71092f694c0a : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"daafc153ed241c8d4030361067fef72cdfd5e239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:26:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08493f158ad049828f77c8349087c431\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3d5712d0518744a4b147977f0ab4e0fb : SUCCESS in 26m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/afcd52ffb48946d9ba241f4661ba49a8 : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf0772969ac949cfa2ab8a008456d09e : SUCCESS in 22m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a548e3c15a0346099432d2d3184b257e : SUCCESS in 30m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7a90bfd274c54616b5824a94f2a612bd : SUCCESS in 47m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b52717afe2f646a788b6864594aab012 : SUCCESS in 1h 36m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b945b0f324764810b828a5dc9fd75ab4 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7165496062194962842adfcc1e735ead : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c431e07da95d44228154c7912f6b5024 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd26960461cf4082a82c3f2f50c70f6e : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20e27917fa08443ab207e8c86c427ae0 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e76299415aa34d17b5426a174cab46f5 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7840a8ab30f4445bba27fc4076d56885 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e30a483713e24d1384ebef2a7921b664 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e0e5e85e17145108ae5bafb4f43754b : SUCCESS in 26m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/271cf844a86a4334be0a07ab6a110651 : SUCCESS in 34m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8808d9ccd926469a98eb0df82677640f : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/930e8bfe129449c293f8f815ea49df0c : SUCCESS in 1h 24m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1770596aeb5e4b3fbb9ac2d4282e91da : SUCCESS in 1h 34m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/97339fa1dee7467f9e0743eba0b1eacb : SUCCESS in 2h 43m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c9f076f5d394462fbf1e15c83b98619a : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ed73bde492034974b4dd0a5c2287e488 : SUCCESS in 1h 26m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6c363c8047d445719691d155695b6d57 : SUCCESS in 29m 42s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/064701f049b849039425f5910780d5e8 : SUCCESS in 54m 54s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/da7878769c8e4211a1c08bd10b66db90 : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/16efb56dd4c543d9b5809b0b00306e9d : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/477928e7f4394210af723f8a4b7fc9bb : SUCCESS in 24m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/80ef1f2003324e74921ec2854703a91c : SUCCESS in 23m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a89353bc0abb4f71bb13b3bbc61a5105 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d166691f34ef468e8fd85fe3c66cfb6d : SUCCESS in 47m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ca0eec3cb001436fb266f02c587dd1cf : SUCCESS in 1h 08m 24s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/42b804d2819147438d04ec17cb115b59 : SUCCESS in 1h 20m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2cb60fe3c1ac40a8816851eca3ec0c11 : SUCCESS in 1h 58m 19s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fda5014820a048bca76643689d6ce852 : SUCCESS in 56m 03s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b8a8ad68eecf4c519977d2918afe49d0 : SUCCESS in 1h 15m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d5ccf512230e443dbf1ddf75ee7a20ec : SUCCESS in 23m 17s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a445c06a2ee04dae83b4efcf703d5344 : SUCCESS in 43m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3d9d9f4fedcc46c08b794f7590f66c82 : SUCCESS in 59m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ab49a323496146828c3a6f1d5dc3eb35 : SUCCESS in 1h 01m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b8c1fa264a644e7d92984e1e98cac210 : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e6fbafc5116593c4182c3a1e7776d3fc64d7c62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:26:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e73f52c9184cdeeb358355543dd417265de4a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 17:09:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9eed73f2bd04cf7967ae6162c45702a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/903e865380f346568592e45a99cb3642 : SUCCESS in 17m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f931da116f54d5fb748d2e7034e789f : SUCCESS in 31m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d198d64f20624326ab5e34dcdd37d357 : SUCCESS in 1h 00m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b8a3f688d0944386b7205d7e96ea1763 : SUCCESS in 1h 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c89402bf384d0280e4121e851e458f : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20a04fd990da4d0c91ad9da17e24f7f6 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1140d23f2004c1a8b49be86948247b3 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a18cbe93f12f49e191474fa692741104 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e08bd39db28425b89f13d314ad720ad : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70939f3fb2934a619ad8c013fc45219e : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/72b496baa83f4984a824fc6b7f0bdf32 : SUCCESS in 28m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d8ae168d56342de93c0de89cba219cc : SUCCESS in 36m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ececdb074b9849e5942ac4d6dedb79a8 : SUCCESS in 59m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/66d5585d107a4cbdb525eb21cf995f4c : SUCCESS in 1h 22m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b62f280ccc3e48ce87e7933955b018c0 : SUCCESS in 17m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3692f8a30154173bb3ac61fe330b5b3 : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b1db5f12539441b286932f85366c5168 : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3a91c1245ca74430af833c5618881bd6 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4ea70e5c1824e9043f9377d53806eccef4f968b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 17:09:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"956277723d9f270419ebb628caa769e0f1d62865","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 17:10:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb70464889ab45a59f761daf6439a573\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67199511ae694d92bb0ca61bd7931321 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94bb22dd632c41febca3c7edbe4b232e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73051ccde779f6d5a3c91d6fcd87ab00b1ac809a","revisions":{"81e08f34938664e819aea92b856552202ecabd17":{"kind":"REWORK","_number":1,"created":"2024-09-13 15:29:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/929301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/929301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/929301/1"}}},"commit":{"parents":[{"commit":"1f156fb394370b3dbb2c26a1655d830f9720f6a0","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f156fb394370b3dbb2c26a1655d830f9720f6a0"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-09-11 14:48:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-13 15:29:38.000000000","tz":0},"subject":"rabbitmq: Switch to Cloudsmith mirrors","message":"rabbitmq: Switch to Cloudsmith mirrors\n\nPackageCloud is discontinued as of August 18th [1].\n\nAlso corrects the versions docs for RabbitMQ and Erlang.\n\nUse new RMQ Cloudsmith repo URLs, these now use\n*.rabbitmq.com domains [2].\n\n1. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#packagecloud-will-be-discontinued-on-aug-18th-2024\n2. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#mirrors-now-use-rabbitmqcom-domains\n\nChange-Id: I2cce3dd2540c878878ff0aae80ee5a7e0f455da7\n(cherry picked from commit 528b7eef167996041f7cd3b2184fbcd6b5ff57fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81e08f34938664e819aea92b856552202ecabd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81e08f34938664e819aea92b856552202ecabd17"}]},"branch":"refs/heads/stable/2023.2"},"73051ccde779f6d5a3c91d6fcd87ab00b1ac809a":{"kind":"REWORK","_number":2,"created":"2024-09-13 15:30:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/929301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/929301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/929301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/929301/2"}}},"commit":{"parents":[{"commit":"1f156fb394370b3dbb2c26a1655d830f9720f6a0","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f156fb394370b3dbb2c26a1655d830f9720f6a0"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-09-11 14:48:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-13 15:30:52.000000000","tz":0},"subject":"rabbitmq: Switch to Cloudsmith mirrors","message":"rabbitmq: Switch to Cloudsmith mirrors\n\nPackageCloud is discontinued as of August 18th [1].\n\nAlso corrects the versions docs for RabbitMQ and Erlang.\n\nUse new RMQ Cloudsmith repo URLs, these now use\n*.rabbitmq.com domains [2].\n\n1. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#packagecloud-will-be-discontinued-on-aug-18th-2024\n2. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#mirrors-now-use-rabbitmqcom-domains\n\nChange-Id: I2cce3dd2540c878878ff0aae80ee5a7e0f455da7\n(cherry picked from commit 528b7eef167996041f7cd3b2184fbcd6b5ff57fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73051ccde779f6d5a3c91d6fcd87ab00b1ac809a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73051ccde779f6d5a3c91d6fcd87ab00b1ac809a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
