)]}'
{"id":"openstack%2Fkolla~929302","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I2cce3dd2540c878878ff0aae80ee5a7e0f455da7","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-15 12:37:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I2cce3dd2540c878878ff0aae80ee5a7e0f455da7","subject":"rabbitmq: Switch to Cloudsmith mirrors","status":"MERGED","created":"2024-09-13 15:31:44.000000000","updated":"2024-09-15 15:26:06.000000000","submitted":"2024-09-15 15:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"929302","cherry_pick_of_change":928928,"cherry_pick_of_patch_set":5,"meta_rev_id":"0e8ae567c2928b523e68870d121b8dc33ea83231","_number":929302,"virtual_id_number":929302,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-15 15:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-15 13:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-13 23:14:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-15 13:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-13 15:31:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2024-09-13 16:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-13 23:14:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b24e2fd1caa8bf7ce5f636474ebc8aac2ece4d0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:31:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* doc/source/contributor/versions.rst\n* docker/base/rabbitmq_rabbitmq-erlang.repo\n* docker/base/rabbitmq_rabbitmq-server.repo","accounts_in_message":[],"_revision_number":1},{"id":"ccd01ca1a229e0ac0d419d7df12c9a774c463513","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:35:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7daef4b63936532b048bee6ff7935bf71f1e9ee9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:35:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"fc3a710f5504e07fc249f2e5c3c4ade560774da6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 15:39:25.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1be4d5da31caaa0cd279ad9d5944c75971534d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 16:31:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a364daaf4bb4ebeac64e381c1a99777\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9b21d5d599f84b54a35e19afb2a8b460 : SUCCESS in 25m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f9e0f8c4bfaa40abaed199d732e4bd3a : FAILURE in 25m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/19f38d656a954bea88b117fdc04ee07f : SUCCESS in 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0817ef1b805f46c2b76b42f9327685d9 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41a1e189557d4b11a93a878256264584 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc4d4db4eeb4dab8083633f100ccc1f : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da65dd0689324e9f99b5eb3648d91656 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8480759dafbd43a3bd22cae7bb77b2da : SUCCESS in 3m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47a66f4c85574571b101865a519e1259 : SUCCESS in 28m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/30b3fd10a4934792b0896b7f8da09e9d : FAILURE in 29m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a2fc4427702941d8a6e1f2b384a9614d : FAILURE in 34m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc94bc09f4b34a69883ae19a92ae0f27 : SUCCESS in 21m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/349ddbcef03f46bb81393c3bc0cb771d : FAILURE in 23m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/04503c485cd04ffb9f94adfa0ee820f6 : FAILURE in 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e823a841dd444782fb5a143165b870776e550f62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:57:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73d7362aa41f4a7a9c38884ad913e4da\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0aec05b11aab4ad6a2412c0e54603bcf : TIMED_OUT in 3h 02m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b4fa7b7fbe2436c99b80efa8d04e1c5 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/41c271fc67404c7ba26db14b5ffc7162 : TIMED_OUT in 3h 04m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a7cfb02fcf18464f9f3b00af2b52ca8b : FAILURE in 2h 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aacc4dc42d5f63bb3a3e40fd115eed5180c4739b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 23:14:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"34daf8c51cfcfccf557e45d8ec0a293908bac3bc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 11:43:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27fcbf53b10159bb4509ce2f6878052c982debdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 11:43:29.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"9a20d0a9f9c469ec03a0c135f41482080fb4645b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 12:22:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e4275047576407ba17cfd4a9bf4185c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b2f086be1da740ea9d6d7c50488c7339 : SUCCESS in 24m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/acd493aa852d4a0c901d8b25de8d7270 : FAILURE in 24m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d1c6a2ad33814e04bfc2ecf61f1f59d3 : SUCCESS in 22m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b74b84f2b94460ba14fe6c44c740be4 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c4fc51cab140f581cc5802e9a4acc6 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece1c4cda371424aa5f59a0609fc4299 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb314f50c4f143ef8267fe99e267a9d7 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bcd5005bcf4bb5b326b8b9512e2f9d : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e8ddf12b90ec4d549454c0ab29e4f7eb : SUCCESS in 20m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a62820960d9a4e0e885db14467778b42 : FAILURE in 24m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aac984732f6242a39a5ba440bdd82deb : FAILURE in 28m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/316c0e3d2fc14a1380612f113e51372d : SUCCESS in 14m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/19ec604581ac47af9cfc00a30e303f0e : FAILURE in 15m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5bde3baba4ee4f62baa17f41856ac17c : FAILURE in 19m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ee41616c9729cc764c28901d5e932edaffb68c5a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 12:37:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"149682f9110ae9598ccab3218b73d60d5844c6ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 13:29:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18441c5cd455b62c89740581146bb7efca32627f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 13:51:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f00f7becc3745e2841f98ff9cf95ad6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/61012170177c4e27a503a9a03206c1e7 : SUCCESS in 21m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d3f9218b60a34558af9884f670883dc2 : SUCCESS in 52m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a2ba4db4d254c9f90a072a0e27d6d5e : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88754b23a499438e971ce9073aa41f9d : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf748459c634bb48736e8da1a8a2d99 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30c47ab48e0846f8898eefe1e9ac069c : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4381150e65bf42cc86b4d7126d2a110d : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/058fa157e0ba43a9bfb4d5a02345aad2 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9291cf76d2304c74a1551e03d3a2146b : SUCCESS in 22m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9242b4bfbb3149e498371f58e80b6c89 : SUCCESS in 1h 04m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d077c287ab8948aab488b6e5fd55846e : SUCCESS in 52m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dc24dee6cefa4f2ab0b556cf146a157d : SUCCESS in 25m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9788fae5ddf14e55b6b51603219d8175 : SUCCESS in 44m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/618fd54bcb8c4af5bda62e5ab43e0a26 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":2},{"id":"14250229beaa1baa287f471bad379bef252348e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 13:51:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e43a41cd45f024821e1b4c65f5dfd0d1874cd25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 14:26:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd058f8761714431bfa22b2a1ea40415\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f29dd03f83154a8aa95863a9a0de1f9d : SUCCESS in 2h 40m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0608b3468f074dc5aa34899fc789c722 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e2399b2995c348e2a052c9306ba090f9 : FAILURE in 2h 25m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aefb382377e94667a209ee3abfaf53e1 : FAILURE in 1h 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35c95e64d0443832aa2d3481f6a8ee9dc953cb5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:25:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f5be2a02744c98a6713ca5c79550df\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9792ade9d474492eb38a2becf6360f81 : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e05f2352e964745957d023b3bc53b95 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8f002a5c5db4cd086b62705055dc398 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2af60d1be244ed5bc32d0606be98b79 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/767e6cd8d0de4c45956c54cff68ae622 : SUCCESS in 3m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/75251b30f35b4876abb023884acefe62 : SUCCESS in 25m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82d1911946704a74bcf238e749c25fc5 : SUCCESS in 40m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/580a5c645a5443768243a64adfa808b4 : SUCCESS in 1h 23m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4b2a5e3598114c3a8938d5cd97800cbb : SUCCESS in 17m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/03aec3fde87c4206b754515bd9457cc2 : SUCCESS in 45m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5bfb4a0aa02b4c2fad7520c202234f0d : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5e32fd5f8fc9cdd32161c2af843bd87ada810076","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0e8ae567c2928b523e68870d121b8dc33ea83231","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 15:26:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472755e190774878b48412137205534e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4010c4fd0324815a32c1897980aa4a4 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/653520ac96eb4e13ac2bc1a93de3256a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f9dc8fbd3532ae16ac31042d77ab75aa6c001c1","revisions":{"80f1136ad5569555d6eb2b0817c5e6940b84ba87":{"kind":"REWORK","_number":1,"created":"2024-09-13 15:31:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/929302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/929302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/929302/1"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-09-11 14:48:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-13 15:31:44.000000000","tz":0},"subject":"rabbitmq: Switch to Cloudsmith mirrors","message":"rabbitmq: Switch to Cloudsmith mirrors\n\nPackageCloud is discontinued as of August 18th [1].\n\nAlso corrects the versions docs for RabbitMQ and Erlang.\n\nUse new RMQ Cloudsmith repo URLs, these now use\n*.rabbitmq.com domains [2].\n\n1. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#packagecloud-will-be-discontinued-on-aug-18th-2024\n2. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#mirrors-now-use-rabbitmqcom-domains\n\nChange-Id: I2cce3dd2540c878878ff0aae80ee5a7e0f455da7\n(cherry picked from commit 528b7eef167996041f7cd3b2184fbcd6b5ff57fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80f1136ad5569555d6eb2b0817c5e6940b84ba87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80f1136ad5569555d6eb2b0817c5e6940b84ba87"}]},"branch":"refs/heads/stable/2023.1"},"6f9dc8fbd3532ae16ac31042d77ab75aa6c001c1":{"kind":"REWORK","_number":2,"created":"2024-09-13 15:35:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/929302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/929302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/929302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/929302/2"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-09-11 14:48:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-13 15:35:12.000000000","tz":0},"subject":"rabbitmq: Switch to Cloudsmith mirrors","message":"rabbitmq: Switch to Cloudsmith mirrors\n\nPackageCloud is discontinued as of August 18th [1].\n\nAlso corrects the versions docs for RabbitMQ and Erlang.\n\nUse new RMQ Cloudsmith repo URLs, these now use\n*.rabbitmq.com domains [2].\n\n1. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#packagecloud-will-be-discontinued-on-aug-18th-2024\n2. https://www.rabbitmq.com/blog/2024/08/11/package-repository-updates#mirrors-now-use-rabbitmqcom-domains\n\nChange-Id: I2cce3dd2540c878878ff0aae80ee5a7e0f455da7\n(cherry picked from commit 528b7eef167996041f7cd3b2184fbcd6b5ff57fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f9dc8fbd3532ae16ac31042d77ab75aa6c001c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f9dc8fbd3532ae16ac31042d77ab75aa6c001c1"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
