)]}'
{"id":"openstack%2Fkolla~929446","triplet_id":"openstack%2Fkolla~master~I7510b68f4096134bc9d106fe14bfae2bd7b8b012","project":"openstack/kolla","branch":"master","topic":"kolla-patch","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-25 19:18:19.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-11-25 19:18:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7510b68f4096134bc9d106fe14bfae2bd7b8b012","subject":"[DNM] [CI] Test kolla patch","status":"ABANDONED","created":"2024-09-16 10:57:57.000000000","updated":"2024-11-25 19:18:19.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b12d8e554990f1286b97dba03b873781636ead29","_number":929446,"virtual_id_number":929446,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-14 09:46:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2024-11-14 09:05:07.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 15:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 13:17:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"69658d777a6ccdff65a97aae79f3123784f03282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 10:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae009c6162d7353f0ef7e06e5ed821482af8fc5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 11:30:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5ff5ca5574c2e367920cfc1a1c2ea7537b816530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 11:58:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9b4cf07f5fed065eb480e541a9368394c8728c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 12:06:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e2890b03e6d2cbf2c71a8503edea8c788c04099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 12:19:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f8b625e310a082822ee063a82e05fa4558b55f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 13:06:06.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c7b1a5cdadff0bb3a577c7619e85e4a0006f7362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 13:28:24.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"f2244748b7abd9f5fb755381757433c7720bd071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 15:05:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595c1ede1c2f4602b8542c08475bcf2a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/03e7d1c6704d452ebe05a68e7d98f92a : SUCCESS in 24m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bf0a30259b674ad9823043f3c0020f05 : SUCCESS in 1h 19m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d16c7481f1843608ecd0a5b8285e32c : SUCCESS in 16m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/463eefca54324d978d56d870cbfc87b3 : SUCCESS in 40m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/06375ef3ef3e4b5d88e7a67874733ac6 : SUCCESS in 1h 09m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/af109d933b304333803eaeccc0f77290 : SUCCESS in 1h 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27293d58d13f4e21b680f0ba7b0a8609 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3349f8b26564e3eb2f2be2f75b4a745 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9725605bed004c98b539e52a6f5ed321 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51de89f0fe2d4a8d9e0c16fbd4fed69d : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/880048afa63e4343ab64e4f7ca4287a3 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c3ac0a898242d18823c4dc8cafdc8d : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/edbd1fa339cf4c3493d3f934003ce458 : SUCCESS in 28m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b534c213a467476c94713adaf690a956 : SUCCESS in 41m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d79f00e537604dbbadfbefe529365ea2 : SUCCESS in 1h 06m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/807ba021680f419c9ee9d8dc4ecc26c4 : SUCCESS in 1h 19m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8feb1d6ef66145e4897605d7f5e32fe4 : SUCCESS in 24m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5e8272381a67418f98affd3ac497845a : SUCCESS in 1h 15m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3cfb31b24b7c42da9fedbed31e512092 : SUCCESS in 46m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/43a711fd8dab410e9ae0bc366eadb53d : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":7},{"id":"2ed43091e5ce677cd74cb1e695975ddee59c4e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 16:36:24.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f5d8ad23f205dd308e2c39243f5b3eec3091759a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 17:33:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"816f04c56877172fabf62e52948f78f893f31d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 19:11:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0171076504d74239888c18a61ed65bfe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/538ad1d6bfbc46bd95a76552c7542dc7 : SUCCESS in 32m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff82efa788994f75b17131cf45408c4c : SUCCESS in 57m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e3f446ed32a94876a1f42ff56c93a4af : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72670883592247b4a4fe946d1a0c27b5 : SUCCESS in 32m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f951103577c4cebbf12f008e5591645 : SUCCESS in 55m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/552565d84233497e94e2e6c6070a7126 : SUCCESS in 1h 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa98da4cc15245cb9bd8056c2eaf9437 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0705d3b53434c3fbdf8368eee504542 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d065106a96f1474b9f948572fd15cd94 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1403d6ab113d4d38ab219be9960b2665 : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4b634599fc547bfb22f9ce422fafa5c : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9d80970e9248a38df41305289117d9 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ad6ee1c4bae3418182471c320b2b87cf : SUCCESS in 30m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d384f2268cd448149624e347eeeeb0bf : SUCCESS in 37m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6c248e69c4ec4cd097f42bc76709a015 : SUCCESS in 57m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9af91f25d7f45e498d4fc8ca4a42ee2 : SUCCESS in 1h 27m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9cbb9a5eaf3c43b8a60d766cd1738f5a : SUCCESS in 24m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a03c45c364ba478d90f83a120887917a : SUCCESS in 1h 00m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/78fbff2b296a414db3bf3f61b52e9a06 : SUCCESS in 43m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bc7f6191d743496daec2e75ae0da659c : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":9},{"id":"0179a0f968c611cf2ea9cf683b0846fe63e79553","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 22:06:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519bba07b1fc48d0b0d4dffc24f73fdd\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e26462b8655f4f76a4b01a9963ef1972 : TIMED_OUT in 3h 03m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ec6e93c0bc2412ea9d2a7b219fee31a : SUCCESS in 1h 58m 24s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f22cd2b639dd4654a79b078586ade31d : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3beebc4541e4f279a223bf3ee17df95 : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"02558bab6a28c2bc225be4c99eeb6dd5928581b4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-20 23:18:22.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"dd54b6e5d20c5cabff702b979eaba3dbd25c6b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-23 13:59:00.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"d8c7b0ab35f38fd4b1db1602e893c26ade619af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 15:45:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc330921ccc4483db5dcdc9777f90683\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0dbe65acbdc64f2fb859f06ad8983264 : SUCCESS in 23m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0efb5cab4d134b9fb408e730b66d7985 : SUCCESS in 56m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1c954b3b04a642339f97dba10be258fc : SUCCESS in 19m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0328d5d9fa9f4033b660e9721c468d3d : SUCCESS in 30m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/061e0f65357148b08bd9443d43c47fad : SUCCESS in 1h 00m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/272154b65d9e47a4b9c2ee33d330f7cb : SUCCESS in 1h 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28912336fac84289b72228ccfa61fe3c : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b008d96938e482f9cfa10449a3b1180 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec65f573b8674e048e225a3fa05a7ca2 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93eecbafda3f4fd7a7bf66b777648b04 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f00805288b4ebc817041903c3f434b : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3566968379b44196b020051c261041f3 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ca5a5b3a18f47cc822dd0eeb6b8d1ea : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d1c8229da064028bf5b7aab99227de0 : SUCCESS in 39m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f80ace2ce0734cbbb7a490c864c5d506 : SUCCESS in 1h 06m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/14b0f46c082e4f38bda27a8af0209bf6 : SUCCESS in 1h 33m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3107a517cf444c2b91aa85121601583b : SUCCESS in 20m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/54ef53fe7fc449f1b7bcfc1fe469be18 : SUCCESS in 1h 38m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af14ea68f0a945249848694ac7517fbf : SUCCESS in 51m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/836adc5706fa4cc2b5692342cc17fe2f : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":10},{"id":"524ff7e26a618170722249ba948c50fc50c5d2b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 20:02:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/796ca35f4e094439bdbcf751b8671f8e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b46e0b6a361445ab85ce1d79e7797174 : TIMED_OUT in 3h 04m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41dd9fe4644746bc815633c5d85c0b36 : SUCCESS in 2h 39m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2ddcebd4cb34953bae0852e1252ee2c : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eecfcfb403e04290a87a02a6feab9762 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"895caa257a4d86df2ba7d24fe11094e15243c36c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-30 21:10:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"03f304502a63b6bdf25b69fabc4f5e7745cb0e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-30 21:14:33.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"17214ab358801db2c452c0ae43506ac5a25744c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 23:16:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f608046029e4d07ab0b1714f770f80d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e6ca416afc5c4ce28eb42291d95df440 : SUCCESS in 28m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de6fc99f82b5456388a633fd16fc5e6d : FAILURE in 8m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/552b7cc3e8344430b0a041d2230f9285 : SUCCESS in 24m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/023c142fc2384a779da9caa6d12eb457 : SUCCESS in 33m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7e66351ac5e54d52a3050cc8ec86b284 : SUCCESS in 55m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e16e3fc3c4274a9882c919fb3570b14b : SUCCESS in 1h 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b176fd745c1646aeabee7c2fb17d9828 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43632e97215a46bfa03a4e3764b18f0d : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b0c42c5022d4a969469286057269c14 : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/709bc629c7df4b7192420a13947fe902 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c231230614461aa0118a8bb489dbe7 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/52321f28690e4993a50a252ab0ad2725 : SUCCESS in 29m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/96502668f62146e2a383d0289cf84844 : SUCCESS in 42m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e80580d208d242c4b868ab668539163b : SUCCESS in 1h 00m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2a8d2bbf960b4a26b7cae8b2277d5377 : SUCCESS in 1h 12m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/16e8f59dba644dd8ae058c0bd4e45852 : SUCCESS in 20m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/da71aa4b14be455ca71be9b2d156bdcc : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/abb3928eca544b75a80067e88e625957 : SUCCESS in 51m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/68bf581b3e5340c083d95ad3dddf23cb : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":12},{"id":"9d939308eb14ca9c045c92097a784130942be4c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 01:11:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73fae287137d41df8db4cdd964f518ea\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c71c33f1e3742f8a6ce303860976c64 : SUCCESS in 38m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/72ce8ef6924240e8b75ce594fa7d5902 : SUCCESS in 2h 00m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/250aea2cb2534494a1e8fddd3f12d6ad : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8dd61de7d44e4ae897180493f02f30e2 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"275dcefc48aa5029eb7958c9aa6bfeb5a3715561","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 13:17:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f221ee9a5545a185c98c4876dbe1ccdaee3f08f7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-07 13:43:48.000000000","message":"Patch Set 12: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"90d205d05e982d3e004b4ecb61575daa4603cf4f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-07 13:46:31.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f0bad3d5702337718d418e257c31a77897cd80ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-04 14:06:42.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"d86fae5d4f1c87074836c5598a9d2fea33e68d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:55:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a4cf01f16df4cecb8bbf5f57154828f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/abe8866b3eb548a09b84e9d0ea6b9607 : SUCCESS in 25m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f2d07ba4d5a247d9a11e083734bd3c78 : SUCCESS in 57m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a13e03f0fcd44a119631a39c04338ac8 : SUCCESS in 26m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c2436f5b15384ecba9a578d3e8a0c051 : SUCCESS in 36m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f2d6585bd0544e858c0ef9630f2400c1 : SUCCESS in 1h 07m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/54a0177e99d6449b9e9f2d107075d49f : SUCCESS in 1h 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e9964f24a5d4f148851fe2310185695 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc8529c9df6e4e20a1b958b417cadc8e : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/099c1e110e1644408abef34bcbf96de3 : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e79f4a4d493b4b70860eb7a1502218ef : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab8204abbff4fec966d1d5270d83570 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fe2a59670e96489fb696a1b287c0a501 : SUCCESS in 30m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/656065e9f8d44681842103537f7c1665 : SUCCESS in 37m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9b599f60b0d2489fa3da21ffd7ba24e0 : SUCCESS in 1h 13m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd54574374744365866a5933b96bf706 : SUCCESS in 1h 25m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/092adc630e844bccbd3de316d44130b9 : SUCCESS in 22m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/72b84c4d1ad94041b1e0f74b74e54633 : SUCCESS in 29m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/12aae8e853764233831b44e0e034a387 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b5d23ac23dd24bc3b4b29c2f252e2843 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":13},{"id":"520beda6f6179e5700a80082bd88bfa05db4167b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:37:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157f1af0d0ea40c09022e271ac4081a5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/233bdaf8900847cd9df29e93b9686e30 : SUCCESS in 2h 14m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ec024e5717346bbad001f576f011596 : FAILURE in 2h 28m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34a02638986841e0a5635f5508c61bf5 : SUCCESS in 2h 25m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71a2fa5f1f6d4abaa3f997f6039f8294 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ab31c32b7d24fc0a7695330c8069d18 : SUCCESS in 4m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7296f39caef3462791afb8dcd73fd464 : TIMED_OUT in 3h 02m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/da7a1db9726540e7bc856e8fccf0ee9c : TIMED_OUT in 3h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c6e28325b261ac004c5555213bfdba30b02fdfab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 07:50:12.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6e1f9087ccca6a143c2610da7b7860129af9f84c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 07:52:47.000000000","message":"Uploaded patch set 15: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"fefeca8a1d336744122e0c4fb210862b5c219d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 09:37:06.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":16},{"id":"de84e08447bad9466a0fb30c5986be9c2c56be19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 09:53:08.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":17},{"id":"90bd139a5d0e5d400e6b43b8fc83ecee6456c2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 11:43:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24f9dcb9cec412cb1de7b328f64287c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/96bf1905ab8048edb5c6121a71973829 : SUCCESS in 24m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a2a14a82c1b84e239c318538b210d3a4 : SUCCESS in 1h 07m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/62deb02a139f4c729c59ec9b03c11f2f : SUCCESS in 23m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ca60cdaaf95c47fbaa5ab004b398e69f : SUCCESS in 40m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0adaab6a50ce46a4965b3da8193beba5 : SUCCESS in 1h 04m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b7a51ff6c5e34fd79dbaaca021afb43f : SUCCESS in 1h 39m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c414173ae5e4f859ab7a2b5917a2c46 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53e93dfe2aad40628d73316ab09765f8 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d0d15ae38204de584f5b1e1cc32512b : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/130cfbe4f4e840e8aa93a02f48bb735a : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8587a9a766d546cd83169a097e936ff3 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bc1c1c9372de4a85b4639c5f1f27d8f4 : SUCCESS in 29m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f9d1d5765173455584cab6091421b533 : SUCCESS in 43m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/87644e17846e492cb1c48325fb2b289a : SUCCESS in 1h 21m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c59e7cbf82a5442781de773ceea0c59a : SUCCESS in 1h 27m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0de58b9500c945b68c22d3e85eeca3cb : SUCCESS in 23m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8c48925f9e2a430eb13876c938bd23bb : SUCCESS in 38m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ad24e30cf99e40fdab6beb7b9883e058 : SUCCESS in 59m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4fd73809c44c44168a2014bead794310 : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":17},{"id":"98847213121f1fc1a0ba2dcfb6b7e574a8c3e72b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 13:11:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ec0a67f0b84240b08b87dc788e6508\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e0c0449d4813464ab22d7bba42c0e646 : SUCCESS in 2h 48m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4c7c82bd121d45d9b1a3d394000c057d : FAILURE in 1h 24m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01e5f7b9505c47658e425b61a8173b84 : FAILURE in 2h 09m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f97fcdee973440f7b6b10e829f38b67f : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db561dd6a7a44e98b09f9228a55c809a : SUCCESS in 4m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6b3d6d2f6ccd41b582bc06dd0ac79922 : TIMED_OUT in 3h 02m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f89f2286099e49019b9705b9acb1bdcb : TIMED_OUT in 3h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"27a15147e02dc86f5d2ade4d7c3c1e2e0b6f7e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-14 08:52:08.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"b0adb4ddc71313c1e407f4d5dbbfd13af5d55a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-14 09:05:07.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"6bcb274625897814325d5ac0a81f5ef7fe2707a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:28:32.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8349ba129bd940669849c284ba50ba27\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0358610d682744cb95cf12a5aed44366 : FAILURE in 5m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a82892c8df74e6c94c13d10b7d02638 : FAILURE in 4m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15bf6809705c442d83df2aedff0d3f71 : FAILURE in 10m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8b421aaad1e4ef491b73e03c6c34744 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc7d5b0e987a43d69b0d797078a68e96 : SUCCESS in 4m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b1d4d3d47d1943ef982893aad39a1099 : FAILURE in 8m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ad77ea7cc7d740898178ddb3c0d3093b : FAILURE in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c4d24d6d5d8c72f839a8493211ad1f13199342db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:46:37.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dde7e4be6abb48b88bbcd0217114c3f7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/97fd973597d4428b861e20182f412b03 : FAILURE in 4m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2084ddc4f61b44e4aec730b63880db0f : FAILURE in 8m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f4cdacfe18d452282e5ab10fdcb4673 : FAILURE in 3m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e69994b681ac4a1589d280cb8efd829a : FAILURE in 4m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d83c415641b6448b89e3d9b6a50c52b9 : FAILURE in 9m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c297804e5a54f7290eb45bd42883859 : FAILURE in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c43ad5583e49f4b13fbd6af00f7c58 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c20bcb0e78bf4ba498f03cd242112116 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/022c8ea77fc14029b6ee3a2e76996d38 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22f7a8acc4ab4ad0948b8a1a365ed56b : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b67ad61df0c4d0e858783c38094df27 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64ef8a381cc04cdcace25b3d55b5d3dd : FAILURE in 6m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd03bf7de2d64561a467db8d017cf243 : FAILURE in 7m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/80f5fc4f21b2478d94cac57e81b151bc : FAILURE in 10m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7ecdb27e8ab6430381f012d391ed1837 : FAILURE in 12m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fd6388bdfbb5424ca3255a48666175bb : FAILURE in 4m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f8a7ba6efaf64f2ab4c17d52f1d208b4 : FAILURE in 3m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/202959b6801c41aca3866e259ba168a3 : FAILURE in 7m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7416e2ed7d984d548ef8f37316971f8c : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":19},{"id":"b12d8e554990f1286b97dba03b873781636ead29","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-25 19:18:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"09172247ca71515fcbc55cf3d5cfb30d3661a14a","revisions":{"28045e05c11ac0c889a6a15872dc7964918a2ba8":{"kind":"REWORK","_number":1,"created":"2024-09-16 10:57:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/1"}}},"commit":{"parents":[{"commit":"8e559e16ee3ea469318ea435cd69a91b91135d69","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e559e16ee3ea469318ea435cd69a91b91135d69"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28045e05c11ac0c889a6a15872dc7964918a2ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28045e05c11ac0c889a6a15872dc7964918a2ba8"}]},"branch":"refs/heads/master"},"e56e405427f2c93d38561f9d891c5a6cbf8f20f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-16 11:30:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/2"}}},"commit":{"parents":[{"commit":"b998011c19fa8678415cd42031c6b215842bf9ae","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b998011c19fa8678415cd42031c6b215842bf9ae"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 11:26:12.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e56e405427f2c93d38561f9d891c5a6cbf8f20f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e56e405427f2c93d38561f9d891c5a6cbf8f20f4"}]},"branch":"refs/heads/master"},"88d25a88b2c70434c258fd2c0faef6322bbfc73d":{"kind":"REWORK","_number":3,"created":"2024-09-16 11:58:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/3"}}},"commit":{"parents":[{"commit":"b998011c19fa8678415cd42031c6b215842bf9ae","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b998011c19fa8678415cd42031c6b215842bf9ae"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 11:58:36.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d25a88b2c70434c258fd2c0faef6322bbfc73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d25a88b2c70434c258fd2c0faef6322bbfc73d"}]},"branch":"refs/heads/master"},"6178b62b7c82dd5db6d60a8fb0e25f0f33ad68fc":{"kind":"REWORK","_number":4,"created":"2024-09-16 12:06:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/4"}}},"commit":{"parents":[{"commit":"b998011c19fa8678415cd42031c6b215842bf9ae","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b998011c19fa8678415cd42031c6b215842bf9ae"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 12:06:20.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6178b62b7c82dd5db6d60a8fb0e25f0f33ad68fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6178b62b7c82dd5db6d60a8fb0e25f0f33ad68fc"}]},"branch":"refs/heads/master"},"538ec286728e70e4827bcfe15b5b7827a01cc3c2":{"kind":"REWORK","_number":5,"created":"2024-09-16 12:19:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/5"}}},"commit":{"parents":[{"commit":"b998011c19fa8678415cd42031c6b215842bf9ae","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b998011c19fa8678415cd42031c6b215842bf9ae"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 12:19:12.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/538ec286728e70e4827bcfe15b5b7827a01cc3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/538ec286728e70e4827bcfe15b5b7827a01cc3c2"}]},"branch":"refs/heads/master"},"279ecd8fef5c249ca57373b74e7d3155dd554fe1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-16 13:06:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/6"}}},"commit":{"parents":[{"commit":"a1c6e63dd81e29e9ab1de159c816cf3ea50947f7","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1c6e63dd81e29e9ab1de159c816cf3ea50947f7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 13:05:24.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/279ecd8fef5c249ca57373b74e7d3155dd554fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/279ecd8fef5c249ca57373b74e7d3155dd554fe1"}]},"branch":"refs/heads/master"},"269bc43f1e1ce7c8f1ac6202923465d43bcde23f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-09-16 13:28:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/7"}}},"commit":{"parents":[{"commit":"03025884272809ad391b0c374e2ed096ab270813","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03025884272809ad391b0c374e2ed096ab270813"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 13:25:23.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/269bc43f1e1ce7c8f1ac6202923465d43bcde23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/269bc43f1e1ce7c8f1ac6202923465d43bcde23f"}]},"branch":"refs/heads/master"},"c132259dc52ff730dbb2c7bc1c3c76bdcb19f9e4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-16 16:36:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/8"}}},"commit":{"parents":[{"commit":"c2665b902a01ed16f64281633f0cb23615273421","subject":"Add macro for patch files to all images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2665b902a01ed16f64281633f0cb23615273421"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 16:23:22.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c132259dc52ff730dbb2c7bc1c3c76bdcb19f9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c132259dc52ff730dbb2c7bc1c3c76bdcb19f9e4"}]},"branch":"refs/heads/master"},"064e64d6a11228169f0019c70e06a8710e375fa6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-09-16 17:33:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/9"}}},"commit":{"parents":[{"commit":"790f4be6a39ba4df2b7bf69a83481233f4302bc9","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790f4be6a39ba4df2b7bf69a83481233f4302bc9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 17:32:53.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/064e64d6a11228169f0019c70e06a8710e375fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/064e64d6a11228169f0019c70e06a8710e375fa6"}]},"branch":"refs/heads/master"},"4e0a064cbe0d5bb6ab4ba396da720132fec98ee9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-23 13:59:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/10"}}},"commit":{"parents":[{"commit":"4d3dfce50ad285ae1c46385add896bfad840c15b","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d3dfce50ad285ae1c46385add896bfad840c15b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-23 13:58:45.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0a064cbe0d5bb6ab4ba396da720132fec98ee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0a064cbe0d5bb6ab4ba396da720132fec98ee9"}]},"branch":"refs/heads/master"},"8eb736508356bbe9da868119785323fee193c6e4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-09-30 21:10:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/11"}}},"commit":{"parents":[{"commit":"a53ae18993e829ffa2d36aaa81583dba812187dc","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53ae18993e829ffa2d36aaa81583dba812187dc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-30 20:36:08.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb736508356bbe9da868119785323fee193c6e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb736508356bbe9da868119785323fee193c6e4"}]},"branch":"refs/heads/master"},"8af3adec7d6ed8248431bb90e8a5307f2893a8a1":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-09-30 21:14:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/12"}}},"commit":{"parents":[{"commit":"9ee918d9095fa44a31e4834574529ca8520d4239","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ee918d9095fa44a31e4834574529ca8520d4239"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-30 21:14:12.000000000","tz":120},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af3adec7d6ed8248431bb90e8a5307f2893a8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af3adec7d6ed8248431bb90e8a5307f2893a8a1"}]},"branch":"refs/heads/master"},"0b841dde0928a9f059e79cc026b155e0e9ec1e8f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-11-04 14:06:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/13"}}},"commit":{"parents":[{"commit":"6069f610b6ee38cea09d4bd333370080c335e23f","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6069f610b6ee38cea09d4bd333370080c335e23f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-04 14:02:31.000000000","tz":60},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b841dde0928a9f059e79cc026b155e0e9ec1e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b841dde0928a9f059e79cc026b155e0e9ec1e8f"}]},"branch":"refs/heads/master"},"5c33427fa75cfff97ee00a6277fa3b6213d28160":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-11-07 07:50:12.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/14"}}},"commit":{"parents":[{"commit":"a7419d171a2e688a3427f258efa1bac5ed432888","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7419d171a2e688a3427f258efa1bac5ed432888"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-07 07:49:50.000000000","tz":60},"subject":"[CI] Test kolla patch","message":"[CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c33427fa75cfff97ee00a6277fa3b6213d28160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c33427fa75cfff97ee00a6277fa3b6213d28160"}]},"branch":"refs/heads/master"},"94de8afe9ef2e984632b0623d90bd578fd6afbf8":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-11-07 07:52:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/15"}}},"commit":{"parents":[{"commit":"a7419d171a2e688a3427f258efa1bac5ed432888","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7419d171a2e688a3427f258efa1bac5ed432888"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-07 07:50:16.000000000","tz":60},"subject":"[DNM] [CI] Test kolla patch","message":"[DNM] [CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94de8afe9ef2e984632b0623d90bd578fd6afbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94de8afe9ef2e984632b0623d90bd578fd6afbf8"}]},"branch":"refs/heads/master"},"0c2ff0b39f37059d09273dc50bea23e9469c9801":{"kind":"REWORK","_number":16,"created":"2024-11-07 09:37:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/16"}}},"commit":{"parents":[{"commit":"a7419d171a2e688a3427f258efa1bac5ed432888","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7419d171a2e688a3427f258efa1bac5ed432888"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-07 09:35:10.000000000","tz":60},"subject":"[DNM] [CI] Test kolla patch","message":"[DNM] [CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c2ff0b39f37059d09273dc50bea23e9469c9801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c2ff0b39f37059d09273dc50bea23e9469c9801"}]},"branch":"refs/heads/master"},"127b9852bf8b2267b3ed695f26df87cd142fa3f0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-11-07 09:53:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/17"}}},"commit":{"parents":[{"commit":"99676fbcf86243085d95f3a3f410faa007ed2196","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99676fbcf86243085d95f3a3f410faa007ed2196"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-07 09:52:58.000000000","tz":60},"subject":"[DNM] [CI] Test kolla patch","message":"[DNM] [CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127b9852bf8b2267b3ed695f26df87cd142fa3f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127b9852bf8b2267b3ed695f26df87cd142fa3f0"}]},"branch":"refs/heads/master"},"1b9eb0557164cdde4a60183606964686192c372d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-11-14 08:52:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/18"}}},"commit":{"parents":[{"commit":"83b5c515f8c3a65b7a8133364b68fb19cddd5e9b","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83b5c515f8c3a65b7a8133364b68fb19cddd5e9b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-14 08:51:23.000000000","tz":60},"subject":"[DNM] [CI] Test kolla patch","message":"[DNM] [CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b9eb0557164cdde4a60183606964686192c372d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b9eb0557164cdde4a60183606964686192c372d"}]},"branch":"refs/heads/master"},"09172247ca71515fcbc55cf3d5cfb30d3661a14a":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-11-14 09:05:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/46/929446/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/929446/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/929446/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/929446/19"}}},"commit":{"parents":[{"commit":"d59da651f6845776d2ecbf33d53acdb35ee9801e","subject":"Add mechanism for patching files in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d59da651f6845776d2ecbf33d53acdb35ee9801e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-16 10:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-14 09:02:16.000000000","tz":60},"subject":"[DNM] [CI] Test kolla patch","message":"[DNM] [CI] Test kolla patch\n\nChange-Id: I7510b68f4096134bc9d106fe14bfae2bd7b8b012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09172247ca71515fcbc55cf3d5cfb30d3661a14a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09172247ca71515fcbc55cf3d5cfb30d3661a14a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
