)]}'
{"id":"openstack%2Fkolla~929457","triplet_id":"openstack%2Fkolla~master~I9359df23a41a600dd46dbc1812a54be57fd33f71","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2025-02-02 16:28:31.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-02-02 16:28:31.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-02-02 13:23:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2025-02-02 16:28:31.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-02-02 16:28:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9359df23a41a600dd46dbc1812a54be57fd33f71","subject":"Prometheus update all exporters to the latest versions","status":"MERGED","created":"2024-09-16 11:54:18.000000000","updated":"2025-02-02 16:29:27.000000000","submitted":"2025-02-02 16:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929457","meta_rev_id":"1a833eca1de8d33d75f020d4616b3ef4c8bf5db6","_number":929457,"virtual_id_number":929457,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-02 16:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2025-01-30 16:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-02-02 13:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-30 15:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"value":2,"date":"2025-01-31 15:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"value":1,"date":"2025-01-31 15:23:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"value":-1,"date":"2025-01-31 15:23:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 13:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-16 19:26:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-17 18:03:27.000000000","updated_by":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"reviewer":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"state":"CC"},{"updated":"2024-11-19 20:16:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-01-30 12:41:58.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-01-30 13:24:20.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2025-01-30 13:35:53.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-01-31 15:23:00.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"15337d5739d1646f949ec2173c460a7eb11d471d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-09-16 11:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5de6d805faeeebf8032d65d0d4dfd3adea074b6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 13:21:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d6272cd547e4e62bd7858b5b9ede16d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f61702d408b7412e8fe2de51bc4a1a95 : SUCCESS in 27m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cffce07c99a74831a2c9cf6d2c789fe8 : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3bdb67b0e6d44e8b801e5fbb96062c6e : SUCCESS in 20m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a7229ae189194450a60d4bf06beb90a1 : SUCCESS in 28m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87718cfc008b41bbb743ca9ba4d75783 : SUCCESS in 1h 00m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/019d61119ce44ab085e73f7fdfcd6a9f : SUCCESS in 1h 23m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8226187d20c0423b826e808ce9914634 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/000bbf88d6614073a47866b419bfd032 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/301d712db34d423faed6c26d1f43acf6 : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5362e6360124ac98a2c2837539aa24c : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a141514e4f4d459b694fa091aba205 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a204516f5f4d898ab3dbc37549a9d6 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/11d44854bfcc43318954809289d58693 : SUCCESS in 28m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6245a9a2c5d64881a6570684f182c8db : SUCCESS in 37m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b5ba3adcc470464b83b42608f82455c0 : SUCCESS in 1h 09m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f39be9ee89c94feeae21cb9a09cfcc34 : SUCCESS in 1h 22m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f28ea833058044dbbe18cbfc28e15b5b : SUCCESS in 23m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a3856754181543abaa99f4c0e4b3c727 : SUCCESS in 48m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cb9c2400da14f95b1c17d79cdd96399 : SUCCESS in 57m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/82be8946d29a477a93e607f8f2c4ade5 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"13eb10928dd16451f9599ba1673d03c517282c01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 15:09:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0343939ff664d1c83de06ed592a789a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4ed77055f1447a1ac259ae5b47e65a6 : TIMED_OUT in 3h 03m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4ba82bc4c83d4b0c80e40d72766e5d66 : SUCCESS in 2h 45m 17s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12ffe9b76a0f4a70994f05e0379be883 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/806b8f5e31d74201adec6be7a764705f : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b77b79d394c17b5153b95519e785164ce23821e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-19 19:40:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4159e42413f68572a5a1b386eb0871768793baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-19 20:06:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2a1818c867c7937ccd414521651156eae2bd3942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-19 20:06:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4167f2cdcd96c748ace0aaa218a5e2862c2cb7c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-19 20:16:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ee124c596adee88da85ad0d27490f279ee3d6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-19 20:46:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"36b42ce8e420437cf18d51cb7d95f5f8477950ff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-19 21:08:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bd0ac0cc357dcaef369de16cf9e156690070ea4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 22:34:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c6bee6239204012bc57b33ebab4a3d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c652c1d2e88046bf93cc26ea35484a27 : SUCCESS in 29m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/22a1ce8680e347f7bc1cf3299accd6ae : SUCCESS in 1h 01m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb9f0a8ea0cc4f30adec7990f7a41100 : SUCCESS in 20m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f745c9df5ea4464f9d603417865cfebc : SUCCESS in 27m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce50a4a0d72c4fd88936403ac26601f7 : FAILURE in 11m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/801d14a2ca5b4612811d545b3b017c97 : FAILURE in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9c58e466d64fe2b127907e211b78eb : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4312385ae9bd4d229a425cab89a11bad : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27aee30532de498a84d4926877f484c5 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9749813e4dee41279b37cf1af612421e : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afa66ccaa5c464cb97049a7ac07fd4c : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe50f2f4a474814a325eeecee8ebd3c : SUCCESS in 3m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4a5f719fbd5548a59bb584bf6ff91630 : SUCCESS in 35m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f691556109b4c2da9346b97d8be0ee2 : SUCCESS in 24m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c16e71c859bf4515abe87a15e1c1b72c : FAILURE in 11m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5ecad0959253477f9f789727d23dd1b0 : SUCCESS in 1h 23m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba82698325bd4198805bf016dd17f97c : FAILURE in 5m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1aadf0537a2e42b0931c8cc422612b4b : FAILURE in 13m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e15dd68d4d34e6db7bb560f648cc6b1 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e43db2e05436433f83c9e426b9016bab : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b8fd51201ec5552482869b9429ac2a26e38a5a8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:17:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0495e3210fe44ca1a6cebbe792cdae8f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bb3dfae281b74c56912f9eb8545822f4 : SUCCESS in 1h 46m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fc85c40904f94851966a382e80379f49 : FAILURE in 3m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ce0fa363bde42bc88d62af5a19544a8 : SUCCESS in 2h 16m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4ff16a8d78c4959a69d63bd338fb5c1 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c08fea6f69c434eba1da6bba07c7256 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8bf4c43a74ef4d848d67b63c1cc6a674 : SUCCESS in 1h 32m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a30ee43c2f234d2794c1dfe9d7e07de5 : FAILURE in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"249c4a746425b0e5303edac776a5dbbacbec09df","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-20 08:30:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42c390dde6b71fa5b48b4c5fb56927414a15b5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:10:43.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e5c2872958748b7923b7ebea305eccc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/431b95b0eeb348fb817ad48165046895 : SUCCESS in 23m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/81353406357c4f93b1fe5de059d0104c : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/61008bed6f1e4e69aca4527d0c89ccdf : SUCCESS in 20m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a7c40ac5427d46d3b3554a21f45bbddd : SUCCESS in 28m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/221dc7fd6e39496d98b72353dc94adef : FAILURE in 11m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/13428b023c7b43edb83af066f3d93996 : SUCCESS in 1h 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3354597a899243d4a10b54c500f77705 : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf4b86b9a2d84cefa37b3c6900b53df8 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc3955caf7e649be9c4d54e722b72414 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53fd77649ba4d27a76a5369bdaa79e2 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f071ccfeae5c4249a0ee005a214efc68 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b5e1a648d7b41fdacbc0e62bdb50a07 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1ce13e13a85d4c8dbf832f8f9bf400de : SUCCESS in 20m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e32eecf6279849f49a0425cf6ec370c8 : SUCCESS in 27m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0a4b7b310b664792b34be230a51a1c9d : SUCCESS in 1h 03m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cc47b1fd6a1b486dae0256af52e99865 : SUCCESS in 1h 33m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6dfc5760b7504161a1cf9a896080e2f1 : FAILURE in 6m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4f8e883b3f8c4fd5bafb8792f244e1c7 : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5066de5a4364452a81ce018c351915ca : FAILURE in 9m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/544f4c4e75a44c9c89f19fb0745239ac : FAILURE in 1h 26m 33s","accounts_in_message":[],"_revision_number":5},{"id":"14dc27026a04ad1bde62ece3631a1c6cebe27945","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:09:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c3d9a5835f4b3da020b3ddb7a40abb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/46a2bca91b8e4ed08b8412bc23885d6b : SUCCESS in 2h 00m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e988ef3c4d24b0cbb0344a01278e9cb : SUCCESS in 2h 59m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80d074dd236b4c0889e3ee16a2bdc201 : SUCCESS in 2h 03m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed51c4e778c14699a57c52e1cab0c962 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d2b9e79820f443087cd63bbc87de9ea : SUCCESS in 4m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a8d66d074d0c430bad9a6dd834a2cfb6 : FAILURE in 1h 44m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c111789626f7432082ad50f40648ed51 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7cc4937fd62e56b9cc9ab3cbf058977dbd9ed919","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-21 08:00:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b6cdae26302f41f2db2074c815a4d47a111a6048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 09:04:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e242f0fed5004a269add3fc4af3ff6b1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ce3656202252433c81dc5f59c26eac7c : SUCCESS in 18m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/03207fc59dcf424987e70ef5711f8569 : FAILURE in 8m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/38ced5e1271547cb9bc1656325c6549b : FAILURE in 3m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ec484f6d6a34430a92e6c1162c5089d6 : SUCCESS in 30m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b3a9fc17225c4dbebda8953ff50ec61a : FAILURE in 7m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6bcd2f3efb3d4ba69a181cc4abb65e83 : FAILURE in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ba9fef50e244b89e2add5f9bf11a71 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b16a2eb668f4d518945760898c09ad3 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f639e07594fc4b4abf621b737357e3c7 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9a065cc7fbe4a139c9691a083cdc2d3 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cea88331a44fc3badf4e0b7971dfdc : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/968a9ce417844c33a3cece9eeb524b94 : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8c176582b5e445580bee96b5e1524d1 : SUCCESS in 38m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2eb6e09c139b49a4be84c17e90023ac0 : SUCCESS in 35m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ccc35bddd4774febb1de4eeaff2b654c : SUCCESS in 1h 03m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/622c011d8cf74a3ca95003a6c0816e20 : FAILURE in 8m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f78484fe9b374578b30fe0d8ac9f8cd3 : SUCCESS in 35m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f92192aa4c074965982170af76c94f1a : SUCCESS in 40m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4cb9512b142f4d2fb3ab3df1121772d9 : SUCCESS in 1h 03m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bc5a9bd379ef4d1493f12bdc7633b976 : FAILURE in 7m 11s","accounts_in_message":[],"_revision_number":5},{"id":"09df7762055cbcb6788757a5bd74324fc9ceffb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:03:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ecc828a470412a9b3a235883abb9b8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/744c2a1f2808413fb4541ec312388151 : TIMED_OUT in 3h 01m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/818afb27fee645cdbd7490b05991cca7 : SUCCESS in 1h 43m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e0415038c7b14a45abaf10d1bacbed35 : SUCCESS in 1h 57m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de9ff8b370c542a9ab61c3f80dac3fc3 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c67996a5046452d9ee9fde65a221153 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a892dbcbef4d48858d7096d78321184f : SUCCESS in 1h 47m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d878682e375443eba2458ed892f397a : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9f1d45dd34dc22b8ff5ecc006985f73d1d1d2ad2","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2024-11-30 13:55:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"043e99c6ece992c04d245d5f415c2b8792afb488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 15:42:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b832751951224b868d89910f2d713a78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/06db4fa9cea74368b8a2a186a368b515 : SUCCESS in 22m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c2cebd32a0e645dfa614f0b3f87abfbd : SUCCESS in 1h 06m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83d5a7a6e3b4441a880555d10ca1c476 : SUCCESS in 22m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/796b8776bc29477ca07ac2fd3d10d310 : SUCCESS in 29m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f4088dc398f14f99901d317dd3358b9c : SUCCESS in 1h 09m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f58f9574e63d4c62a270da9b5ff72d4f : SUCCESS in 1h 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45f97b53f39479aae874680d6123b8c : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d76c10995d14daca5216bf2c74459ef : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af31735f7156413e8931c211eef2613d : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b8528974d5f45698327ab9b707f2dfd : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca485af59dca456a9325f24d890f1856 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63266f90962f4bcf9fe526ea295bf76e : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/110e2da43d26440cb23ef08a5f0f905e : SUCCESS in 28m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/85503c18371d43d9b5852b7c83c42474 : SUCCESS in 34m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5df06e953e54493db5f8a3b16d39793e : SUCCESS in 1h 18m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8ba9a24a96324759aa07683b37b31ae8 : SUCCESS in 1h 29m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb1ad9170f6e40a7958b0562eadc45c7 : SUCCESS in 25m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ad3382816f3496db22fec000d34e7e9 : SUCCESS in 29m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/44b0d730191b4e949216a69044599260 : SUCCESS in 58m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/41737913c52d43efa4100ca25c60e045 : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":5},{"id":"436d50e46c9ca6ab29c1ddbfd541f98f2ae9f162","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 19:44:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da42aeb0102545be81901d142f81f016\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/778ddee5ac6442daaff8feff92405812 : TIMED_OUT in 3h 01m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1f2eca9217d4d2c8e1b098ce25ca03b : SUCCESS in 2h 36m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d4d790e631c40a59cefe8744a154e43 : TIMED_OUT in 3h 07m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05265ba1cf1843d89632951b8d7245af : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/328959ca2dc04b7c9ea14436a7f85fb5 : SUCCESS in 5m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0cf1f9296870439ab983d837ea961617 : TIMED_OUT in 3h 03m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/688be2f8b46d48dbbfe1f017d3c59dea : TIMED_OUT in 3h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b74cdb4efebf18ee6af58e8369409d8433b7f77c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 12:41:58.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"69cb425014028643f0933f677440b738d526105b","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-01-30 13:24:20.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"faf80a1bc7ad82ad917e887013465bad68dce0e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-01-30 13:24:43.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_37306\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"_revision_number":6},{"id":"27d2d1e89b86022254f4272922355609cd0de9df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-01-30 13:29:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"100f7e21f46053d28af4386c924d2efb4b8bf237","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-01-30 13:30:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d87a64152969b987d5afce89f146b68d4937b020","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-01-30 13:35:53.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"2db49437222d0e974d8ade8dc15cb706642b5a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 13:55:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0c3de057c1ff92621efa55d2a68eb8dc59a1b78a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 13:56:43.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"65a1b3af2775928ecc5a9a5ec1350bfa87bb36af","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 14:00:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5a72edebf4f1e777a96fd9eea16c63884600b68a","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 14:03:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c0b38564601f3d71e736099325a8c3b781c5211d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 14:18:30.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"e5f42b851432e08c4b817b1c70968594681fb5e8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 14:20:34.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f0c1c34964fb2c1ba80f6e59ff9a36d30073c30b","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 14:24:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e24d0474dc8efed39bef0a339bdfc602fc98174e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 14:27:10.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6aa836ca306725ba051dd5a2f888e255b73dcfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 14:28:00.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"83395850263b41998aa2c0b9f5d99c3f93a7ff9e","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 15:01:18.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"18d6aa92408c97e869623d97bfa5a8397251f942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:11:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a64de7fa7559440eba6a6349ecefd48b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f0bf3250369d49b79075246edef6bf6f : FAILURE in 28m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8bfd9784c842483e958942eb990e3203 : SUCCESS in 1h 12m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1243b90f680e4524bfc298dbaaf490ab : FAILURE in 21m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2113a071dfbd4f3d8e2d48aa2a480cc1 : FAILURE in 23m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ffdb6634ee1f4b63a3c1b104d62580d6 : SUCCESS in 52m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b892d2dd05a14da0adef198a30465f29 : SUCCESS in 1h 36m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fa948c0ead43d6b2fd11e5626b3d9b : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a92adb02592e4ff0be5750d6f898b827 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb25682fa90a4c778e9ebd0a61f6e233 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a78bb7882ef458f92e332f17f0bfcab : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea26f5909a945ad9910499dcc1786b0 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da93fa6ebf874872bef357c2c49a72e6 : SUCCESS in 2m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5c56007b99b746ed92b1a61c70264ff3 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27beda00878a41e3bb4d34bbecd8aad9 : FAILURE in 31m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/733f61b1355d4f37840f7e1607c00958 : SUCCESS in 34m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76423360c2d542189ea5588fa497dc36 : SUCCESS in 1h 00m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d07fc2f577b54abab2b99cb7d6b90232 : FAILURE in 18m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20571e5c363f454d8d68f01539b07903 : FAILURE in 22m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/927f10681a9b45e98df266f95ec6d137 : FAILURE in 17m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e489509bb47f4cb4ae409570a94fe6aa : SUCCESS in 50m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/81f6ba41ec724e7188ba77f70ea85f92 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":10},{"id":"691277b6f44b29be960a00ffbbf799c38fc7c105","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-30 16:16:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"72c4445e4eff415c660d944f8e9783f06cb3b117","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-30 16:51:59.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"84572299c0c8b56cfbc774dce9bf5d120ab7462f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 17:36:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd511dfc89114996a8efca308b025218\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fcad25de1a7e48f3ba9597f49e4e33d9 : FAILURE in 2h 37m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/157833aa07da4cdfa518d9cb535378ab : TIMED_OUT in 3h 01m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb3fc4986e0142c9b6e59ee1cde1d139 : SUCCESS in 2h 29m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16f6eba6036940ab96e1096c71e31b69 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6943bf30d2a4f1abd258978d4122d60 : SUCCESS in 5m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/44e9b1382bf74a198d5fd63ca13b8abc : TIMED_OUT in 3h 02m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e91905f025164165bdd075645a2e0548 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"aea23dca73b10db789d7e61d6ca1976b4d6f4b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 19:32:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1cc9bc6a34147fb83065434b6010396\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b0a139b2c2fc44689cc500c178ec8d61 : SUCCESS in 25m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/908c09ad7e9549ce814dd2d68d2d31ad : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3e711347c3e54fcea15b51e7a4d970ff : SUCCESS in 18m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9699920efd53492a9d656ddabee459a3 : SUCCESS in 32m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fb2e5f1642364f9982ccce5a92d58517 : SUCCESS in 1h 07m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f0da719631564190b238bdc7ce890f0e : SUCCESS in 1h 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5c7ffaf92140318945a166a053b6a2 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29aa7c09dd53485b9b91d611fc7e4939 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b3b558906a3415c963faef5def38cb1 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8146fd00e83d45d5a384860f95639075 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b10edf275d48fb80b48f508a444064 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5560f25d1b9a4c319c1871094786e49f : SUCCESS in 3m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/770b9a70e14e44ebbc5220af9426b2eb : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/45ceb48d6d91410b8a46052ab26bc996 : SUCCESS in 23m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e12e36136e2f4e8e96f8f26e20fafe1d : SUCCESS in 46m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff3ee1c6331e47a482ed4ce3e9258deb : SUCCESS in 1h 04m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e5ec9465086e454581b572e9ced0d772 : SUCCESS in 1h 35m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67166973ead440b3923a1d60a3618222 : SUCCESS in 23m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c65e0ad49ff14f258d752bbf90a2469c : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/483ce6be5a934d3fa2974bb5e45ac6b1 : SUCCESS in 1h 09m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/79f027c2948e4f358b1a8f1e9febe355 : SUCCESS in 1h 33m 00s","accounts_in_message":[],"_revision_number":10},{"id":"cf9685aba2c4838c2bf8d3e95f4463b8bf483381","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 15:23:00.000000000","message":"Patch Set 10: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a5688b1d051b7b28211bc23e6ec2f7eba055787a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:23:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"61f9e22739ac60445fafb6d0896c48f8bd18238d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:05:13.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bc3b74d438744308f6c633d4e152f79\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/82b9d5ff5ade4f89b1b8c683e1f0a7ff : FAILURE in 13m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b119c755c43143c28fe4abed33ec4533 : FAILURE in 13m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4f230d26a80741bebf2898c0c4f868ac : FAILURE in 16m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/95421641f79843bfad7ca8918e743be1 : FAILURE in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8832b6ffe5944a78ab4369d7a88ac81 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4300b406766c443bb4b43a66b6bf292a : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8d15714848d47c59dd576a85e108bee : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd7389b4bf24f5b80a97125e98541e1 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f6aff4b0bb348d88d6688942b333136 : SUCCESS in 3m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/155dbbc8548943c7a69b68e5de7fa7be : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3744862bfb3f402d931c44b95999325d : FAILURE in 16m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e8a7676e3b6349c691df2b6e16b25bcd : FAILURE in 18m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0847590c5c6f48db9d8ed88f7e9c31c5 : FAILURE in 13m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/808405b139d3464c8c2980e10413588b : FAILURE in 14m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb78c3e6e9744148bb3bb6f9d515445b : FAILURE in 15m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d9a65ac417f54bcd86e2c02bad968b76 : FAILURE in 18m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/25ea1eec76c94591a2caf9f82197b4fc : FAILURE in 11m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5ace978fe1184f40bb4f52a7c4064230 : FAILURE in 14m 05s","accounts_in_message":[],"_revision_number":10},{"id":"385c4ad0a480e9df82e99ca1d9332a3536561fcc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-02 13:23:25.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"226a0b6e80fae97af31d218e42dd6550db5a6dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:48:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4d546dc0ce4ee2b6b7b82337185102\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a06be86679b748b593025b0144f274b7 : SUCCESS in 21m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/43bdbe632a4949adb945ab3791e15900 : SUCCESS in 37m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b6cfb6ee3c724d02b7275f8ba5e252d1 : SUCCESS in 20m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bfccfd9be282461fbd2b02fbf5cfe045 : SUCCESS in 30m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2da445f416734b45bdd68780762c8616 : SUCCESS in 55m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25c51132687346dea78246737db59e8d : SUCCESS in 55m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1787844d9d9c469682d700450e69ff38 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d22983a7c2410c9ffb2723c9118b0c : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f0dcbd8bf6245fa99191b60b6144fa4 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafa8b2bc90f477fb2233e1f918d398a : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef11e02851554afcb3aafcb9a37bbd4a : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e58de6775c0467a963e0d95b88adf8a : SUCCESS in 3m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bf7603b11cf84847812233a8c0f5c54b : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8b28486669254ed7b4b4c98ea421627e : SUCCESS in 28m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/21ca289d1ca048879df0499525954410 : SUCCESS in 31m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c8ec75a5d66340449d985e880aa42d59 : SUCCESS in 1h 11m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3fb50e97e0684c469d9b43c56ea14ebe : SUCCESS in 1h 12m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f0ff48b8964f45ffb57db45b878b7afe : SUCCESS in 18m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9310f2ea806f41b6b67beaae8053ab64 : SUCCESS in 17m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e96e62da0c764dd582a8b814152ceef8 : SUCCESS in 54m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/124fd32404b245c0b6405ed8cd38a041 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":10},{"id":"08ed6d6a4d0b497ba3dc1e8f951a2381ff34d6c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:48:54.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"50b581e005c88a037911576c4d51bf7a414f9b42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:25:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/badde13a6361402fb4615e28ab04a4b9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/57f3425cca214fa9ac69edba7d8cc2af : FAILURE in 2h 12m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e88f9de1203e4406bdd656806673b93e : FAILURE in 2h 03m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f120b96f92204d18ad5f00edf5fdcd20 : SUCCESS in 1h 22m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7009053d655c48d69657f23f164bbe47 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e869b83a72644edfb58e402c5e40ac3e : SUCCESS in 4m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/42e8589e58984c72a04d79636fb6ecb0 : FAILURE in 1h 55m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f3803bd6967b4309a291269b81b8a1c8 : TIMED_OUT in 3h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"19c7114ccaf0e30d0f8a3923d3beeab35a7cddd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:28:31.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca369689bd04a5f9d5eb2bbdc3875f8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f9dfbc0edfb8419abda017564ddb6166 : SUCCESS in 20m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f1965543d8a4f86a5a421f4e85c202c : SUCCESS in 31m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/904144b3ced742818f5beab9698ec27c : SUCCESS in 1h 03m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fe1f5583f8f2419190fce4f86ac0233f : SUCCESS in 1h 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf65fa639c44f11b242e3669323939e : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0d6e540ae734554b5e5d347fd788b2b : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8aa419fae7b4af793c5c61d25439374 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1fbb7ebef741398534677e003f40f1 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0931a42bf1de43b8bf9f88dca5c40c0a : SUCCESS in 2m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4386808e354e49f48f0c2317d42bb7b0 : SUCCESS in 2m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/53b0276b2c4340a1b46fc50f8fd827b8 : SUCCESS in 16m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2f1e9f212a4545efa52d85502436d606 : SUCCESS in 24m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2849a954b9bd442c9523e5ff110ae530 : SUCCESS in 36m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/19a0b53abe794c46b697a7f1b99e364e : SUCCESS in 1h 11m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0cb9ab4d82154942bf828f9b6eca76c7 : SUCCESS in 17m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a6a67477fe7d4308a1de4c92f7d267a8 : SUCCESS in 29m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6b6377bc7dc540d7b2194a49c7d15ab4 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2dc56276b2b744aa9333939faed13567 : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":10},{"id":"8c8ec91104fcb382891deb2727fbabc43a7c4873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:28:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1a833eca1de8d33d75f020d4616b3ef4c8bf5db6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:29:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af2ccb2fa25640849f46b76acc146e7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709898ae515741e4b96c7a45026f4c4d : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/751ac1a9b8c84dd2aa395e34545c0637 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"0f5c679dc93d25d0ab8818924adfd8037a18b289","revisions":{"cdcfb494b4f28caad07f838e6b445b26b593d256":{"kind":"REWORK","_number":1,"created":"2024-09-16 11:54:18.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/1"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-09-16 11:54:06.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-09-16 11:54:06.000000000","tz":120},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdcfb494b4f28caad07f838e6b445b26b593d256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdcfb494b4f28caad07f838e6b445b26b593d256"}]},"branch":"refs/heads/master"},"b251aec6027c82415cce9a9f1e08fe09eb331b85":{"kind":"REWORK","_number":2,"created":"2024-11-19 19:40:36.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/2"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b251aec6027c82415cce9a9f1e08fe09eb331b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b251aec6027c82415cce9a9f1e08fe09eb331b85"}]},"branch":"refs/heads/master"},"3042ddd99772ec351d74af569dd6f6ceea59c035":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-19 20:06:39.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/3"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr","email":"vurmil@gmail.com","date":"2024-11-19 20:06:39.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus-v2-server: 2.50.1 -\u003e 3.0.0\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3042ddd99772ec351d74af569dd6f6ceea59c035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3042ddd99772ec351d74af569dd6f6ceea59c035"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fee519a55f9dda9f137374d8e181d35b7061a7b9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-11-19 20:06:56.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/4"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr","email":"vurmil@gmail.com","date":"2024-11-19 20:06:56.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fee519a55f9dda9f137374d8e181d35b7061a7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fee519a55f9dda9f137374d8e181d35b7061a7b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1c21b07631c3efe90068ab7c784255b2f3ff89fb":{"kind":"REWORK","_number":5,"created":"2024-11-19 20:46:34.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/5"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 20:46:09.000000000","tz":60},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c21b07631c3efe90068ab7c784255b2f3ff89fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c21b07631c3efe90068ab7c784255b2f3ff89fb"}]},"branch":"refs/heads/master"},"40b921df93accaa925e477a55be21edd5b6a3833":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-30 13:24:43.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/57/929457/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/6"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr","email":"vurmil@gmail.com","date":"2025-01-30 13:24:43.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40b921df93accaa925e477a55be21edd5b6a3833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40b921df93accaa925e477a55be21edd5b6a3833"}]},"branch":"refs/heads/master","description":"Rebase"},"b35e076bae3e80f97647333bfadccb826faba359":{"kind":"REWORK","_number":7,"created":"2025-01-30 13:29:48.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/57/929457/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/7"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-01-30 13:28:57.000000000","tz":60},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b35e076bae3e80f97647333bfadccb826faba359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b35e076bae3e80f97647333bfadccb826faba359"}]},"branch":"refs/heads/master"},"c391e409f29a59b4b1050241ef4c7dc0821a432f":{"kind":"REWORK","_number":8,"created":"2025-01-30 13:55:08.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/57/929457/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/8"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-01-30 13:54:47.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-alertmanager: 0.27.0 -\u003e 0.28.0\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c391e409f29a59b4b1050241ef4c7dc0821a432f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c391e409f29a59b4b1050241ef4c7dc0821a432f"}]},"branch":"refs/heads/master"},"05f2f8057f06dd54d6ecbf33c2b66aae7e65cf29":{"kind":"REWORK","_number":9,"created":"2025-01-30 14:18:30.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/57/929457/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/9"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-01-30 14:17:51.000000000","tz":60},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-alertmanager: 0.27.0 -\u003e 0.28.0\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-cadvisor: 0.49.1 -\u003e 0.49.2\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05f2f8057f06dd54d6ecbf33c2b66aae7e65cf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05f2f8057f06dd54d6ecbf33c2b66aae7e65cf29"}]},"branch":"refs/heads/master"},"0f5c679dc93d25d0ab8818924adfd8037a18b289":{"kind":"REWORK","_number":10,"created":"2025-01-30 14:28:00.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/57/929457/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/929457/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/929457/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/929457/10"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-01-30 14:27:02.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-alertmanager: 0.27.0 -\u003e 0.28.0\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-cadvisor: 0.49.1 -\u003e 0.49.2\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f5c679dc93d25d0ab8818924adfd8037a18b289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f5c679dc93d25d0ab8818924adfd8037a18b289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
