)]}'
{"id":"openstack%2Fkolla~929893","triplet_id":"openstack%2Fkolla~master~I6e726d8ebf9dbc6d019dd72d6d0be9d9957a3f22","project":"openstack/kolla","branch":"master","topic":"mariadb-proxysql","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-07 12:30:39.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-01 15:50:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-10-07 11:15:44.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I6e726d8ebf9dbc6d019dd72d6d0be9d9957a3f22","subject":"Bump proxysql to 2.7.x","status":"MERGED","created":"2024-09-19 08:29:38.000000000","updated":"2024-10-08 10:00:04.000000000","submitted":"2024-10-07 12:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929893-mariadb-proxysql","meta_rev_id":"93986db61dc3d973eb9ccd8936fc6bed9f738dcb","_number":929893,"virtual_id_number":929893,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 12:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-07 10:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-01 15:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-07 10:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-07 10:57:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-10-07 10:57:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-07 11:15:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 09:25:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-19 10:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 12:19:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-07 10:57:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b1300e950d59796400d9b6cc63c57f365aed6d2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 08:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38def57c12a88116ded04f931cc35c7e5147d0c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-19 09:25:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d6cde269aa3bc35cddd86defd7a922fb714488d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 10:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9c12dca85644e3c9e2b99cc3af20e27\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ade3e22c1d124021a611d534b0436d81 : SUCCESS in 33m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/633b1e5090d14ef3ad00c6dce7da6230 : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/885ea102e1084eafb0c7320d8228ee89 : SUCCESS in 28m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/68c5c05864ce438a994fc14eb0e11e4b : SUCCESS in 35m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4b450a1e7cb842ecaaa58d3bf99f3f0e : SUCCESS in 52m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/80269c3cff7e41c4a3ea57b90ed4ff82 : SUCCESS in 1h 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83233ecacfb3450d81aee6978695ba03 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5959f35b70454df38ddd97e4f5e322f2 : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b09485b8cfff48929b9b51490193c88b : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/429be8e451cb450c8cc485d3a071f352 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f900994e04e456697d8d6112931eb68 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb444af282b54fdea30966fe5484f211 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f854e00091684c308966ace2bac69941 : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43b594f47a644876b14feab9aa4833bb : SUCCESS in 29m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2598a7f8d7f0438ab55a1ddbfdafaea9 : SUCCESS in 31m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f4bc1acee453414b9077b48d6ffb91a3 : SUCCESS in 1h 05m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/448da4635cc24c7ebba4ca0124be3d0e : SUCCESS in 1h 34m 29s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5419263828524f0e96c3d3d96156b75d : FAILURE in 1h 15m 11s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a1f0130c826e45bc822383c10c03a71d : FAILURE in 1h 40m 37s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/224f2374a0da4325a3fb548a0ed76b2a : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/de52f2476c1a41299a52a7e66de87613 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/19de6a2adb64401dade59263d37fc4a0 : SUCCESS in 27m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bae466d854c44601b627564d2401ba40 : SUCCESS in 53m 53s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0da0b3fc6ff745539fcb8452d9c3822a : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0fbf677e3411477086500a75fa8e3711 : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/899479cb779d46518611b4e1270e61ff : SUCCESS in 31m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/21d0791a2b1f405c8035177f7ba26417 : SUCCESS in 18m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/42c2e13ed48644de97fe5da083c94580 : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8d8442a612eb4858a52ab1cc41d1f50a : SUCCESS in 47m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9497ef62f67645f39d7eb19d205f122c : SUCCESS in 1h 09m 15s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6b9b23f1f9194ec8adbf4249b6197e0e : SUCCESS in 1h 37m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f92448a93e0b443fa737efe806e34a6b : SUCCESS in 2h 20m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5ce735da3ebd496c895fe68964d1cd8b : SUCCESS in 56m 38s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/29395bb9c4824feead7aa8f78d1fd43e : SUCCESS in 1h 06m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2c14d4eb97f5426b86f4a6981b9de3fd : SUCCESS in 28m 46s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fc89e3d201384c9ca2541ef19ecaa4c1 : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0e515601cf3145f6a00ec4cce6c50a37 : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/74b1d18bb50e4e21b480f124a4865507 : SUCCESS in 1h 10m 51s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4f6d0abf4f3d45ae972423fa2e20f09e : SUCCESS in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f8f643c8daf46ceb24197ffd00c57c6634cb157","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:53:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cd5de64c06a4e1481456fbfffd1fa4d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b343b9af62a46b3b4667caf61ab93c9 : TIMED_OUT in 3h 04m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0db3522077864a4dae516899d3237d92 : SUCCESS in 2h 57m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7432abe7369941969b88f507cb4757ef : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00835d9915c54ab59db6b23771980f48 : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39527ef46c4f5f1a808884356f86ed9795b7d640","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-20 23:17:31.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"5cdeac566a335c727724b30173d01cd45ed3d03c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-30 12:19:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b27141e5eddfa969750eb54a5e2274f38da7b63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-01 15:42:13.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"abf323d7ff6e56cd3480c9d56784559cff6e9e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-01 15:43:34.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c28e7bdc20703a952b428e18e291d35e9e046acb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 15:50:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a4d6186a38202e0c1592ef001f1a7f6ce733742e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 18:41:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7ccd26ca9834f49a52649a9699e7506\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/08fa66f5b0164773ae36e85a6b996ab5 : SUCCESS in 27m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/aead495bfbda4c739603842eef0e68ad : FAILURE in 7m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/af51b6aeef4b4a9dbd97f65275608d05 : SUCCESS in 20m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5144959b4d644c32b666df171ff84f6b : SUCCESS in 38m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2c97e521afdb4a9bac41f5f1c117676e : SUCCESS in 1h 33m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/acdfd80b8f5c41a09c2606ab878808b9 : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eead62eab97f411eafc0b174c9d89d3b : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2915376a333240bbbfc21a29cd057e29 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce0bcb3c3bd648cb8a1e77c5f5be503e : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21c4ce70f00c464db3824d6263c74746 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71a3c39af5244d7a59be194dff53130 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cad1936ba85e4d95b011b7a1f9f391cf : SUCCESS in 3m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea2e5bcaa5a5496a9cdd5d0529b71009 : SUCCESS in 27m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/75ef08cfea91455382d9eec20cc55d54 : SUCCESS in 44m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4cfa30cdc16d4dca982a77d6b55cff17 : SUCCESS in 1h 06m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/22a5766d46344c2fae1fab0f45bd3e6d : SUCCESS in 1h 26m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1c7ceff9b0b74ca6b8e3427821d74e02 : SUCCESS in 1h 42m 05s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e360102de1e44a06a819a2ac2ef95ffe : SUCCESS in 2h 19m 42s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/73fd2ada285446a285a09261d85ca510 : SUCCESS in 1h 25m 14s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8cbf20cb75404cb4928051f9658957b5 : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7ccbcc9483cf4df89fb09ee393236838 : FAILURE in 25m 41s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/41f17a208c6b481eb4c0f45b6c3f2295 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ed9404cd52db46e49ba9f63d3b59301d : SUCCESS in 1h 06m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/10d04915ae08402a90e2d3ea9a5b2312 : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f7c414e391be4b7795e66d6a1cbb5709 : SUCCESS in 36m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c4ac34ff2db742ebb704c326db3414e0 : SUCCESS in 24m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c4eaa72833664935ab05e21221a26434 : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d11f0b280a5940c187fe57ad15ef6b37 : SUCCESS in 59m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7fd8275f1fad444cadec043997caa635 : SUCCESS in 1h 35m 03s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/556aa39d8b3b43f9a10146baa9bb5e5f : SUCCESS in 1h 41m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5206f4f2c4c44ea88f9c7e5de7df7314 : SUCCESS in 2h 51m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1b6b43e790c347adb4112f052276f1a7 : SUCCESS in 59m 21s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a6711435586648e79b83b4125e4f30b9 : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e464752119764e769faa45a683a63ee5 : SUCCESS in 23m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/691cf120ee0c4739b29d229f9b76b641 : SUCCESS in 1h 18m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6f3016ae55614f1babc36e57aec93423 : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7783794ff2f34115a795f4a9503b1d98 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0b176a4fa60b4d9d817ed691d2401627 : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88d90bc9304b41b3eb3dfee567d5d5ba6d853bfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:21:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a079e9f8004788a65171f521867602\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a9301c4b0cd349299032f8b7d91abf09 : SUCCESS in 49m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77831a36c4f044f6971c4bc3f6bfdfe6 : SUCCESS in 22m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f11baca2fbbc431ab72ed5d9b3df5d60 : SUCCESS in 2h 50m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e1cc2814832433fb0ea170d0f3d4fd2 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ef49ebe9ad8460f9d803fee1274ee9f : SUCCESS in 5m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f7ad04e3d71e41f38e01895c33b2c560 : SUCCESS in 39m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/15a89495b7a24d48a253cae95846bc59 : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47707895f6c2f8b2978dde803cc457f33f2b6716","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-02 10:02:07.000000000","message":"Topic set to mariadb-proxysql","accounts_in_message":[],"_revision_number":3},{"id":"5116f938a25fe406590a8b80a04bd6cab7414616","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-07 10:56:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"83ce3ef2dbd7bc4bc895bc9951a313caef6b88d5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-07 10:56:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"26ede28c65d02857b194053221037639d2061c53","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-07 10:57:58.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"171ab515931adc267affa140c9f40034191e13e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 10:58:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0573347b49ecb64c760d79c204d1cb207c1c8e00","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-07 11:15:44.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e88e352077dffb317ac66a845e7354b3b7f6ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:30:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d03be694e274fcb91e4278baa5733ef\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4874f737fe734e0c9bc6937d86ea0bf3 : SUCCESS in 18m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a1dc2c6a020d47c89996c88864151f7e : SUCCESS in 23m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1921be6f8a5f46e78873b5aa1e9168a8 : SUCCESS in 1h 04m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c71a738327d844ba9dc5f60efcae6573 : SUCCESS in 1h 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8d0d35af974382b202ebf5773b39f7 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7495abf7c0cf4d7e8b0383b99337311c : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/571e29138ff444b490dd3deeacc3d4c9 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8be20e02efe488a91b9605a885dbae0 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c53ec83962664085b09dd1b9cef650ac : SUCCESS in 3m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7b96e2b6e3643e187a3dc5b63d4e635 : SUCCESS in 28m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/994b06b0ca404429a4047e91d397a8ad : SUCCESS in 40m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/405b3a35499e4a2ab121523e68906909 : SUCCESS in 1h 07m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9556375394134ceaaa6673bfeae2ce57 : SUCCESS in 1h 22m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ac4e098bc7948be851e7d5ae4276ad6 : SUCCESS in 22m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0b05b946dd334ec8965f8a563a24ddd2 : SUCCESS in 23m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5cc230e34fc34781b0576d203fdebc8f : SUCCESS in 55m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71bd133f208f4c8b871ebdd475d80886 : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cd5f119b7447f376a49c131a3d68252ca65a2562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:30:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"79563504120ce787df5314f689a2200af3a24a19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:31:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9154a82b51914775a7e9731e5f907de8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634f10d73094494594e8cc81b3d7d059 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5db34c427f0495aa66b939d1313dc2d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"93986db61dc3d973eb9ccd8936fc6bed9f738dcb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-08 10:00:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"046ad524eb579827ca756e371dc935eabdddebc7","revisions":{"5002ac94af9c4325c0858364bf8b08b942b2c11e":{"kind":"REWORK","_number":1,"created":"2024-09-19 08:29:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/929893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/929893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/929893/1"}}},"commit":{"parents":[{"commit":"015fd8d42a42a49243970c703fd513d3d5b2ce74","subject":"Merge \"Fix cinder-backup: add missing sysfsutils,lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/015fd8d42a42a49243970c703fd513d3d5b2ce74"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 08:27:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 08:29:29.000000000","tz":120},"subject":"Bump proxysql to 2.7.x","message":"Bump proxysql to 2.7.x\n\nThis patch bump proxysql 2.6.x to 2.7.x\n\nChange-Id: I6e726d8ebf9dbc6d019dd72d6d0be9d9957a3f22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5002ac94af9c4325c0858364bf8b08b942b2c11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5002ac94af9c4325c0858364bf8b08b942b2c11e"}]},"branch":"refs/heads/master"},"b67859c6a0a3574f1f9e4b50249d294aae301a80":{"kind":"REWORK","_number":2,"created":"2024-10-01 15:42:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/929893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/929893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/929893/2"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 08:27:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-10-01 15:42:11.000000000","tz":120},"subject":"Bump proxysql to 2.7.x","message":"Bump proxysql to 2.7.x\n\nThis patch bump proxysql 2.6.x to 2.7.x\n\nChange-Id: I6e726d8ebf9dbc6d019dd72d6d0be9d9957a3f22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b67859c6a0a3574f1f9e4b50249d294aae301a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b67859c6a0a3574f1f9e4b50249d294aae301a80"}]},"branch":"refs/heads/master"},"046ad524eb579827ca756e371dc935eabdddebc7":{"kind":"REWORK","_number":3,"created":"2024-10-01 15:43:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/929893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/929893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/929893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/929893/3"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 08:27:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-10-01 15:43:26.000000000","tz":120},"subject":"Bump proxysql to 2.7.x","message":"Bump proxysql to 2.7.x\n\nThis patch bump proxysql 2.6.x to 2.7.x\n\nChange-Id: I6e726d8ebf9dbc6d019dd72d6d0be9d9957a3f22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/046ad524eb579827ca756e371dc935eabdddebc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/046ad524eb579827ca756e371dc935eabdddebc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
