)]}'
{"id":"openstack%2Fkolla~929952","triplet_id":"openstack%2Fkolla~master~Iad4a93cd7318eb131a092799182236198b5d657b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-19 19:24:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iad4a93cd7318eb131a092799182236198b5d657b","subject":"[DNM] Debug IPv6 proxysql + some attempt to fix","status":"ABANDONED","created":"2024-09-19 16:12:29.000000000","updated":"2024-09-19 19:24:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd7cddda6b33a0cece8e80e83e4c5c4d06fab414","_number":929952,"virtual_id_number":929952,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-19 19:15:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 19:15:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc5dc1b80f658ad5411f2d0385af6708865c3521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 16:12:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45da332c4107d5fbddb972bf6ebd963d8512318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 19:15:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/781772ac96db456199077d3459c511d8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aebc09c4e6da4c5fa95908e7d0d25202 : SUCCESS in 27m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/48057bbcf53a4f70b9da8c7ae5108634 : SUCCESS in 57m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7cf023ce539743bdbc605a6d4cdc6863 : SUCCESS in 23m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3876461b66764dc395c2ca308d23e0e5 : SUCCESS in 32m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f1a16fda69ab422daddfd2d33953fd88 : SUCCESS in 1h 07m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/56a1d55496094311b17b91039ca5071c : SUCCESS in 1h 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4169a73df66e4c4eb262aa2fa6a8e48a : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11875e83b6aa464ebe13e46a4875c8fa : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b26d7e22c9cc4b36b978857fa1a5c4fa : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70c17da875934543b3a8fbc25bafca84 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b64fa0475e8f4e5f88ab562c7e95bc55 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1131f687d24a49955d3527c683945b : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41db68ff51f14ad59ec3d61f27e1c15d : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7f0850f6c59a43f88377f7db381eee58 : SUCCESS in 31m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d429b80bd5ea41f69e7ec203c7d4e3f8 : SUCCESS in 38m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4800820b61314994aeda0811713a3510 : SUCCESS in 1h 02m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9c50056b5144707802a2c5d6c0a30f7 : SUCCESS in 1h 27m 18s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/93db7fd95ccf4653bfe56c6cfc789a11 : SUCCESS in 1h 43m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2df03db0774c428cb8f314fffc2bac4c : SUCCESS in 2h 55m 49s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8371b7a95ae24157aa75bb013d630e87 : SUCCESS in 1h 08m 03s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/91721ae2322e45b7bcb0907d667bdd70 : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/49e45c2945b74c51b3e7beb9ff18dc04 : FAILURE in 25m 22s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bda1df039d524dffa352461fd7b47f1c : SUCCESS in 1h 22m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/487b998fc3ca4c46823d2ee9f0a006fa : SUCCESS in 1h 12m 04s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/86e57e2941a0496c848ed2e27b69d593 : SUCCESS in 1h 31m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9fb05460760247f2a91c74eae8e76c3d : SUCCESS in 32m 53s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/9bb6692c81a248608262065a168932c1 : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02ccbdb931b04fae8f299c13998d40d7 : SUCCESS in 19m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/28a893c5ba974f0db40523ee1b73f271 : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8919de99cc584ca4ba025e63645ef0d3 : FAILURE in 33m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e51dfcfdbd504d6bb3c6fd2f2d65e68a : SUCCESS in 1h 14m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/10311d09d6dc4adb9d5853df72e92196 : SUCCESS in 1h 25m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/986c8dbaf68941d392ad55791c415bc4 : SUCCESS in 2h 09m 22s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/221ee69dc34b4c02a93db82258e6dd5a : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/828ee840ebc04c2bb8febcde7b076ccd : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/576f67feccac496eb470e62a8f1f8941 : SUCCESS in 19m 50s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/1bcb36dfd97e4c928386e431b9dc8dfc : SUCCESS in 43m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d661a4a1397241f3982e5535ad862e0d : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c605cfdc7087459096dbf353d9cff73a : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9086d12d1170410c9360138f5467ff8b : SUCCESS in 29m 43s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/e2b55d214f9b458383cd621b49968556 : SUCCESS in 1h 04m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd7cddda6b33a0cece8e80e83e4c5c4d06fab414","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 19:24:25.000000000","message":"Abandoned\n\nMission accomplished","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed21454f18999afa45c9e0d363db0aa08aede844","revisions":{"ed21454f18999afa45c9e0d363db0aa08aede844":{"kind":"REWORK","_number":1,"created":"2024-09-19 16:12:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/52/929952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/929952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/929952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/929952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/929952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/929952/1"}}},"commit":{"parents":[{"commit":"015fd8d42a42a49243970c703fd513d3d5b2ce74","subject":"Merge \"Fix cinder-backup: add missing sysfsutils,lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/015fd8d42a42a49243970c703fd513d3d5b2ce74"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 16:11:41.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 16:11:41.000000000","tz":120},"subject":"[DNM] Debug IPv6 proxysql + some attempt to fix","message":"[DNM] Debug IPv6 proxysql + some attempt to fix\n\nChange-Id: Iad4a93cd7318eb131a092799182236198b5d657b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed21454f18999afa45c9e0d363db0aa08aede844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed21454f18999afa45c9e0d363db0aa08aede844"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
