)]}'
{"id":"openstack%2Fkolla~930686","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ib22a7ffaecc97408dcffe351232abbfdf7122c7c","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-27 09:31:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-27 14:17:13.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-27 14:17:13.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-27 12:38:57.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ib22a7ffaecc97408dcffe351232abbfdf7122c7c","subject":"Revert \"Pin OpenSearch Dashboards to 2.15\"","status":"MERGED","created":"2024-09-27 09:28:35.000000000","updated":"2024-09-27 14:18:17.000000000","submitted":"2024-09-27 14:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":928596,"submission_id":"930686","meta_rev_id":"498a1c20655a2e13d8fdc61f6456ace4a9301319","_number":930686,"virtual_id_number":930686,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-27 14:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-27 09:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-09-27 12:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-27 10:04:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-27 12:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 09:28:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-27 09:28:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-09-27 09:28:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-27 09:28:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 10:04:43.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9b52bd72e26bc63ee1efac533c94a7866c100cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 09:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"871f1b4678ae7c73ffe0f1e20d4e842f32eeab48","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-27 09:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b3b7783cf775dc7d9fe6726c84830d1fc3228c0e","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-27 10:04:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f599146016b2a3bf5aa4fb83f4fee327cc0c632a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 11:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d19e069b61be4a9dbadd1c8332192261\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ae47ee22033e4c44b2e50b96be3d05f7 : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6b5cec0555242d5a7008188096b353e : FAILURE in 5m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3124227396114576bbf6d7163ba10797 : SUCCESS in 21m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a887cd9fe95540f7b20b45db112bfd41 : SUCCESS in 27m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7357d16c80cc4992b8579879ee6f8f2d : SUCCESS in 56m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/17b92dfcee4c49b08ef94e7efa7771c8 : SUCCESS in 1h 33m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2935bb4c7d49440ebb8dbd3c1114a92e : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d825011764ce462a81ff8266ce483ffb : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b31fa80ed64d18b0405f66c919904b : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64c51264aa6a4883a245391980bf4986 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea4f2980ab04fdb81010506287ba92d : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab9ae0afb4f14886ac07d160cf481bdc : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/95441ef274604bcaa4884e1cfb3a1ec6 : SUCCESS in 28m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a3795c25ae104c96a0cb81fcbb302382 : SUCCESS in 28m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cc156bbef6224976aed354eb6e261805 : SUCCESS in 1h 11m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6d7ad1b778164a0682a392c874270595 : SUCCESS in 1h 26m 52s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f9da96fcfc7146bf81920feea1b1818a : SUCCESS in 35m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b7d99f2e5b7445e9a13cbcdc5c3d4eb : SUCCESS in 16m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fd19bf8d3a8a41b88e0ffd01b342aa0e : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20b6214d128c48ec95408d09ef7d31d3 : SUCCESS in 49m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5c1bab258c8840c78ae148f10f0b2251 : SUCCESS in 1h 13m 41s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/60d711d63a4148f9a733e5a48bc62bae : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824dbb5eb11b9657b359c162a42838349d35391f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:33:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f473afde84d4c96ac06737aea3ebc1e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2dec5150d8147d6b148e3c77fd4f917 : TIMED_OUT in 3h 02m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a5f526a8d75847b693d8892df99e461a : SUCCESS in 2h 21m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac4d9ba101364a58885adf4b86aaa468 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad7b8a29ce1d40cbbe48e95d5e927fcd : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d66979de588e650e4adc95c1ed840bb68d735c0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 12:38:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01fa747ff61505ea6c659f548ea5849117a5cb18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:39:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdbe00a0f038a5f985d8532b09e981b3bfd93576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:17:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d4499cb89714fdd95eae3a56446d760\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1c75a48856324cd0b3e793ca86d88a7b : SUCCESS in 17m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f72ebaa4fc45435ca9806c249cbef78b : SUCCESS in 27m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a820f09c60043e6ba0bb9472cde691a : SUCCESS in 1h 02m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64927ff882d24112ae87d8d15e0eb451 : SUCCESS in 1h 35m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1ae5f5a0fc400f97116708c0aaa079 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0e0032e8cf24f6b8a6907753ce3e283 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebc7c216f84b4ecdb01adbc82668dafd : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dca4826d26343139d003a414abe50fb : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3bf0db195014858bbb1aa7a4786375f : SUCCESS in 3m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc79ec75a19b46fb9620153de880bf38 : SUCCESS in 24m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/14e6cfa955e2459e9f2097148c167448 : SUCCESS in 33m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/03aa8fdc86c44f0995aa756017cc2f31 : SUCCESS in 1h 05m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/08d1114359e0420c8f46f02dc90143d6 : SUCCESS in 1h 28m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4e37511019b4ca0b63b4b07a8e538a4 : SUCCESS in 16m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b0a277ba387b4f279cc0ec96bab4c312 : SUCCESS in 1h 07m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe5199ed03764878a0fee9d21aab5a19 : SUCCESS in 47m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/60590437ab8f4a56b9460f4194433a57 : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"11f28decef4d89a895fb8587aa59842ed543918b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:17:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"498a1c20655a2e13d8fdc61f6456ace4a9301319","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:18:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80189f806aa54c29851910cc68673f4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd38f87e807c453a803cb2119f477ea4 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591b326f1c2c4cefba4036fc4f351ad9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9639f05c86be497fd426c77c124d7d2b006d4e3","revisions":{"d9639f05c86be497fd426c77c124d7d2b006d4e3":{"kind":"REWORK","_number":1,"created":"2024-09-27 09:28:35.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/86/930686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/930686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/930686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/930686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/930686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/930686/1"}}},"commit":{"parents":[{"commit":"a58a7f88dea997a2a98ce4f5d1894b13d1f0d4d5","subject":"Pin OpenSearch Dashboards to 2.15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a58a7f88dea997a2a98ce4f5d1894b13d1f0d4d5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 09:28:35.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-09-27 09:28:35.000000000","tz":0},"subject":"Revert \"Pin OpenSearch Dashboards to 2.15\"","message":"Revert \"Pin OpenSearch Dashboards to 2.15\"\n\nThis reverts commit a58a7f88dea997a2a98ce4f5d1894b13d1f0d4d5.\n\nReason for revert: bug fixed upstream https://github.com/opensearch-project/OpenSearch-Dashboards/pull/8097\n\nChange-Id: Ib22a7ffaecc97408dcffe351232abbfdf7122c7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9639f05c86be497fd426c77c124d7d2b006d4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9639f05c86be497fd426c77c124d7d2b006d4e3"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
