)]}'
{"id":"openstack%2Fkolla~930687","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I5dd2af9fc394ee0e2adebf375310f056823a931c","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-27 09:31:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-27 14:13:10.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-27 14:13:10.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-27 12:38:38.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I5dd2af9fc394ee0e2adebf375310f056823a931c","subject":"Revert \"Pin OpenSearch Dashboards to 2.15\"","status":"MERGED","created":"2024-09-27 09:29:05.000000000","updated":"2024-09-27 14:14:13.000000000","submitted":"2024-09-27 14:13:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":928597,"submission_id":"930687","meta_rev_id":"faf2f364260f82287f5afa5d28c855a562193315","_number":930687,"virtual_id_number":930687,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-27 14:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-27 09:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-09-27 12:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-27 12:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 09:29:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-27 09:29:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-09-27 09:29:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-27 09:29:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73a05c1da6882b14b1884f2ac671a9c4983cc973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 09:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d64344a14c3fb6a5c0a7eeddfea2018066592b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-27 09:31:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cff6a9ef512459fb4b30e6e14e100ac59e34152a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 11:18:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab26a4ebe5746d98723191893a6fc68\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d14ea275055b41859114db07a22907a1 : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7d6f7c10fd08458690e3d1d705a6e513 : FAILURE in 6m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f6803a4e52f542498619e677276aab4b : SUCCESS in 17m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/20f5f1a407214b198bf04ba1f5ed3c84 : SUCCESS in 27m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b5b42582e757414180d27360202eb518 : SUCCESS in 1h 09m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9d9ac5fdf05b43759707180c68b741a8 : SUCCESS in 1h 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2d727b9bb942b188eae35b592780a3 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08ee44119ce64d37b3f9486e6f113a1a : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cb38182202e40be90b7d4fa89899d69 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ccf5e0a27874d609d1d56261b7f60f3 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ede6dddbdad44cd08ab2f949ea262c14 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11b46d0f4ff543929ab0fe64513f8be0 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd600765df83467d8a71eaa8b5e8bfd4 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c976f02d6f4043ad8cdbf54e59c1e79c : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fe3cfe7cc3cb4179b601eb1485348df0 : SUCCESS in 23m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/167c750a56dc40fe850933c4daaca9ed : SUCCESS in 31m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/112e12fa48a448f682904fac9610a7a2 : SUCCESS in 1h 08m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c457bb9242fd41868490a4f03997c169 : SUCCESS in 1h 38m 14s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c0319c2b36014a8792802b6de59683a6 : SUCCESS in 35m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e764812b416041cab0be4d7214cc9888 : SUCCESS in 19m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d24dca52eff74d6d9d9d3327d69027fb : SUCCESS in 1h 15m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a88c50ddece94b70bc0ae7627d42ef3c : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d46376eae75941e58185ff0b77ebebf5 : SUCCESS in 1h 12m 28s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/de8deef927ff4a9591e73d7e13b9d11a : SUCCESS in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15cb5024e247be5adcf1ecbb79ada318fbb0ab5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:34:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68aef7b32a947d591356f24c0ad9323\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83f6b454e14f4adbbbe0db3509ec3c1b : TIMED_OUT in 3h 03m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3b8e361709a41dc81bda5cb87b7d72d : SUCCESS in 2h 45m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da59a48533854849a7a7cebdcdce666b : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4baa38a9785640baa1529cd63ac043bd : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d358152faff94ac3beebae9bf830e6b4 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5615c6489d0f46fcbeb3606aaa0548c3 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79437427b6847702ce473305758e0bfbde168091","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 12:38:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8844daa54de5ced8b2d192a239abd54dda61f26a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:39:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a97f6aa2f2e13034c9b1aec4ff779ab322382bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:13:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033f46fc3fea4e7f844866b78c7afe89\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07c8e964d5d34284826305313c82ef8b : SUCCESS in 17m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aa05095a67d54c79aea68d644162d241 : SUCCESS in 31m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4a33da4b95bf48deb60bc2c75b494e31 : SUCCESS in 52m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6fc1b975e3a4ffb994e1b5e49bf0edc : SUCCESS in 1h 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a2939d44514243a1ae9c16fd087a0f : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b568d60a0b498ebe277595e5b231d8 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4d0e1eacddc467784307507914dc962 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17dcd507466246b781290eaa72fc2d6a : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ce45c08e1d4add88aa346214ec5880 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a85566e7de604536b8abf3680d811b98 : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/487698e0be9d46e88284347719b38fec : SUCCESS in 23m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e05a2cbf9cc0423ba05e3e365ed594cb : SUCCESS in 36m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fa1c088ef6834dd694497f7b74a09c03 : SUCCESS in 1h 07m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e298505730d84bc6a13126d97707f04f : SUCCESS in 1h 34m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f695fa82946f436c889d7b3c58d613ff : SUCCESS in 19m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f33fbcd99a1444d9b30574a9107ddaa : SUCCESS in 46m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/530a0560d38d4d18bbe9d01fdf476b50 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5a543869a6104ba19c575285d41d78ab : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"34fa6663dbfac0386d9ef28bf9b11cb11191238c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:13:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"faf2f364260f82287f5afa5d28c855a562193315","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:14:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98519effcb2741399959da81687e70b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e21284894f49f9b6918bbe70048c03 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac1f60c51124d5b90e500bdf456e857 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10291ab3e4ced62d6484089c66bc46d3e4f7697c","revisions":{"10291ab3e4ced62d6484089c66bc46d3e4f7697c":{"kind":"REWORK","_number":1,"created":"2024-09-27 09:29:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/87/930687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/930687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/930687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/930687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/930687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/930687/1"}}},"commit":{"parents":[{"commit":"9dc0371b53997523ddb6bc782802baea4cbe7bbd","subject":"Pin OpenSearch Dashboards to 2.15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9dc0371b53997523ddb6bc782802baea4cbe7bbd"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 09:29:05.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-09-27 09:29:05.000000000","tz":0},"subject":"Revert \"Pin OpenSearch Dashboards to 2.15\"","message":"Revert \"Pin OpenSearch Dashboards to 2.15\"\n\nThis reverts commit 9dc0371b53997523ddb6bc782802baea4cbe7bbd.\n\nReason for revert: bug fixed upstream https://github.com/opensearch-project/OpenSearch-Dashboards/pull/8097\n\nChange-Id: I5dd2af9fc394ee0e2adebf375310f056823a931c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10291ab3e4ced62d6484089c66bc46d3e4f7697c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10291ab3e4ced62d6484089c66bc46d3e4f7697c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
