)]}'
{"id":"openstack%2Fkolla~931018","triplet_id":"openstack%2Fkolla~master~I555efcaec8672130a7942d14a0b514a8375f73a4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-03 16:22:48.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-03 16:22:48.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I555efcaec8672130a7942d14a0b514a8375f73a4","subject":"CI: Move aarch64 jobs to check queue and mark them as voting","status":"ABANDONED","created":"2024-10-01 13:56:51.000000000","updated":"2024-12-11 19:03:30.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"896dd71555839b9146c61da372a10d2900e8f2f5","_number":931018,"virtual_id_number":931018,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-04 14:19:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-10-01 13:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-01 13:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 14:02:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"664594c1b24b3f00d0cf66718f6f0a60ab842cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 13:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"381f8abb5512e929133e03e64540dc68fba2958f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 13:57:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py39-arm64\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n        - openstack-cover-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n      check:\n  ...\n\n  in \"openstack/kolla/.zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"e9f586c1c107de503ffff8d376047ecac7a63ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 13:57:23.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py39-arm64\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n        - openstack-cover-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n      check:\n  ...\n\n  in \"openstack/kolla/.zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"981ffc82c081eb617d753e40e8955ed3eee81166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 13:59:36.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"466d9e676f1d547eaa68e1f65c04193b6840d153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 15:07:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1ba2ae2a53d680dead57bb72f5b1ec6f220b59b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 15:14:33.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"321d63cdd53de4f2127a9d66edbd87406f10dc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 18:17:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1c0e2d9c1634fa59b47174c0fe8a34d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fbc56a998a3d47e8800f18d768771772 : FAILURE in 23m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ca7be82e615d4c89afe729cb7af2a7b0 : SUCCESS in 16m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/81b6416de4944ba0a53297057d94add9 : SUCCESS in 2h 30m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18da5d6af1842a1a8446e1f5f94cd55 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/026aa247df094ee29a449cd81dbc7bde : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c432a5c41f59443ebe75987b6b22a746 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5736725e1f904c779e7349ea2d721f1f : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47205e6e8f6043e7a54b7cb1c1be07a9 : SUCCESS in 4m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc30fa7d9aa341f2bdd1c5e13fd0a817 : SUCCESS in 6m 22s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aebbab28a6e447399e3bf8adf2ca980e : SUCCESS in 4m 33s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ea0bde8ee9334ba2b0b82326fc442ce4 : FAILURE in 22m 58s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/07825d45d2934b59add328f801319af7 : FAILURE in 18m 38s","accounts_in_message":[],"_revision_number":4},{"id":"9b7d1e5475555fd7bfa0efdd2b94ad9e1ca00d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 05:38:56.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7dff57b2c5235813eeeb0c8b46500a47ff993159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 09:10:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d026e39d854407b9fc171af4fb0b8d7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2fd34641998e4f5d91afc89a1dd84065 : SUCCESS in 29m 22s (non-voting)\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/acfb8c23993c45e082c6279ff641598c : FAILURE in 2h 50m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff0ceb4b5adb4558960e67e79eac2bb0 : FAILURE in 6m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dab6cf748a1c4b258eb02603fb1fd5a1 : SUCCESS in 14m 07s\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1e1c9006ab145c3b2b1b386b1df5197 : SUCCESS in 1h 51m 06s (non-voting)\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d5d2aeba5eb4ae9a839c4e0b7f0af24 : SUCCESS in 27m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d2b1bf969a514673b80297d13882fe34 : SUCCESS in 1h 03m 40s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bfc36b4697c948e595151da056aeec86 : SUCCESS in 2h 08m 49s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3dea0895da43471c818d0bb47ad2bbc7 : SUCCESS in 1h 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be791ec20094f1387e3f97b6feb16e2 : SUCCESS in 1m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c7b94044eac4b2bada6b498621cf29c : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f69a34f21914ffab4b96f96228630d1 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53b44246a0a47c2b74c1db061c5a9b7 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ba1d1aaa6d4595863ca3c6bf5a9144 : SUCCESS in 3m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcd43cebbfc742eb872165821fa3b7d8 : SUCCESS in 5m 15s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/558dfa1c1e844e7d88f550cf248fcbef : SUCCESS in 5m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/69cabbc9072c4e4e8d5bfd2e2eb5179b : SUCCESS in 32m 14s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bed8add582624e28bf5970a6bb85d0ac : TIMED_OUT in 3h 02m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/77738531122541d1828ff38bf28a3f47 : SUCCESS in 35m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e32a6ccecc8e4b0faa86f83da875acd9 : SUCCESS in 1h 05m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/225778ca870c46298c52229a678faf0f : SUCCESS in 1h 35m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/17553d2fc216477d8dbe4974199098c5 : SUCCESS in 22m 24s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/40436c09657e43a1b9c5e6573e329b3d : TIMED_OUT in 3h 01m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a8d80661c7fe4e62831584e36491b3b5 : SUCCESS in 29m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b7167f11a5934cc5b35740a584afb091 : SUCCESS in 1h 03m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7a15bb3300f345889160da8c59ef11dc : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":5},{"id":"3a3a2b457085e014f73324c33ed9adf97ee9b054","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 09:18:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"439b120a503f818c8699a3b7ec0768eb6f67b543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:22:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a3710f174414470983ebd3ab212c794\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/12df35d494cb4c68a4b09dc31d06f244 : SUCCESS in 25m 07s (non-voting)\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/41f26bfdaf604710bdeb4afaf1eebeee : FAILURE in 2h 42m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ddc26b07bd3f4b699107f294ccf689f2 : FAILURE in 6m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8bb1e9419bc743f7aac09f945122b3e6 : SUCCESS in 19m 20s\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cdd5e2193dc349e7b5437c759975f41a : SUCCESS in 2h 16m 20s (non-voting)\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9274eaa39f1849d59d87aa12220e18be : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/67e023d7956c49309a884356346dd6ea : SUCCESS in 1h 03m 01s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3be306ef8b5c406f856983595afdf1ad : SUCCESS in 2h 10m 34s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8857572826b146b09501b85a6b2baa7a : SUCCESS in 1h 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1cdede630954211ae519e2da5a99a11 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73e0758bd76d42d0834850be1da716ab : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17eee6064c7e4177b8e2b784606c15ba : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ba021206c764a598932ce4f3501d59a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3013eced45294617b532ebed1c8f30f9 : SUCCESS in 4m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d83217de633f494984056679622b9a4e : SUCCESS in 4m 47s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9503a37438134e6f835acb26fb48373d : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc091250cf2f47aba17ee281dbd72602 : SUCCESS in 26m 58s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/69c17d240ddd45bb9d41af5ad430d93e : FAILURE in 2h 43m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b8befa17629940b4b66f36e24ce115f6 : SUCCESS in 34m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4f0f743a7c364eb798e0a3477ee97fb9 : SUCCESS in 1h 06m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c8c4541453cc42549d8f28822cd38c5b : SUCCESS in 1h 31m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7a553b705af4bb9b3268d27357b3cc0 : SUCCESS in 27m 51s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ede006ddc95d41579977142ff77c7c69 : TIMED_OUT in 3h 01m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aa7cc9962508457ab11b912295be05e8 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/22ceb789c57d4dc7b321d887f3ba31f6 : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6d633338bc8b4edf8f97ec0121779b6d : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4c696d8bd3538372a73f87752b288373fe8306d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:14:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* tests/playbooks/run.yml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3b06ecc6b11441d072ebdbd06630f642cab39f57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:15:06.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"0635a3c48c7e753ae2f4acc5a7ce0f886d4c5f03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:16:49.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"b120a1c04566789a90abe0298220a928710666e6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:16:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"0915bb6258c39fc03b819a1a1d7ab50cbed7ec5c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 15:20:03.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"2e9222b8ed7b4ac720ea5a011cd9cd2cd4cd756c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:28:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d224ec00fd94530b696d46a64c09caf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e4bda70f38aa4237a25917ccf1dcfeb8 : FAILURE in 3m 50s (non-voting)\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bf10d15a20524837a1171e09e035f86d : FAILURE in 3m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/958e20525eaf46dd8f49cdfb2e213f2d : FAILURE in 6m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13f012c3b6344b45a7307921ff869777 : FAILURE in 2m 52s\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/27c8e3274a2f4d92bd0b7b68f5b3c5fe : FAILURE in 3m 22s (non-voting)\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aa448b2b1d3044468b1f4b1d03c5bd49 : FAILURE in 2m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/920c6d5c5cc146b7be687cc129e984b1 : SUCCESS in 1h 03m 03s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8bbbbd8cad494fc4938ec3421972f352 : SUCCESS in 2h 16m 15s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/85e884910c0d4bee9f98de27c8bcf2fc : SUCCESS in 1h 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3440a5c74c40f590d3436de897980e : FAILURE in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6478d321ced649a58bc421e493555fa8 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c7086e3143e458fbbb057265e7f0aee : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72a362bf7c647fbbd01b1e6f983c672 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50839f1667044d8ab9aaa0bb811a3918 : SUCCESS in 4m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e9d16a5ec7345b7bb2c33076947fcbf : SUCCESS in 4m 48s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b08f4ed5aeeb47faa208aceb9fdddcbc : SUCCESS in 6m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d7e911ab3f974632b3ef619c6230a3d0 : FAILURE in 6m 10s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/466d8dffea934d81811579e32496e385 : FAILURE in 7m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/af6630b2833f40d8aefa02b7ddc2a1fa : FAILURE in 6m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/41559b1effc94512afcc5bb794907315 : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/301de128785e4c509ac7963edc791f27 : SUCCESS in 1h 24m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30146cd2ea8e451eb7b202ba87388db7 : FAILURE in 3m 02s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dbd37a83541e431c88a1c36c792b89c1 : FAILURE in 2m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e2f701d031c54cd4be02bc8c776820c2 : FAILURE in 2m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b62aece3124d4cf48afbe9077a93ddca : SUCCESS in 1h 01m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1dae3efe0ead451881e45046b0c2f48d : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":8},{"id":"88002cc2dcf1a6ae4b0b9afd3753e55042cb27d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 09:17:02.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"236c72f6b48e3579a8f95c43ae870aa98ac631c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 09:45:45.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"8d7a56ba5d37edce033ea21d1528c662c647dc5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 09:47:57.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"b0ce674e75e80a1306fc8f54d1afb58a3d5c4de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:15:36.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f9b8614e7d48379bab2b6309d42ee9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ec588e402e89490798a7d23b7c59a0de : TIMED_OUT in 3h 01m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8553faaee8f9402eada94c691673f4d6 : SUCCESS in 2h 15m 12s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1c057137a484fcbaf30fe4163ea3ed9 : SUCCESS in 1h 56m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c347f7cd450c4356b3bac93e4d90fc8a : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e87ed6405a204167a5664e36b889e6a7 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f2774e6ef674993bc4580a14bc79e97 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/702cd9675336466397d7b132dc3f62a7 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124d7370db154647bdf36527c0a95257 : SUCCESS in 4m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/092f93a11005403589b61f4a52da0d0c : SUCCESS in 4m 54s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6da45363a4a44960b7fd42dee698e27f : SUCCESS in 5m 06s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fd20fece00bf4498bbac4fb997a272cd : FAILURE in 2h 03m 16s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ccb55813b64842b4ad3588193f8bf957 : TIMED_OUT in 3h 01m 26s","accounts_in_message":[],"_revision_number":11},{"id":"c5a750586f20a9c6d08c6036538c3d49c0c5c999","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:22:48.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.d/centos.yaml\n* .zuul.d/rocky.yaml\n* .zuul.d/ubuntu.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"66be3a8e6f8dd754435d5a0e31b81ba8e8c1db51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:23:14.000000000","message":"Patch Set 12: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 27, column 1","accounts_in_message":[],"_revision_number":12},{"id":"f2da57fe9bb03b7e1c89e9eba739945973e27029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:23:21.000000000","message":"Patch Set 12:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 27, column 1","accounts_in_message":[],"_revision_number":12},{"id":"835f673d82d8943a99411beb23b14c6b0542eb91","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:24:06.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"bd79bc2ec0591a2ae02db926269832fcb2ad6ef9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:25:13.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"cd1a2f23cf7fd2e02763fe377643d1ece6b104db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:18:55.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abfc0b9745264259a966426d99ef90c9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9a7fa8a8fb4644d39b285743c389b89b : FAILURE in 2h 51m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4aadd1745df4408f92de661dafebb00c : TIMED_OUT in 3h 01m 23s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/641d63bfd1124133b8f981a6c87006d0 : SUCCESS in 2h 02m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ef98a0d9d648538a36ea4a04befd76 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a01e10f61640a8b4e7b18da1fc715f : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3256a202cc6a4c7db51c0e1b0e72199e : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/103e26b3c2bc4b91bb1e8b4dc5038a6c : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc701a802c744dfb3a1f64351c0f8f3 : SUCCESS in 4m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be47621568ee4abfbe9caeb77b44dd3a : SUCCESS in 3m 51s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a014f0a4e70499fa70298ca211dc191 : SUCCESS in 4m 59s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2fff47b551954cbd87d9baf4545d8a31 : TIMED_OUT in 3h 02m 20s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bd6bef0f865e4fdba1f3b3651eaa0ea3 : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":14},{"id":"39d23fc9588ab53ec96dede905ad0681426da9d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 06:54:58.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"212bebcd8a51c6594c66db34799d99ec29472a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 09:59:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d7bbb6eca654cc08d171e9c3aa4c655\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8d8554b38fc453ca09b4c4165487607 : SUCCESS in 25m 12s (non-voting)\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c3cf94f0d7d745ae82b37c1506cecd12 : FAILURE in 20m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1af82eb7e15d474e8267996d56eafe8d : FAILURE in 7m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/063464a42480404abaea3510cde15eeb : SUCCESS in 22m 29s\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc8ccaf7677a4b0c83f5ccb011686c7a : SUCCESS in 2h 05m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/67f4f620ffd845c8b17932653400f3d6 : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7db5793c9b4243a9ace8dfa136ab58d0 : SUCCESS in 1h 02m 15s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15fe5a59677a4ab29d7fc478a7173268 : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dbb6db65617843eb97fa7e33312b5166 : SUCCESS in 1h 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbc625e665f4069a3b9564160995ee6 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed5a3c52695a4bf090d6c9f78c43f3c7 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53a13157157c4b098233bb9d9a2bd52f : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6bace03e30d4955a268fab9d6fef54d : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238064a97bfc4416a6a52ee448436510 : SUCCESS in 4m 24s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fea7f5a25c3640a686d3b4844e10fee3 : SUCCESS in 5m 24s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a105181ac16409b844ddf0f9cee37c9 : SUCCESS in 5m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/842ae234ba5c4dce894b630c6ae9c367 : SUCCESS in 29m 33s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0eb9b03de743418d9f2379bc5de71ffc : FAILURE in 23m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0fff2c00e61f41bd88b8239e3df7585a : SUCCESS in 36m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5f5570b8d1da4273bfa12c8d54483a5a : SUCCESS in 1h 03m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f524614e32cb48bbbde6c9b6d35956cb : SUCCESS in 1h 21m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8a85f70c247b4440accd331a5915b024 : SUCCESS in 22m 13s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cee09ad4a8746ff8c8af52d03f7fee8 : TIMED_OUT in 3h 01m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/54ed8210172f4125a30b168123da5717 : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c7dab91a0e6d42bd899023f2b3b59440 : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e5e9f740684f460a95bb4f0d29975679 : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":15},{"id":"bd296c81bd19a644602cda1c8d0225da2ff74051","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 10:32:24.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"6137cb9561e1c1994e87d9884845ffa2e8b3a86e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:32:46.000000000","message":"Patch Set 16:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2023.1.  The\nproblem was:\n\n  Nodeset kolla-centos-9-stream-aarch64 does not match existing definition in branch master\n\nThe problem appears in the the \"kolla-centos-9-stream-aarch64\" nodeset stanza:\n\n  nodeset:\n      name: kolla-centos-9-stream-aarch64\n      nodes:\n        - name: primary\n          label: centos-9-stream-arm64\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/2023.1\", line 96","accounts_in_message":[],"_revision_number":16},{"id":"db7d3d3fec5feded91c216e733176679de59e766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:32:55.000000000","message":"Patch Set 16: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2023.1.  The\nproblem was:\n\n  Nodeset kolla-centos-9-stream-aarch64 does not match existing definition in branch master\n\nThe problem appears in the the \"kolla-centos-9-stream-aarch64\" nodeset stanza:\n\n  nodeset:\n      name: kolla-centos-9-stream-aarch64\n      nodes:\n        - name: primary\n          label: centos-9-stream-arm64\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/2023.1\", line 96","accounts_in_message":[],"_revision_number":16},{"id":"c9f80b79cf9b7f42ebd341755e845806a9159110","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 10:36:47.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"62aaac1436470b3ecfc72f9b4a8e8f5a95f47272","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 10:37:42.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"c3d48ad214b3143da923289861e93d681bad8541","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 10:38:18.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"44a3928a64d79c68fb84f0cb6f3e2638cf426612","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 11:08:56.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"30c44afd490cef386bb07ccde9a045d9d4aec794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:19:20.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba1ce6cfcecb4a978b5e3c153cc3f6c8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/61cd978183c9407596714a3950611a97 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f719a21d390e4b8394e4e236aa0de564 : TIMED_OUT in 3h 01m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3c2a44cb4932423ca8393879103c65d3 : FAILURE in 7m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a3899e65c373451ea442250a674bea13 : SUCCESS in 21m 26s\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8268959005944f88aefa6358726c8af4 : SUCCESS in 1h 13m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ceb26f56c4e44620af35ab2ba8fed700 : SUCCESS in 26m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0c40cc6e991d4ee3a73b8d9d509e3c37 : SUCCESS in 1h 05m 00s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62400dc8805e4ee7b0737ee2533add60 : SUCCESS in 2h 51m 20s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/20965d5388e341d59bf7841b9d86c733 : SUCCESS in 1h 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029baf69a2dd4dcaabf3938c5f9c2fa8 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44e6cdf45367443287c0344ddb93a512 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/797f5d79230b45c9b09e08832f698b97 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fda95e07bad34f7ab737619473f7cb64 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d7162847eb4cb885e8e0b74956719b : SUCCESS in 3m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78f868f0a9924cd68accb7496707f23f : SUCCESS in 5m 53s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d1a2fbd7e7448ff9f7bd1354edcc6f0 : SUCCESS in 4m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dcdc21dd66b34161852de364c910ceb8 : SUCCESS in 29m 29s\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f758f27625e40de90298260c97bee95 : TIMED_OUT in 3h 02m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/efe87d5e9fb94d8c9bfe457b741fd212 : SUCCESS in 33m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2c6972734ca042bdb247a89e956ff4d1 : SUCCESS in 1h 07m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2da000413624d39b63e3c84a22d7638 : SUCCESS in 1h 15m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f73b5032daee40bca29ae02c131ccfa0 : SUCCESS in 20m 20s\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/186aee9ec3f646a7a1b8dc8fdab55894 : TIMED_OUT in 3h 01m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/18988b88f0614714af627a57c6613960 : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/73dd3d917f6746a39078ce8d382bb6fc : SUCCESS in 51m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86c169400afb497190f8ed803fd3abb6 : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":20},{"id":"896dd71555839b9146c61da372a10d2900e8f2f5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-11 19:03:30.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"13430ce2dccd39357c307c119eb46bee7fffc1c5","revisions":{"540e09241505f4bb5c266a5f6950930b6711c58e":{"kind":"REWORK","_number":1,"created":"2024-10-01 13:56:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/1"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:56:48.000000000","tz":120},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/540e09241505f4bb5c266a5f6950930b6711c58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/540e09241505f4bb5c266a5f6950930b6711c58e"}]},"branch":"refs/heads/master"},"4ff4a46ca31b93367a7d3bde8cff56d1144a4aeb":{"kind":"REWORK","_number":2,"created":"2024-10-01 13:59:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/2"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:59:35.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff4a46ca31b93367a7d3bde8cff56d1144a4aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff4a46ca31b93367a7d3bde8cff56d1144a4aeb"}]},"branch":"refs/heads/master"},"3a00cd7c4973af05ac6be454c4554460a91b23a7":{"kind":"REWORK","_number":3,"created":"2024-10-01 15:07:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/3"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 15:07:21.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a00cd7c4973af05ac6be454c4554460a91b23a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a00cd7c4973af05ac6be454c4554460a91b23a7"}]},"branch":"refs/heads/master"},"8eb99af58878d36c02b97eebb282ab6e9a41df67":{"kind":"REWORK","_number":4,"created":"2024-10-01 15:14:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/4"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 15:14:33.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb99af58878d36c02b97eebb282ab6e9a41df67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb99af58878d36c02b97eebb282ab6e9a41df67"}]},"branch":"refs/heads/master"},"43ef3a9cc335ffbbf9e81cbefc828cc6daa08dbf":{"kind":"REWORK","_number":5,"created":"2024-10-02 05:38:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/5"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 05:38:55.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43ef3a9cc335ffbbf9e81cbefc828cc6daa08dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43ef3a9cc335ffbbf9e81cbefc828cc6daa08dbf"}]},"branch":"refs/heads/master"},"bb656100bf26263b723bc407ea6d21e87abf9408":{"kind":"REWORK","_number":6,"created":"2024-10-02 15:14:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/6"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 15:14:44.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb656100bf26263b723bc407ea6d21e87abf9408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb656100bf26263b723bc407ea6d21e87abf9408"}]},"branch":"refs/heads/master","description":"Rebase"},"ca6c9137f0562d652dbcf6daf4502ba8904e4038":{"kind":"REWORK","_number":7,"created":"2024-10-02 15:15:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/7"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 15:15:05.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca6c9137f0562d652dbcf6daf4502ba8904e4038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca6c9137f0562d652dbcf6daf4502ba8904e4038"}]},"branch":"refs/heads/master"},"2ad52b678cb73d7be76235f2c53b915eb625987f":{"kind":"REWORK","_number":8,"created":"2024-10-02 15:16:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/8"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 15:16:48.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad52b678cb73d7be76235f2c53b915eb625987f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad52b678cb73d7be76235f2c53b915eb625987f"}]},"branch":"refs/heads/master"},"02f4f48e234168373343413dad091ebb62df4f79":{"kind":"REWORK","_number":9,"created":"2024-10-03 09:17:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/9"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:17:01.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02f4f48e234168373343413dad091ebb62df4f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02f4f48e234168373343413dad091ebb62df4f79"}]},"branch":"refs/heads/master"},"4670723434b9c9ca75e4a00a6870cfb988590531":{"kind":"REWORK","_number":10,"created":"2024-10-03 09:45:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/10"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:45:45.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4670723434b9c9ca75e4a00a6870cfb988590531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4670723434b9c9ca75e4a00a6870cfb988590531"}]},"branch":"refs/heads/master"},"d05e326902ce8e227ba05973890041eb9fc584d9":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-10-03 09:47:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/11"}}},"commit":{"parents":[{"commit":"ae441689d8f9ea882742f8ae17e4e7bcc7c37d29","subject":"CI: Fix aarch64 builds on x86 after switching to role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae441689d8f9ea882742f8ae17e4e7bcc7c37d29"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:47:57.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d05e326902ce8e227ba05973890041eb9fc584d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d05e326902ce8e227ba05973890041eb9fc584d9"}]},"branch":"refs/heads/master","description":"Rebase"},"fb078753354c0652b9393fbd48e6fd7d9d344755":{"kind":"REWORK","_number":12,"created":"2024-10-03 16:22:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/12"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 16:22:48.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb078753354c0652b9393fbd48e6fd7d9d344755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb078753354c0652b9393fbd48e6fd7d9d344755"}]},"branch":"refs/heads/master","description":"Rebase"},"398fa11fffcf40319f02a4a210de54af320cab93":{"kind":"REWORK","_number":13,"created":"2024-10-03 16:24:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/13"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 16:24:06.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/398fa11fffcf40319f02a4a210de54af320cab93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/398fa11fffcf40319f02a4a210de54af320cab93"}]},"branch":"refs/heads/master"},"e750af44e5f8426c0583fa5a47cde8a6f2d146e7":{"kind":"REWORK","_number":14,"created":"2024-10-03 16:25:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/14"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 16:25:13.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e750af44e5f8426c0583fa5a47cde8a6f2d146e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e750af44e5f8426c0583fa5a47cde8a6f2d146e7"}]},"branch":"refs/heads/master"},"c9fa382bc56c9fa5beeeb3a90e0d6d7b32e2e999":{"kind":"REWORK","_number":15,"created":"2024-10-04 06:54:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/15"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 06:54:58.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9fa382bc56c9fa5beeeb3a90e0d6d7b32e2e999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9fa382bc56c9fa5beeeb3a90e0d6d7b32e2e999"}]},"branch":"refs/heads/master"},"7c67262b08c266ebb9f57401ff6625b68c9c2eb8":{"kind":"REWORK","_number":16,"created":"2024-10-04 10:32:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/16"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 10:32:24.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c67262b08c266ebb9f57401ff6625b68c9c2eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c67262b08c266ebb9f57401ff6625b68c9c2eb8"}]},"branch":"refs/heads/master"},"8a6b4f09e7556b28a37a1878fef6f49c4d0062b8":{"kind":"REWORK","_number":17,"created":"2024-10-04 10:36:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/17"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 10:36:47.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a6b4f09e7556b28a37a1878fef6f49c4d0062b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a6b4f09e7556b28a37a1878fef6f49c4d0062b8"}]},"branch":"refs/heads/master"},"2d8322fb70f11dd1ab340520246d896fe3f4a461":{"kind":"REWORK","_number":18,"created":"2024-10-04 10:37:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/18"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 10:37:42.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d8322fb70f11dd1ab340520246d896fe3f4a461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d8322fb70f11dd1ab340520246d896fe3f4a461"}]},"branch":"refs/heads/master"},"ead7ee0ed4ce5b4a7fbd44665ee8ef817569bc5a":{"kind":"REWORK","_number":19,"created":"2024-10-04 10:38:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/19"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 10:38:18.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead7ee0ed4ce5b4a7fbd44665ee8ef817569bc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead7ee0ed4ce5b4a7fbd44665ee8ef817569bc5a"}]},"branch":"refs/heads/master"},"13430ce2dccd39357c307c119eb46bee7fffc1c5":{"kind":"REWORK","_number":20,"created":"2024-10-04 11:08:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/931018/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931018/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931018/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931018/20"}}},"commit":{"parents":[{"commit":"672601a3e66c2d818c30192bebaf4229ae468aa1","subject":"aarch64: fix name of rabbitmq copr repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 13:55:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 11:08:56.000000000","tz":0},"subject":"CI: Move aarch64 jobs to check queue and mark them as voting","message":"CI: Move aarch64 jobs to check queue and mark them as voting\n\nSince Zuul does not really cope well with voting changes in multiple\nqueues - let\u0027s move all aarch64 jobs to main check queue, mark\nthem as voting and add them to gate queue.\n\nThis is an effort to make aarch64 first class Kolla citizen.\n\nChange-Id: I555efcaec8672130a7942d14a0b514a8375f73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13430ce2dccd39357c307c119eb46bee7fffc1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13430ce2dccd39357c307c119eb46bee7fffc1c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
