)]}'
{"id":"openstack%2Fkolla~931020","triplet_id":"openstack%2Fkolla~master~I5c69d288a427d708b42deadac872a3a7368181e3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-01 17:31:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I5c69d288a427d708b42deadac872a3a7368181e3","subject":"CI: Fix aarch64 publish jobs on x86","status":"MERGED","created":"2024-10-01 14:05:07.000000000","updated":"2024-10-01 20:58:24.000000000","submitted":"2024-10-01 20:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931020","meta_rev_id":"eb8d1b67be3afc01ae6a11a9e5e3eb61fedbb653","_number":931020,"virtual_id_number":931020,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 20:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-01 15:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-01 15:32:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-01 15:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-10-01 15:02:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 15:32:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-01 15:38:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"589ce9c04cafeede1af56757744a926f4908cf25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 14:05:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af0d89e84e65bad50803489f078dd9b1a1c74569","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 15:02:58.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0022ff4669af8a3d66d5e65963cadd00a52b9873","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 15:32:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5dc334de11ec7f0cfece73672b78a4c59ebc695c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:38:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84aaaa0b10184bd692107fbfdd1f77ba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/54eaefc78d22455c915228de56d659cb : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/10439a7902b14157bfd2e68bc0a03025 : FAILURE in 8m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6010b48377bf49088edd9c23bf4d7475 : SUCCESS in 21m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a1c7db9ad3c74b37b674f43f0101fe6e : SUCCESS in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d225994234424be091ac047e5874d3a7 : SUCCESS in 59m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9a468a6ad14e4e41875c229966bf7a6b : SUCCESS in 1h 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ac2f358dc24e79b72c78439d4b5af3 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c329491c57b4c95b0897f43fb53c939 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3214dcc38eae47ec9a3802974c6a74bb : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1d597acfa0240ef8513a1c9781be2e0 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e53d03e830e4cc8bb07c954d8400f32 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0e8b64a0c13d4aa09a4bfd4b33034388 : SUCCESS in 26m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bda1298b553c401b816e714a8e05d0b9 : SUCCESS in 30m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f8ad26f393f84a40bdb2bef42ac6e562 : SUCCESS in 59m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f56e42ff58c14a06be360cf66d3b7f49 : SUCCESS in 1h 27m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67e090a62cfe4435872b1102d4611c46 : SUCCESS in 24m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6b815727180e4ad6aba43649079fe474 : SUCCESS in 24m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d39255763e4c4a768467c14787079219 : SUCCESS in 55m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2121b3f23db043e19b775c4977588ab0 : SUCCESS in 1h 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"35a104da026ffa2aeb5d933e047a052a5fb240dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:38:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"baa9c6533051fc6bec205938d03a706cac4977b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:00:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d17d3995b84e789fd8fe897f4b8d31\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4df83c0e657f4a6e8a9db40c3c55e2d7 : SUCCESS in 25m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83f799e6294848bc9cd6ac45db016c11 : SUCCESS in 18m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a53cb8bc166246f48a6363333600cbb0 : SUCCESS in 2h 41m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93d10b64c98b4fbdaa9bebdab53b0e33 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fac9ce171ec3435b82b20c49c2dadf17 : SUCCESS in 4m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/934a6fbd5b834f1cb59e1ece9e6456f9 : SUCCESS in 27m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/341459d6fedd4e7cbb990169e4143030 : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb7b86316d7b83823fdec016e3ff9c4dcb619ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:24:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/507b4b033a2b4c80b451a2aec2df075d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/db81b54d7a844de4ae9eb565bb729452 : SUCCESS in 18m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a95bd15a55bb49b4b19a37e9ae1dba18 : SUCCESS in 46m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a172c5c5e0194f5295f9a07c4295a3a3 : SUCCESS in 1h 06m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5af06ca6423744bcacebb977581f40de : SUCCESS in 1h 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df8e226dd2c49dba8123766fd6bac83 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50366c8acf8d40278ef11792d58dd678 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb04d9cd12a045e4988c6f2c60a5429a : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e1f4eabe2d4496b7e2679ef3e23069 : SUCCESS in 5m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c049fb75ca6f4a628681aaf65d5eca7a : SUCCESS in 29m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8ee11985eade4deb949fa3089115e8ee : SUCCESS in 45m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2ddcd6b6c3f44ef38688c869f9871795 : SUCCESS in 1h 04m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d81a169a78f54f88b4f7bac262e24cf0 : SUCCESS in 1h 40m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/812542b6a515493c8cb97f9c4ec1f088 : FAILURE in 24m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ba0042ee069947bbbaa82e39480146f7 : FAILURE in 24m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1db157fca7d443539d102df144121f62 : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/40c5b02eb0ab4f8e88510e4288d49423 : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"366bddd8488227c050c81851a7bc505aa4aea404","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 17:31:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"461336a2223b718444eb7482e58bf4357226e00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec7c0d993104161925b9878d1be8409\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/64e929a6dd794cbaa1f6504dcbcbd175 : SUCCESS in 25m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6e990cc019d342a481cfdb29243b9f10 : FAILURE in 9m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e80a885f8284709b3aa583037fe797a : SUCCESS in 22m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/be8dfa60c29d4f738194d27a15c64dab : SUCCESS in 28m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/57128c9f36a24b91a294198bcc30bb86 : SUCCESS in 1h 05m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7a60715732d740b6a9d3b5be7f359d24 : SUCCESS in 1h 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13579bc961084c5981be9e238b99a237 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3ce5a36b37f49eb884d3cc0bee9330c : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/499f1227b55b4ac5a16178c459460643 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f23e97be59e5480a82f08c3d71433965 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7644c3c1ac4567867c461a414c1bf2 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bcd2ea60f4ab413cba38bcdd861beece : SUCCESS in 28m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/54067b4054694a7ca293817035308f69 : SUCCESS in 37m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/42cea8bbe2064965b8d686dddb8ca6bd : SUCCESS in 1h 07m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a906d0084b64c0d86632660c29721d9 : SUCCESS in 1h 23m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4e591b6a53c1496f96ad26c7cd7a3a46 : SUCCESS in 19m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e8230e5a09104c92bc7391c357346869 : SUCCESS in 24m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4791ec442e6e41e5abf9d9bd24066967 : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/67daa7aa340045c2af79c20475aa7762 : SUCCESS in 1h 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2e00de8f60ed73f140efcb60f05b1979e4ac5fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:12:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fe93338ac68068dc51d15aef5024d840a167c8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:28:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b67b1c6515547b9be179829baa61881\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f512bc03439746dfbbeff395b878a464 : SUCCESS in 22m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1945e0fd4fb64772afcd759fdaea2627 : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e3b6c6cb75d7491996e3956dedadbfd9 : SUCCESS in 2h 35m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/934869c6ade34258bce135c45f4bad0e : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04cb3f7a0b894f2dbf65ecd6fa6fc67b : SUCCESS in 6m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1839114ea08440798a6e57af65c4a439 : SUCCESS in 25m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/575ea2cea40143339c27ad7653264a60 : SUCCESS in 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3f0e974efc968ceaf33c4b662c2d59d12feba20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:55:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0103ad58374e138a00d6460429a6f2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e32e078b1bc4437a218684120785b42 : SUCCESS in 35m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a58da13ae4184ca7a7cd7929ee87f01e : SUCCESS in 32m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cd07350530474ab89e47b7279587a842 : SUCCESS in 1h 02m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dc7279cc31894fddbe68860ae9e03206 : SUCCESS in 1h 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2376d12729d14fa99869aa6d3b6feb3c : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9bf5c352c1e45389e04447f75be8569 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f06bfc6a07d46638b73b7392214e58a : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61aabe340cc74719bfe6c911734badd1 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/af4caef008d34dd7abe7a9f1ebe4621c : SUCCESS in 34m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a0c3b485643b4d0eb95b7a3aadf2f8f1 : SUCCESS in 36m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7789566acebe414fbc7edbc8527b6af6 : SUCCESS in 1h 17m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bb3e09427e0f4ba0819717478b5f9ba4 : SUCCESS in 1h 30m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f0fb9f0c793d4ebcb7491bd046e01de1 : SUCCESS in 18m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/641c4ee2c1584b3b84389570a2f38ed0 : SUCCESS in 40m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/52c8986052d34d0bb6d4ec72478fcf74 : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f68f90176d104d1b98df84731fe435dc : SUCCESS in 1h 21m 31s","accounts_in_message":[],"_revision_number":1},{"id":"762cb11c0fe4247755f3bfdf6c2ad1fd3791e0ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:55:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb8d1b67be3afc01ae6a11a9e5e3eb61fedbb653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:58:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecc727eda1e4fc593c8e580218618de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb080dbf50b9479fa11b3244c185bb77 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b10b7d23579c0d06aa4b75b456f42d4116951f9","revisions":{"9b10b7d23579c0d06aa4b75b456f42d4116951f9":{"kind":"REWORK","_number":1,"created":"2024-10-01 14:05:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/931020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/931020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/931020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/931020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/931020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/931020/1"}}},"commit":{"parents":[{"commit":"602a6f7d4e534b7bacf51bb1752c1550fe8ffd53","subject":"Revert \"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602a6f7d4e534b7bacf51bb1752c1550fe8ffd53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 14:04:51.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-01 14:05:04.000000000","tz":120},"subject":"CI: Fix aarch64 publish jobs on x86","message":"CI: Fix aarch64 publish jobs on x86\n\nChange-Id: I5c69d288a427d708b42deadac872a3a7368181e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b10b7d23579c0d06aa4b75b456f42d4116951f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b10b7d23579c0d06aa4b75b456f42d4116951f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
