)]}'
{"id":"openstack%2Fkolla~931195","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I219418411470bdeae9af84277f5150a623c05e69","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-02 20:21:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I219418411470bdeae9af84277f5150a623c05e69","subject":"erlang/aarch64: Move to openstack-kolla owned COPR","status":"MERGED","created":"2024-10-02 15:19:25.000000000","updated":"2024-10-03 07:51:43.000000000","submitted":"2024-10-02 20:21:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931195","cherry_pick_of_change":931194,"cherry_pick_of_patch_set":2,"meta_rev_id":"994b2441fc579f256545c8963e4dcae6bd9894f4","_number":931195,"virtual_id_number":931195,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 20:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-10-02 15:21:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-02 15:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-02 18:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 15:21:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-02 15:22:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-02 17:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ceb57d54e57b60da8c38ad8f5f19eeb0761fbef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:19:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.\n\nThe following files contain Git conflicts:\n* doc/source/contributor/versions.rst\n* docker/base/hrw-copr-erlang-for-rabbitmq.repo","accounts_in_message":[],"_revision_number":1},{"id":"c64540bb4bebcb6034a518ac1e537e9298687f85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:20:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c2db2f16f04989ec3b7682c6d95098893c294e76","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 15:20:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"33e6a9136bd4b20373f5c6ca63c8fbe5b674d6c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 15:21:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"852ece4db1d983dd315b405cc158eec7c9109952","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-02 15:22:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e0478813992e6ddd087b5b9b4b17a0813803cc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:21:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8e1cd6fa3044643887683e9aac29e2c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6cb869ecba8a42f694cb73a6b8faf9bf : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/21ffec37f07541e39c8b4bab78dcfcac : SUCCESS in 59m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/40339c9837054ca083bf64fb4f0d7e5a : SUCCESS in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d5a95f3b744fa7a3cea4e1278a00a0 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61664486ab7b44c18b8c7becf27ad92f : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b321d2df3474f14a050711c1b9029bc : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79f21e271984262a8651f7ddc195cb3 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0ecdd7c994498d8acc45237a06128f : SUCCESS in 3m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/95becd9215dd437682ec399cdf57635c : SUCCESS in 23m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a02b7483b9ba4692a3d2b22a927d306b : SUCCESS in 1h 04m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2cd3650c727a4007a392a947a923f063 : SUCCESS in 1h 36m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd997ad489fc4251a1163d415289b09f : SUCCESS in 19m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d6a85faf6233463a832b9a1780233e33 : SUCCESS in 57m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/355a9047303647c88d0090e89023556f : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7e2b144b9e54c0b80b0435c858f618a08fd3f808","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-02 18:16:05.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7dd3cab84c850a85fbf0c78d6ebefad1e1b74d06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:16:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4caf3ed17214a0b0619f538290e18809f2f16bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:21:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbcf15f3cee549a99e6488ba9004dbb9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e5aa5adfd79b438f854030225ce1793a : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3d340ea77524960812a8b2e778335a6 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cb51df0316d47b0a445184e15874bb2 : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832b1e44c71544b69f32cab0cff73b9d : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e9464eca9e43498d0d1d01131f9227 : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db5b090bd7564758b4807c72df43058e : SUCCESS in 22m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d53b645753c84e68af6e989e59dc0bae : SUCCESS in 1h 06m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20dc8f61a5434b3998c17abe6166f364 : SUCCESS in 51m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/417a79a4e32e4a1795fdee989c842414 : SUCCESS in 22m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1dc57944ec474097bf01ee2775bf8db7 : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c1838a0242c14cbba9076e3f5699a7b2 : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":2},{"id":"70b64255b989d4a64fe39c3ab5917241645f77e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:21:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bb4ad87c5ee277c62e298a48a09d7d9eb3079b49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:22:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dcd1b1f063b4565bcab45881b54327c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99223c22f9c940968f4e379eedf5c98b : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aa32d9025f74929ae0b93c084e98d4e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2},{"id":"994b2441fc579f256545c8963e4dcae6bd9894f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 07:51:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a152e4bfff48e5a90b0494092429ce\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afe1074ace50466480cdb0383ac6892d : TIMED_OUT in 3h 03m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98dd9ade15404b4386e8cc900d9e8876 : SUCCESS in 4m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b522658997164512bced827bd51aaa37 : TIMED_OUT in 3h 04m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1ef3764ab8f482494da709843961442 : FAILURE in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9066422a22a7ead9514a09fb2af991373dead8c8","revisions":{"fee92b93c022cdb7cce57b1a609a61386f1cbf8e":{"kind":"REWORK","_number":1,"created":"2024-10-02 15:19:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/931195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/931195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/931195/1"}}},"commit":{"parents":[{"commit":"b80b2be04942850bb261f99b0100e580b6cc6c34","subject":"Merge \"Revert \"Pin OpenSearch Dashboards to 2.15\"\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b80b2be04942850bb261f99b0100e580b6cc6c34"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-25 09:59:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 15:19:25.000000000","tz":0},"subject":"erlang/aarch64: Move to openstack-kolla owned COPR","message":"erlang/aarch64: Move to openstack-kolla owned COPR\n\nChange-Id: I219418411470bdeae9af84277f5150a623c05e69\n(cherry picked from commit 2480da6d987da338fa2688a31a33658da0abeb23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fee92b93c022cdb7cce57b1a609a61386f1cbf8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fee92b93c022cdb7cce57b1a609a61386f1cbf8e"}]},"branch":"refs/heads/stable/2023.1"},"9066422a22a7ead9514a09fb2af991373dead8c8":{"kind":"REWORK","_number":2,"created":"2024-10-02 15:20:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/931195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/931195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/931195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/931195/2"}}},"commit":{"parents":[{"commit":"b80b2be04942850bb261f99b0100e580b6cc6c34","subject":"Merge \"Revert \"Pin OpenSearch Dashboards to 2.15\"\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b80b2be04942850bb261f99b0100e580b6cc6c34"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-25 09:59:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 15:20:16.000000000","tz":0},"subject":"erlang/aarch64: Move to openstack-kolla owned COPR","message":"erlang/aarch64: Move to openstack-kolla owned COPR\n\nChange-Id: I219418411470bdeae9af84277f5150a623c05e69\n(cherry picked from commit 2480da6d987da338fa2688a31a33658da0abeb23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9066422a22a7ead9514a09fb2af991373dead8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9066422a22a7ead9514a09fb2af991373dead8c8"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
