)]}'
{"id":"openstack%2Fkolla~931257","triplet_id":"openstack%2Fkolla~master~I4febdf6b391bb7320ea30e1ca71fed81e332021d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-03 14:29:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 16:02:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4febdf6b391bb7320ea30e1ca71fed81e332021d","subject":"CI: Fix aarch64 builds on x86 after switching to role","status":"MERGED","created":"2024-10-03 09:47:41.000000000","updated":"2024-10-03 20:33:30.000000000","submitted":"2024-10-03 16:02:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931257","meta_rev_id":"93dc95179b4fd36dc7986bdcfae5d5ba3d68279e","_number":931257,"virtual_id_number":931257,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 16:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-03 14:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-03 13:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 14:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-03 14:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 14:28:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-10-03 14:29:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 10:18:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-03 11:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-03 13:45:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"290464ed6f1aaef3a785dd76ca1a5472ac938cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 09:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d19f8a5b262d48cc393361cb6e55648125b6619","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:18:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"598adb8fb3d116591cae18095499d29a08f32c9b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:51:40.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"0467262614f44b33d3621604a91a703dc4f1bfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64356c793a694f6ea7bd7865993c2e8a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c226087bb41b4a0fa5dd4e9f9de7c2c6 : SUCCESS in 26m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5c9ee2c8c2e04a5c963eb8c4cff8b013 : FAILURE in 6m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d40004aec842471eb805ce394468f758 : SUCCESS in 21m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fa79bc5c393f487facab0d2dea870bfd : SUCCESS in 33m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d700bebd94948ce856ae4525e6e95a8 : SUCCESS in 54m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f10f05021317466b9fdad396642724ab : SUCCESS in 1h 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c990a21c5e4dd9a26189327ab717dc : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b74346826d4447519e49b18c9139b6f4 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f23654760cc45de9dee5d8fdc288ec4 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9edac6bf54d445cbae2ba93db0cb646 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f474bd26aaf04dd0ad58b01cf82ec28b : SUCCESS in 5m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a757cac83f34fe299d34b23ce3529d0 : SUCCESS in 28m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2f14ea2b1ca04efd93e6994b05b2c5b1 : SUCCESS in 35m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e4b079e4c1eb4102860cdc08c8bb579a : SUCCESS in 1h 01m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7d826f20d82244d1aee233ec0d643dcc : SUCCESS in 1h 21m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9cdf725814b24f03abc68699332e6ffc : SUCCESS in 22m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fecb37fa5ff44ebaa8207705df153442 : SUCCESS in 24m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2446dbd6df08436aa15069d2e82a8904 : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f9aac5a086a4920a86e64f1ea3a2529 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ff5000963460131dbafbbc13ca0c3526fa0a31ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 11:25:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a5d6d91c12e460bc71801feb8dd75f76d1a42b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:56:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8484d25aad8547f38797e65ad16f3925\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8b7198095f7540afa56070cfceb07e8e : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d431b07cdc42470d9f0c0d33d85d14ed : FAILURE in 6m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8793f28439f34ffc98a55a68d6c8c16d : SUCCESS in 22m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb983d3ce46d4cfd9eb7095f82800e44 : SUCCESS in 29m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8670f331eb244edb8808f5b350328fd8 : SUCCESS in 1h 01m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3746b6ba98e044f9b9e83f426f5270de : SUCCESS in 1h 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8821907745fa4592b3b768c409157cac : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ba9c2b3252b4eb79c71d38d96bf85d4 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d54ac71beea3487e8a8d12db26bfe497 : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9abcb3924adc453bb5990ac0ca48948d : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e99f35a2fe4bcaa870b8b9c72637f4 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5f4b638243234a2aaf00f67b99c0faa4 : SUCCESS in 25m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/078078bacb254d738e0955a263b545a3 : SUCCESS in 37m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc17737b8a2649e188f4cf4570b7cbe3 : SUCCESS in 1h 03m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1f819ca2544b4e77b0be1bdce1fdd7fd : SUCCESS in 1h 21m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8401b9b3dc0d49b7a6426eb23ca6fabe : SUCCESS in 24m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ff599f26438c4cb49191541a33268fe9 : SUCCESS in 27m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18c33d824ba84e2691aa8e74542095ba : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d581f63d11cb4d8c96f962741358a6f9 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"294e922b8a4ef4f15d4ddc1fef1cebf7766e0f7a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 13:45:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad6e1124f2e5a8428ce9bd829c7c1b566dc0daee","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 14:28:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c6ca6e3225ea61d4a7503dc985faa8dc3253e1a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 14:28:48.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"278c8eee697a73e25f2fddda7264fc289c517814","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 14:29:26.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"174cbcde3e19043a0be68ed1c72106409170b906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:29:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64199f67af9499c4dc2f1a7804a3468b18e08992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:02:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2f513a199884a9e919f08c6b8a1b9e1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fdc74d0e2e984e3693371c928b3fef83 : SUCCESS in 23m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/79aa0d59080c44248409304ca1e4ddf6 : SUCCESS in 32m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa50295dbdb04ea2b52ba7af8e29a45b : SUCCESS in 1h 02m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3aafcb4eebe244bb98a359433c5d221d : SUCCESS in 1h 25m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39af3b648a240c5a191fb37aad1a993 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0381f64ad5184b478e809ea624a9ac19 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f3071f14afe44b1a4cea63254c5d9bb : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0cc171f1bd47c4b854eeb881951096 : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/281accd808634de39e9eb2d29b767f6a : SUCCESS in 30m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/91ec7561073b479fab5c6651404c775c : SUCCESS in 41m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e5e72ebf272e4b0dbb7047b4b490a3ad : SUCCESS in 1h 06m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f6bce3c7570f4e64bd87ddef57f7bd50 : SUCCESS in 1h 19m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af0b21bf5fd74668bc0b50a0bb90c6ce : SUCCESS in 23m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f876e9b8422f44c29d66fd81e82f2c91 : SUCCESS in 23m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fa5ac4e5eccf48969ca617b71dfd8b76 : SUCCESS in 1h 04m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3dc982f94626477cba363cd6f0a5c5c1 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d73af5c9c918909e577bfe11235d0d6c56053096","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:02:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1486e76877ae9fc0411ffddcf616d49eb98ddf29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:04:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4636bc3bee646f2b07629b941384edb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d94657bc754b428b69650254840548 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2},{"id":"93dc95179b4fd36dc7986bdcfae5d5ba3d68279e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:33:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9552f15e85a4485a8d7f77806b134131\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f94e12b7625b42f685ce55aaad414b61 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d637862f690848dca92445cd470464c3 : SUCCESS in 1h 38m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f92f7014e3e4866aa8c735136d89ba1 : SUCCESS in 1h 57m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbe0823256524280afa308ddc0352169 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9dfe620822ae4fe1b41a1b39684d197c : SUCCESS in 5m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/62fda45764f7434887623fdf5ef1ef7c : FAILURE in 2h 43m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/785bc62d1bbc4a468af90d11df71b60e : TIMED_OUT in 3h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39d328da32524aef5c1bcd3435f26869c9daa9cf","revisions":{"ae441689d8f9ea882742f8ae17e4e7bcc7c37d29":{"kind":"REWORK","_number":1,"created":"2024-10-03 09:47:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/931257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/931257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/931257/1"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:46:48.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:47:38.000000000","tz":120},"subject":"CI: Fix aarch64 builds on x86 after switching to role","message":"CI: Fix aarch64 builds on x86 after switching to role\n\nbf04bd90a15e7d3949214d15a8b8e25f2050bb65 omitted the --platform\noption\n\nChange-Id: I4febdf6b391bb7320ea30e1ca71fed81e332021d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae441689d8f9ea882742f8ae17e4e7bcc7c37d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae441689d8f9ea882742f8ae17e4e7bcc7c37d29"}]},"branch":"refs/heads/master"},"39d328da32524aef5c1bcd3435f26869c9daa9cf":{"kind":"REWORK","_number":2,"created":"2024-10-03 11:25:11.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/57/931257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/931257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/931257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/931257/2"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-03 09:46:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 11:25:05.000000000","tz":120},"subject":"CI: Fix aarch64 builds on x86 after switching to role","message":"CI: Fix aarch64 builds on x86 after switching to role\n\nbf04bd90a15e7d3949214d15a8b8e25f2050bb65 omitted the --platform\noption\n\nChange-Id: I4febdf6b391bb7320ea30e1ca71fed81e332021d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39d328da32524aef5c1bcd3435f26869c9daa9cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39d328da32524aef5c1bcd3435f26869c9daa9cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
