)]}'
{"id":"openstack%2Fkolla~931260","triplet_id":"openstack%2Fkolla~master~Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 14:26:11.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","subject":"aarch64: fix name of rabbitmq copr repo","status":"MERGED","created":"2024-10-03 10:46:37.000000000","updated":"2024-10-03 20:06:43.000000000","submitted":"2024-10-03 16:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931260","meta_rev_id":"3be68810dbd71f57e42814f8bf012eca97358d60","_number":931260,"virtual_id_number":931260,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 16:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-03 11:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-03 13:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 14:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 14:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-03 11:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 14:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 10:59:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-03 13:44:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-03 14:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f77ef4590f19bf3356cd2371ba82c70b90afeb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:46:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a83bf5e5231388bd150487fd608b8e762bc91479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:46:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c7af067860bfb7c5b28ebe4fd5f6e9ee4c12fc99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:47:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d7792382e7277e91bb362ddc03063a7fb1bb6d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 10:47:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f7f56195281dcef75201138d5c58cd670f0f5c9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 10:59:30.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2209c4efa8e43bf87e070555fc64e3c79f8493ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 11:26:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"0b0efebba3cdde747abfbcf9873879867d3398b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 13:44:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5916f3e05516250070c858ee5567065d0647aa32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:03:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e22a832162b40e798622657944dbd6d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd840db77dd242f9a6bdf406802bd643 : SUCCESS in 25m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8f2726d9bd4b47d0a0a72b775171e1d8 : FAILURE in 5m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c0a8f65ac3db4c38abaa93ca3cfb9de6 : SUCCESS in 19m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/843352c4012943c3b734cd9e653f2c49 : SUCCESS in 31m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/332b38184a8a4b3e89e00faf2c2b9e15 : SUCCESS in 1h 00m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4583bafe7ecc4e8989817893391821d0 : SUCCESS in 1h 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4286ea20f04fb59ba21d438bbfff2d : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61aeaa359e724a068790797056a7f955 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5729e0bd253246258a012391e4f53c64 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2400cad187bc4e228f183b1ba0bb983d : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6628cd9ff44b368c6dc3b494d5a11a : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/866d1638b9724b8f999c8815d74979a0 : SUCCESS in 31m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ec4967ab21f34899b02ad5a1a51b98e0 : SUCCESS in 32m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/42ebf24ca1d046afa351adddb7d6bb8b : SUCCESS in 1h 03m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/63ca74f20c82443da15570d0a3fcb954 : SUCCESS in 1h 35m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a84564f859be4515bd87452ca56ea905 : SUCCESS in 1h 43m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/661aae6c52944c328fdc40002523b0e3 : SUCCESS in 2h 34m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/18ee146a1c784da5975075feabb4945d : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c0ce75c964844dce819c8082c32101ce : SUCCESS in 1h 22m 54s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/664f10d05e544b50b93c8460bc3c54fe : SUCCESS in 28m 19s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f4d6149a0dd940d6b98e3799aa4f9f80 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1e07bb7b70b04d5ab2e925c2380baba9 : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c71469bfee6d496d83ca7a8305269721 : FAILURE in 1h 00m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b6bf43f588d0410ab2a15fc742cd6aa9 : SUCCESS in 30m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e80511836b5f47b4958ecc3d2e2931ee : SUCCESS in 24m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ee83cd8c7e8d4247ae399fb9c18c0947 : SUCCESS in 31m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59739d45681a444c86d6280032e7c2a1 : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/19b6704508444d61ad61fc07488e29a2 : SUCCESS in 1h 21m 24s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0ddca821a2ab4665ae706550b0532544 : SUCCESS in 1h 30m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/17fd8dbf90cf4fb6873a57c0f3531f41 : SUCCESS in 2h 17m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/af96ee1eb8a441cab5f7ef7d7edee8b8 : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/44b92122e87c4e8a8858fcf6c0683638 : SUCCESS in 1h 06m 04s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5b8fc3e4bf4b446794a55c6d58fa2caa : SUCCESS in 23m 53s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/02d27c0ef9c84ca6bb3aa8ea099a2047 : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dad846727a7a47ea8ec0bd3a47d7fc2a : FAILURE in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/04ee54354fd6489098abd72cef7c8a63 : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/416bb175fb044164897289c4c707b8fc : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c49f892c7d206e16782224d2da5a93ffb62a9af4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 14:26:11.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d3a5adacd991d187f2512d703f559d216e3bcac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:30:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"14836004a462801fe4759e525523b924d487e854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:07:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bbf6f87de444467b9624522a6826d03\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8fe1514e948c4fad96b40029785fca73 : SUCCESS in 21m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/987092ece4ab494482012f5b7f32a667 : SUCCESS in 29m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc82c290f16940efa2113b9985a1937b : SUCCESS in 1h 09m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f8950a6c0fc64027b1d19e3fcc97e643 : SUCCESS in 1h 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78fe68f9b0064876ad96c7ef2b26644c : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/320ba4781c274ff590c175a5e300d41b : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d40a22bb42c492398b829a2eb93672d : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe649d71af143f3b5b67b4d0b21f620 : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1f9fb04959e64896bd322d17db774f9a : SUCCESS in 27m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8bc1c679ec9e406fb80ba2e99b942933 : SUCCESS in 37m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c95cd3e8140641498e65993862e4cea8 : SUCCESS in 1h 09m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da3340e38ac9489e882eab5b2c5705d6 : SUCCESS in 1h 29m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/21a63f06a2cc4b0b8fab3984983a226f : SUCCESS in 24m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e7c325ce603c4026bad256eba569d4b0 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/11a70cc0a8b54e378477364d275a2d6d : SUCCESS in 59m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7962d3f910c64e9195b71e2119703fea : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":5},{"id":"54fdcf44cb37a8a53299c1dc8b1ac47521f82a75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:07:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c6d437396e03f07f2907f87430b49022dc74acaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:08:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f2f2c213de247e3a563b6f3bfa502f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0166e4f1004ae69ba50d61f54fc389 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5},{"id":"3be68810dbd71f57e42814f8bf012eca97358d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:06:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f170d0bb445d40e9870e110915c31cb0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34a5c69a1af24171a56846d330095ee8 : TIMED_OUT in 3h 01m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/63ca13f6817e4db9a0015d479228474a : SUCCESS in 2h 30m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0b86ad62c2a0478ea3b8ccd2dbd70ac5 : FAILURE in 1h 21m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9be2c7a65ef144bea0f4d2ae0240b3be : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38aebc8ad5c34b1fbfb1b65ffdf21c14 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f5dec7f5716f465baa73e0157a9464dc : TIMED_OUT in 3h 02m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49fcfffaba9149a1bc826b6a9b7cba3e : TIMED_OUT in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"672601a3e66c2d818c30192bebaf4229ae468aa1","revisions":{"f002bc375e10be560ce0556d884537f403ff8f0a":{"kind":"REWORK","_number":1,"created":"2024-10-03 10:46:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/60/931260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/931260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/931260/1"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"subject":"aarch64: fix name of copr repo","message":"aarch64: fix name of copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f002bc375e10be560ce0556d884537f403ff8f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f002bc375e10be560ce0556d884537f403ff8f0a"}]},"branch":"refs/heads/master"},"cfc01053d873d01bcc2f155d7b89a74d5b0d9427":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-03 10:46:58.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/60/931260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/931260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/931260/2"}}},"commit":{"parents":[{"commit":"ae441689d8f9ea882742f8ae17e4e7bcc7c37d29","subject":"CI: Fix aarch64 builds on x86 after switching to role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae441689d8f9ea882742f8ae17e4e7bcc7c37d29"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:46:58.000000000","tz":0},"subject":"aarch64: fix name of copr repo","message":"aarch64: fix name of copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc01053d873d01bcc2f155d7b89a74d5b0d9427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc01053d873d01bcc2f155d7b89a74d5b0d9427"}]},"branch":"refs/heads/master","description":"Rebase"},"0c17aba19d19c3d329d23367eed949676d39c009":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-10-03 10:47:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/60/931260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/931260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/931260/3"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:47:17.000000000","tz":120},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c17aba19d19c3d329d23367eed949676d39c009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c17aba19d19c3d329d23367eed949676d39c009"}]},"branch":"refs/heads/master"},"184f858ba54888beba18182b84b2d53b0873bcdc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-03 10:47:56.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/60/931260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/931260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/931260/4"}}},"commit":{"parents":[{"commit":"ae441689d8f9ea882742f8ae17e4e7bcc7c37d29","subject":"CI: Fix aarch64 builds on x86 after switching to role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae441689d8f9ea882742f8ae17e4e7bcc7c37d29"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:47:56.000000000","tz":0},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/184f858ba54888beba18182b84b2d53b0873bcdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/184f858ba54888beba18182b84b2d53b0873bcdc"}]},"branch":"refs/heads/master","description":"Rebase"},"672601a3e66c2d818c30192bebaf4229ae468aa1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-03 11:26:57.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/60/931260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/931260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/931260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/931260/5"}}},"commit":{"parents":[{"commit":"39d328da32524aef5c1bcd3435f26869c9daa9cf","subject":"CI: Fix aarch64 builds on x86 after switching to role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39d328da32524aef5c1bcd3435f26869c9daa9cf"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 11:26:57.000000000","tz":0},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/672601a3e66c2d818c30192bebaf4229ae468aa1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
