)]}'
{"id":"openstack%2Fkolla~931316","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 18:48:03.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","subject":"aarch64: fix name of rabbitmq copr repo","status":"MERGED","created":"2024-10-03 16:19:01.000000000","updated":"2024-10-03 23:28:55.000000000","submitted":"2024-10-03 20:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931316","cherry_pick_of_change":931260,"cherry_pick_of_patch_set":5,"meta_rev_id":"ef44d08cbc38ea36ecfb1f74a6409fe694e6b502","_number":931316,"virtual_id_number":931316,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 20:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-03 16:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-03 17:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 18:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 18:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 16:22:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-03 17:40:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-03 18:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b9e0aec4cd9b3080f0e7d4445f337d0f65470fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 16:19:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f771722fc570b6f0d73e12c39c92d9afed072301","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:22:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eacdc2fa285f62dc7c828f6ab67d4c3a792a874b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 17:40:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3086860dfabca9e05d1edcff24f9716c89da1015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:43:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c48e832a8e542b7a0aecaa5ae2b30ee\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0cec1ebd668d484ba9de49529c06fbbd : SUCCESS in 21m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0d5fbd0c87bc4d48ad78636f05c5803f : FAILURE in 6m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ffd0a5726d9046bdb6bb1e937a3644e5 : SUCCESS in 20m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/983a85f676c04d148a01130eba93bf8c : SUCCESS in 34m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/36d95cbe05d54730ad1c77ff2a371bf5 : SUCCESS in 1h 01m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f71e53e343b14647a4594d73d15e2a75 : SUCCESS in 1h 30m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54086e12658f46ff952682f7160dafff : SUCCESS in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1be30ba34dc49a49d3a53cc7b39f2ff : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e27dc528217e44688d14917640609a5e : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/186a26e6fba04ed2afe38227da15a44b : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e6f3d718d147fcabfbc4e031fcdc5e : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cb6045e351d74f4da6a0e0d6c997d7f6 : SUCCESS in 29m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c3eb2d75e9e94c0caca74fa5f2ec0fd3 : SUCCESS in 34m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5be1d96707344395b38fd9fb27ad5031 : SUCCESS in 1h 10m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/431213df15ee41ebafdad755ae2f7e50 : SUCCESS in 1h 29m 16s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/33ff5c6ec1c1456f80f907fded875d36 : SUCCESS in 1h 44m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ec68bbf96036433abe907c170bd96dfe : FAILURE in 1h 17m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/69c4d566caa84510b807640f53ed65b2 : SUCCESS in 1h 13m 19s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f49a42a396714519872292c660935f82 : SUCCESS in 1h 24m 28s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/538c2ed138354ba1b05bbef7eab8c3f6 : SUCCESS in 27m 34s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/512dfafca11b45768fe1f4e1cf3a668c : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ed90f9eb235f48b0ad5f5ad3f397a510 : SUCCESS in 1h 14m 01s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/652117f52de24fa09654381be12f53e9 : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c9468dccfc5f41d19346245de5566c68 : SUCCESS in 31m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0f28d0de64146eda531ac1f035f8acc : SUCCESS in 21m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6115b13f8c5f4cab9bcd15f558303592 : SUCCESS in 50m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18ea5269bf744a8b82b30236a2eb8008 : SUCCESS in 46m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0db05a9dfe30448d87bad51f50587160 : SUCCESS in 1h 14m 58s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4cae629e5a3a4b36ab21a7c9006cc080 : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/70f2dfce8c0e41b6a11058865c1a5e69 : SUCCESS in 2h 16m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f68271a2d0484a09938bd99cbc80d745 : SUCCESS in 57m 22s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9185ae06485e40e8afdd607ba209f35e : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/97083578a88d40898e9f3fc59d2294b5 : SUCCESS in 21m 45s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2f7d91ec66c047bb945648206cba51ba : SUCCESS in 44m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/01adb1b3ffe44dff9fdeaa4f20094a02 : SUCCESS in 58m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/60128425d6cd446ab0cb0eb2f8118e5e : SUCCESS in 1h 09m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/29c9daff9d3f41b7893f2f38f6fb9c8f : SUCCESS in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc0b5985e2001ff7fbffdcc9f92c26e3d4ceaf35","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 18:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dbb387982c1a22d427de29951339ec402876378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:48:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb68322a2d2e6a87adb96e785ef3e42e92aeab0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:08:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627dd70d606249168f808d6097f4febd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/498a2dbf6b5c472a934b213ab6aac5f2 : SUCCESS in 16m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/07ef97f181f747b99b207dc937f89528 : SUCCESS in 28m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5baf325359be4708b14da16b30540dd0 : SUCCESS in 50m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f408a10b396a4b39ba3cc4423545e77d : SUCCESS in 1h 09m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2801d89b1f04f53b0634ecc61c5bd1c : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea2c682950504cea96c7f6eaa8829708 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b42af3fc38ed479ea05e4dab5c99f074 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a981d424aa944474a2e3be440a58455e : SUCCESS in 4m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0c2b9322dfea480086e17d6220c81856 : SUCCESS in 28m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/20d3342607434fd19ca0a1f6d5d5c0a0 : SUCCESS in 30m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/77b5e045d18c4fdda60468ad53168522 : SUCCESS in 1h 04m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3e2b9436f03d4674855c9d536da0a563 : SUCCESS in 1h 16m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e5f4953177d4053861b88b6f7e03250 : SUCCESS in 20m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/50e5658691ae47818a35d09f34c13b9a : SUCCESS in 46m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1e0c092c539946939d086d66389e0053 : SUCCESS in 42m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/26f06356789845e5acc40ce3adc67191 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2ccb7a9ba68b883083877a6651a79336800de2d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:08:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f99dbe9cc90c778d819989df448dce9c94e0b41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:09:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0dd447ec004c7b8425bcabe5a185ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb45ca5105c147818a47da3ae99a9a3b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1},{"id":"ef44d08cbc38ea36ecfb1f74a6409fe694e6b502","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 23:28:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae2a9756338477abfae5f0cc48b1e3a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e723c07235e458c92bf39a28fda6e6a : TIMED_OUT in 3h 03m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/49ef3928fadd4f84b0305aea0434c22e : SUCCESS in 2h 14m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d30b0872354747d9a99d8f5cdc971459 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02c5867d5f834bfab0904d02bb555ee9 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4537f2b5348879515d639716526bfb67c695bb9a","revisions":{"4537f2b5348879515d639716526bfb67c695bb9a":{"kind":"REWORK","_number":1,"created":"2024-10-03 16:19:01.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/16/931316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/931316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/931316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/931316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/931316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/931316/1"}}},"commit":{"parents":[{"commit":"9a0914d10d0e83129d1723cb7105849a1ffed3cb","subject":"erlang/aarch64: Move to openstack-kolla owned COPR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a0914d10d0e83129d1723cb7105849a1ffed3cb"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 16:19:01.000000000","tz":0},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n(cherry picked from commit 672601a3e66c2d818c30192bebaf4229ae468aa1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4537f2b5348879515d639716526bfb67c695bb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4537f2b5348879515d639716526bfb67c695bb9a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
