)]}'
{"id":"openstack%2Fkolla~931317","triplet_id":"openstack%2Fkolla~stable%2F2023.2~Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 18:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","subject":"aarch64: fix name of rabbitmq copr repo","status":"MERGED","created":"2024-10-03 16:19:38.000000000","updated":"2024-10-03 23:32:24.000000000","submitted":"2024-10-03 20:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931317","cherry_pick_of_change":931260,"cherry_pick_of_patch_set":5,"meta_rev_id":"59ebafa04d0a5191d68f7b53f792c76e86c34d0a","_number":931317,"virtual_id_number":931317,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 20:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-03 16:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-03 17:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 18:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 18:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 16:22:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-03 17:40:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-03 18:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2a8d51057322e14f0489e6ebfc684781ac75b71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 16:19:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cee6f9db8f5f0abd4a059028acfd1f7593f0e75f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:22:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1298010b1273e46686c1a19ad644409c5ec53363","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 17:40:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"58bf7323fd39679454c5d0313a97e4866b964c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:41:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edce38620c874d75806f7fcca095f240\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2d6dfca13c844a2796eeab607c718c9c : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/18bd1176da7642eb8a04a4978c93dad7 : FAILURE in 9m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/32346d2f33b943709742b1f612f6bd41 : SUCCESS in 18m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/41ab99c677f648fb8c8fb0e37b846c5a : SUCCESS in 33m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6c0672cc7a824b44ac3181957090dbe3 : SUCCESS in 1h 02m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2ac0e2d0b41845f1bd24346835ecbb38 : SUCCESS in 1h 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d711289f484f85b6a1adeec75e91c6 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f1b8cae9b3640b18cc24786d3266eb2 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4058b0e031af4544868a4e75cce8db05 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8ccc2125788440bb951b353afc4a2aa : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3cde8ccc368488ba4b4952dbdf10499 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58087e449796428c97ec4386c4dae3df : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9020527ffcb4069822dc0f273ffc2e1 : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8bd275cd57d2491d832fdd904c7ffd89 : SUCCESS in 32m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/81d1b96666c7499592507e7064bf1f78 : SUCCESS in 36m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dddb3f8196da405b8f2d78e2222825d7 : SUCCESS in 56m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ac9d06e9d53e443aa691c8d9d1ade24a : SUCCESS in 1h 22m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8bd52576e48c4c418ae0da5c18469fca : SUCCESS in 1h 45m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09f119b24b0a4cd2904c04dc52b4e4f7 : SUCCESS in 2h 21m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/34f258d44c884802955cb3d29a09ad6d : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f346fe5951d343c7a279e1a253f95f64 : SUCCESS in 1h 20m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/811f8c9c2d814803858f17438ecf7fae : SUCCESS in 25m 34s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/34e2415d018e430f86dc511a844f6dba : SUCCESS in 56m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5ff50eb45d204ba884a0dfc9b48a1557 : SUCCESS in 1h 12m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1c3e0a965bbe4f288961667f53d4e95b : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3f046b5a95ee4e2ea645871bf591d79b : SUCCESS in 35m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e987fbfcbb9040caa2786baa599ddd4f : SUCCESS in 19m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0947fcf182204db9927527b01d966155 : SUCCESS in 1h 02m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5cd2d64edfd741939a030debc8226060 : SUCCESS in 59m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e739ca0590e845819066e30aeaeb5a77 : SUCCESS in 1h 30m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a003cb6494e54c6da9943850e86583a5 : FAILURE in 1h 07m 11s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/908975bed44d46e28772e320bd53eb20 : SUCCESS in 2h 16m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0e057d609f974e3d9a173da05334515e : SUCCESS in 1h 06m 02s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/03705fd3f10645b68970a46f007e4636 : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cfbd0d4097194423b6290680a8cc985a : SUCCESS in 25m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/67f449935ba34843b811981fec8bcea4 : SUCCESS in 51m 50s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/37258f9339334120ac14554328e9ad6c : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cd4835ad21694b6fa7796cc967978a76 : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5ab44051e0c34cc4b149a1be9f06a518 : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c007a69a440ae16d8f1edb5d166b36ba60e9dc1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 18:48:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94371ece362bc1158a4703583b63370ca3f1b2f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:48:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8953028aec5efee46b20cbfd6a796e9229be48cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:36:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bae27397d484a6e836e27b39eeb435d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/489d0c0f68934249a204e837dde11e2e : SUCCESS in 20m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3caaa0f65fdf42e28b6ab28559276850 : SUCCESS in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0efcbcad42044b97abeb15ffec8983d7 : SUCCESS in 1h 07m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cb4b82b897e345318b3e36e1303378b0 : SUCCESS in 1h 41m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c54c6ac99f4fbcb178540659a538dc : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d06c295b0da94de79e6e63bb006c0429 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a170cda255e4505a790723bc542993c : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2148bc9347f4f699ca81d389a98efd6 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2521ff3a0ac4c11a5ac4ec6f1ab95fd : SUCCESS in 3m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/61d508c855d54aa7a0796225caa5b37c : SUCCESS in 26m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/02b30c58c1ea4b4db9bf05144240b300 : SUCCESS in 28m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd1c6cccd38942aba052a427d3fe77bc : SUCCESS in 1h 03m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/47f11c343db14d429e9790a39a988123 : SUCCESS in 1h 17m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4af07c5f6b442dd93c130875aada3a2 : SUCCESS in 18m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6de72354e0ae4f40b137d05730cda5da : SUCCESS in 59m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9065bd43d572436e8d547add7c5209df : SUCCESS in 43m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6823c9c2d9474a63bf75332906bfa2ac : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"91c0a38ab2be13054d3e0afae1d39be48a16c5ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:36:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6fa127b6d47c3cfaaaa45e031acd52e712dd181e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:37:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeed536e172d4fc485d3a656f4af519f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bc9f076f2b46d7bc2cf2e07b963c31 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"59ebafa04d0a5191d68f7b53f792c76e86c34d0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 23:32:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7199d0c9373847dca3e1453a4f7373e5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff943fd436014ad783189ed9433fefd3 : TIMED_OUT in 3h 04m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e871a53dc2b04042895fff9af75034c1 : SUCCESS in 2h 32m 03s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0138b0aecad8410cb8f9825742fbd096 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fc1b08c7af84fe9bf882f80a00072b5 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b718a98deb32452bbb028917c51b7150 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/95c894dfb30c4c0784fb7128c84405a6 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c913df4be4c664640137d2956945bb13be2bf633","revisions":{"c913df4be4c664640137d2956945bb13be2bf633":{"kind":"REWORK","_number":1,"created":"2024-10-03 16:19:38.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/931317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/931317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/931317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/931317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/931317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/931317/1"}}},"commit":{"parents":[{"commit":"56e773e5cbf4eba0f209d6b9de3002687dece6c1","subject":"erlang/aarch64: Move to openstack-kolla owned COPR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56e773e5cbf4eba0f209d6b9de3002687dece6c1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 16:19:38.000000000","tz":0},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n(cherry picked from commit 672601a3e66c2d818c30192bebaf4229ae468aa1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c913df4be4c664640137d2956945bb13be2bf633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c913df4be4c664640137d2956945bb13be2bf633"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
