)]}'
{"id":"openstack%2Fkolla~931318","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 18:48:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ieb413f7a4bd1989d7f29309368b6b218d1395e0b","subject":"aarch64: fix name of rabbitmq copr repo","status":"MERGED","created":"2024-10-03 16:19:52.000000000","updated":"2024-10-03 20:37:53.000000000","submitted":"2024-10-03 20:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931318","cherry_pick_of_change":931260,"cherry_pick_of_patch_set":5,"meta_rev_id":"ad3e2764d66f21015418be7f8d0614780b3aa5b1","_number":931318,"virtual_id_number":931318,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 20:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-03 16:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-03 17:41:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 18:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 18:48:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 16:22:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-03 17:41:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-03 18:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e9dc11ba55ac286ca131c03df8f09899b89fb48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 16:19:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"453283fa52d5a4a7a187e3d103ddcda0ce5c3fb9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-03 16:22:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"271e9400cb64009103feaee1b4640c66f57204d3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 17:41:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05f5c7947b9667b0b3839e6d088e837f97abae7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:01:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e980e21a77c4472fb48e82a8b2cbe0da\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2ea6b2d234794b6dadf2fe128add599a : SUCCESS in 24m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/084d1200632342038b0533d8e112663b : SUCCESS in 1h 00m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b261c7e84b44ef4888ec99c50fe6319 : SUCCESS in 24m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4190abdfa5426d9ef00165d9eddcec : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94025c9eb000489dad599e3cb8ffcc61 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70e52a07cda24af79a5e1ea54174a373 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e08a57cf0c4fd5856d9fd002266508 : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b9a40bb3d9634eeeb737c59d4b0790a4 : SUCCESS in 23m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76d4815d0fea4b89bdc1022e5b036e4c : SUCCESS in 58m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e71cd9f0c37d41a8af79aa85f4ac2b9e : SUCCESS in 1h 35m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/decd87682072419e9cf426853ac13c77 : SUCCESS in 17m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/df2d29fdde884384aedcfd9bb1a1723a : SUCCESS in 51m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c83c9b02cd14e27b6afb4a0135d2cd8 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7cf0c2b27b81a75e8d34955d4f83c45361fc0e23","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 18:48:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69ec1f939c9aba5612210734b7ee818ffb87bc52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9353379c2de201a7f56cda5f2d7538a78978426b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 19:39:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e5dd1394c2461394140f736d1d75cc\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1912d46564f8464da1709909b6e2a702 : TIMED_OUT in 3h 03m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e08ff8f152c41c9a332f49815af9426 : SUCCESS in 3m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5d17b0bbfdf646daa44445f5740717da : TIMED_OUT in 3h 05m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4665ff7ca4604adb9102967f779eeab6 : FAILURE in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4f7337ea6ffc2afc483da730b842dc41640e92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:36:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6f166e3efc4e6d86e51a34988b535d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/375df3b1fff64a518c12dc6a7aa8393b : SUCCESS in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba315e647e204cc7b26e311afed61789 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79ab89b6e7a4e5a9d795e6725e896e3 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439db4164c744d8b9e20e759b0f98d85 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/840126ee187748ce8ed0ea46c5a88d0a : SUCCESS in 24m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e5a0544e92504520829abfa5579c779d : SUCCESS in 1h 02m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b75549a513a6445da6d0703af8033b55 : SUCCESS in 1h 07m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/66e2bec7ad1549dcae4dd982d6eefb45 : SUCCESS in 20m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6491be29cdde49dbab644152a00c41db : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/519c08c55a5344e9a2dacb472b198c1f : SUCCESS in 45m 27s","accounts_in_message":[],"_revision_number":1},{"id":"37543e6cfb7d3c7f76a2d10c395f83f292d344f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:36:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad3e2764d66f21015418be7f8d0614780b3aa5b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:37:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486a4034936c46dd92b596bdc953eaf2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054fdc6620b64185a8b578e7f8e6cfce : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26eda6a27341ae5c5f9e805a2aee274fd0c6f803","revisions":{"26eda6a27341ae5c5f9e805a2aee274fd0c6f803":{"kind":"REWORK","_number":1,"created":"2024-10-03 16:19:52.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/18/931318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/931318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/931318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/931318/1"}}},"commit":{"parents":[{"commit":"9066422a22a7ead9514a09fb2af991373dead8c8","subject":"erlang/aarch64: Move to openstack-kolla owned COPR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9066422a22a7ead9514a09fb2af991373dead8c8"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 10:45:37.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 16:19:52.000000000","tz":0},"subject":"aarch64: fix name of rabbitmq copr repo","message":"aarch64: fix name of rabbitmq copr repo\n\nit should be the same as in repos.yaml\n\nChange-Id: Ieb413f7a4bd1989d7f29309368b6b218d1395e0b\n(cherry picked from commit 672601a3e66c2d818c30192bebaf4229ae468aa1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26eda6a27341ae5c5f9e805a2aee274fd0c6f803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26eda6a27341ae5c5f9e805a2aee274fd0c6f803"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
