)]}'
{"id":"openstack%2Fkolla~931336","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I3eea308a0a70e47dd3b185eece106185b25616f5","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-10-03 20:04:55.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I3eea308a0a70e47dd3b185eece106185b25616f5","subject":"aarch64: Exclude syslinux from ironic conductor packages","status":"MERGED","created":"2024-10-03 18:30:41.000000000","updated":"2024-10-03 23:50:57.000000000","submitted":"2024-10-03 21:34:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931336","cherry_pick_of_change":931263,"cherry_pick_of_patch_set":4,"meta_rev_id":"d0ac1b935ae708d2d8f3208e34bf478270ca9610","_number":931336,"virtual_id_number":931336,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 21:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-10-03 19:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-10-03 20:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-10-03 20:04:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 19:14:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-03 20:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7aed959de68ac336e7daf0b701294bd1513f1e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 18:30:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a1d2755301b6221ba72f4b84baa8c5b044387cc2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-03 19:14:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"61eff948ee4522584e005b1cffa92ca4179e5a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:02:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6198710e27194d70a343343c1eb7b4b7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/661f39ddde64474d9058385c844a2c2e : SUCCESS in 23m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2e6dcc0a18a941529390b4df23bdabfd : FAILURE in 6m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cc2ca762299c461dbe8b2526c1382e99 : SUCCESS in 18m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/87ebd39501d54068ac76635ec75fa972 : SUCCESS in 1h 10m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5345ad2c53e4464f844e87946ba6b8e0 : SUCCESS in 1h 08m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/da5f5702e8064350b0eab8688a2c1f5f : SUCCESS in 1h 27m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb17983728b4f91b46a8b57a4f1e347 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab66e89198e4447085eed204d4903a7b : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7b18a1749114decb5a7a480f76913ed : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45682ad475b9455f87095bd81bfc4cf6 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e07403e4ec4dba9ed6f2384adc3f48 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8ec6a513f456444fb6e53f2f18855927 : SUCCESS in 24m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/54ca8452f3484d8a858589cfbc4c94cf : SUCCESS in 34m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3de07eb86262450b8810b3d27b933089 : SUCCESS in 1h 05m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ea0b3947d3f04555a532efe3c1b1f9ea : SUCCESS in 1h 18m 29s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/28824ac3ae4b47d18554ff2773cc53d2 : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/19fb68fa249146f2a3a9523ae56e1147 : SUCCESS in 21m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d109a667006f443b92cc4a945d105a55 : SUCCESS in 48m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f0b3d16ffa784b4390105c8542ede00f : SUCCESS in 47m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5a1edd29a57a4bf1a61f52b68164ca46 : SUCCESS in 1h 15m 31s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f5ef145cc1e04db180a4e490ed5fd538 : SUCCESS in 50m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63b32e3a5c4bff7b63cd6b045b5cb9b9ecbc159b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-03 20:04:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b185ff0d1ec1695affb7d89a2f50e681ce1d3607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 20:05:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61e7131dc4b7ace738c438cbb037a64112746cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 21:34:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752c8ebc9292486293bd9feefab41642\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a6d5dab3503343bca48e7364fe2f7c08 : SUCCESS in 17m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f3a3c3c96c804c63a6041b32a96cd14b : SUCCESS in 26m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/347d20671ad3499eb4a2e908bc359d07 : SUCCESS in 47m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2d839ad27d3740848712e23519570124 : SUCCESS in 1h 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d7b3aa0c204759837dc6a6b17dc863 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bc7822984cb48749b35e463c79ed3da : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaab400d97594817b315a612df808974 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb8bb2bc9c34651ba69a2b10d180971 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c28a2a723fbb4eb99570d98d921d45a4 : SUCCESS in 19m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8bd884f54b3b4f2698b0ec0c3c0bf579 : SUCCESS in 30m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d7caaf22bf27486c885d310bdd0abeb6 : SUCCESS in 1h 06m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e2804769d53f4ee3a546eb04a18d5d8a : SUCCESS in 1h 23m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c5968dbe0f5f4a9cbeacdb57b4dbed60 : SUCCESS in 18m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6e8f4d5a89c4494cb4336acd14ec6813 : SUCCESS in 46m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cdb1a4105cb649b1a385ed786aa0bd17 : SUCCESS in 44m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3dec1cb7e9834e0694530c5c1d480b90 : SUCCESS in 1h 17m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5bd72b07601700d0f88790d9a49c25db3b7ea13c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 21:34:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e0af89f63ba261e464dbd6c379de7d91e5f0645","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 21:35:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ad448bbde248a09fe48bf42ce3123f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1699988265ef46419022e7c23e954360 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"d0ac1b935ae708d2d8f3208e34bf478270ca9610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 23:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03cda5d5d114e668772439dcb9243c3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/acb72a0203ef472292d20af098ef2fe0 : TIMED_OUT in 3h 03m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7eeb78b827644a1ba6e1de496e0eb93a : SUCCESS in 2h 09m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dcc782af19d45fb8eed530bc3bcdc3e : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af545c0e7c3248a195ee62f0a2f33661 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97a6d193be081d4d37e38db7d0dd8b4fafcb961e","revisions":{"97a6d193be081d4d37e38db7d0dd8b4fafcb961e":{"kind":"REWORK","_number":1,"created":"2024-10-03 18:30:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/36/931336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/931336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/931336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/931336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/931336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/931336/1"}}},"commit":{"parents":[{"commit":"9a0914d10d0e83129d1723cb7105849a1ffed3cb","subject":"erlang/aarch64: Move to openstack-kolla owned COPR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a0914d10d0e83129d1723cb7105849a1ffed3cb"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 11:00:25.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-10-03 18:30:41.000000000","tz":0},"subject":"aarch64: Exclude syslinux from ironic conductor packages","message":"aarch64: Exclude syslinux from ironic conductor packages\n\nThe \u0027syslinux\u0027 package is not available on aarch64, so it is only\nincluded in the ironic conductor packages for x86_64 systems.\n\nRelated-Bug: #1749780\nChange-Id: I3eea308a0a70e47dd3b185eece106185b25616f5\n(cherry picked from commit 7b4913e6f8ee64c462727ee495a3bc8522331f91)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a6d193be081d4d37e38db7d0dd8b4fafcb961e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a6d193be081d4d37e38db7d0dd8b4fafcb961e"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
