)]}'
{"id":"openstack%2Fkolla~931476","triplet_id":"openstack%2Fkolla~master~I860e7bb5424ba42e44e1eca4852407d0508387a3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-04 13:15:49.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-04 13:15:49.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I860e7bb5424ba42e44e1eca4852407d0508387a3","subject":"CI: record kolla-build output to a file","status":"ABANDONED","created":"2024-10-04 11:08:27.000000000","updated":"2025-02-21 10:29:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"6ed241d236ee2b44b63ad105b9939fe31b5493a3","_number":931476,"virtual_id_number":931476,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-04 14:54:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-10-04 11:50:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-04 12:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb76624cd2900105633ba28f17f7e037575b8c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 11:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce808752f8f241da6b6db4122010c6a83e72a7e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 12:32:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71bf221b11564f17b7351423721ee129\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fd21257687a2442cbef58a2077534acf : FAILURE in 5m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a72cd33732a847309295ad8c1748024d : FAILURE in 6m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/796ea54d4fdb4f4dae01fb43a6157202 : FAILURE in 3m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e93d0080b42248b792155a2abd66001f : FAILURE in 3m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef0cbafd102645009a9b346952a0ced3 : SUCCESS in 1h 04m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf6e8da6d65d4c62a21361d676173edd : SUCCESS in 1h 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b62f281bc641d79b8bda8ab4e26caf : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c546cf16ba4b04a231d0d53a80622d : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66a4d93ecbe24be5a8c164b84287f845 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b9aae6f2ce64e73b91d78badaac4139 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684561816a184e8f9c449f17f74fbc13 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8242f4c8797c412aad11b90d83a3d934 : FAILURE in 7m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/937dc4c8ca2d41758f7ca7680f2bbca4 : FAILURE in 7m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7e16674628e241b581c769276c9ccafb : SUCCESS in 1h 04m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/277597ac5e2747c2bb4ecc2ab1e876f4 : SUCCESS in 1h 18m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02d7c132f0e54f818401f0472d4fff4f : FAILURE in 3m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ce2ac95179b147839db136643f1867e0 : FAILURE in 3m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/824660a3e1774142b9cca600c5c8dae8 : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/09c5dbef57284503ae90c4a39799d7e3 : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e02106b94712ef7c6c0110e744c73c33a120f915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:01:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d05d50e8569472482f2ba392d211fd9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ce85733eccde4844a3f4ba86a2b2b92f : FAILURE in 4m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4aa66e34d70a4aa1bf39f50ffa0a089a : FAILURE in 3m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2d78a7eb8f8419c9a5a28d588830c15 : FAILURE in 1h 50m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/191ac23f786b4feab30204a2fbac8962 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4701465d39624e1c88f5652c7431023c : SUCCESS in 5m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/12392f6a96fc4b239412c9c7f89962ac : FAILURE in 8m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/151acfce7bd14fb8a077d2fda09d2321 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"414e86fcd7c6c9d873cebd3807187bbaaba46b5b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 13:15:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5f30622febbf36704d6f727da647a0a8dc419adf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-04 13:17:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"534daeac77fb816745d5ce280b0baaa2c104ccfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:54:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bb60f7a10a44d25a61ba553d20f89d0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8c7d7040eccc4b2db7e7104b338d418e : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4362e73953d14eaf96e2669e89a59d24 : FAILURE in 5m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f2f22c6eb1154f778e4499c1802ddf95 : SUCCESS in 22m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d8691684b0f741baaee280f0841d76ce : SUCCESS in 28m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4a0cc0c69c1c49169cc8d0ada9737323 : SUCCESS in 1h 08m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/69a946646f784aaab6b96eeac0fbcd5f : SUCCESS in 1h 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fedd52f50c4b04a4ed1dda0701db6a : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1e8b7bd26b4e7885a1f771298d0597 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c3b014dbb4143439d439af264bdf9c2 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abc5d47737d94da09f0eb6fe6aaa4f62 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e10b3d94cf403ead4e7b8446d04c42 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d1f33316061e45e0b30ed082fa99b937 : SUCCESS in 28m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d16c26e9727642eb85faace46d307ffe : SUCCESS in 35m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4014f05c13554d198e38b28092e7c811 : SUCCESS in 59m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d25dd9179c974da5b23f41b1c4e33525 : SUCCESS in 1h 31m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc29a8150fe34a91b9eb23a6caa99464 : SUCCESS in 24m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fb3aff81727b433781e7e66ce8b2f72b : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/303f1c06894347da82ca76981c6d4fcb : SUCCESS in 1h 06m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e206d950943a483cbad35531d9201dba : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":2},{"id":"88d05744e2a63cc75755e03501d8a3d4c17afcf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f5e465a0acd4dea8fa38200403eaef5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/999c54cd8a6a4ff6a260bf8420d79514 : SUCCESS in 2h 28m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/50b49b95e21242d3b7b3281a6eea9d9f : SUCCESS in 1h 39m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/441cbccf382849c39dbb9f27c518546c : SUCCESS in 2h 32m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8db2da086d7a4e4db1d3938068eb7c76 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/613c998a172b4404a29f009ae7900a11 : SUCCESS in 5m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f5eb27557264c8180bd00103cfe5e58 : TIMED_OUT in 3h 02m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a474fd1657f54797948fd2cf71dcc759 : TIMED_OUT in 3h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ed241d236ee2b44b63ad105b9939fe31b5493a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 10:29:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbddf3e9f3f579f5ea7059e839be0e7f1f6202d9","revisions":{"613c670467610f4dcb8581bae39f181a3a5b7fea":{"kind":"REWORK","_number":1,"created":"2024-10-04 11:08:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/931476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/931476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/931476/1"}}},"commit":{"parents":[{"commit":"7b4913e6f8ee64c462727ee495a3bc8522331f91","subject":"aarch64: Exclude syslinux from ironic conductor packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4913e6f8ee64c462727ee495a3bc8522331f91"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 11:08:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 11:08:24.000000000","tz":120},"subject":"CI: record kolla-build output to a file","message":"CI: record kolla-build output to a file\n\nChange-Id: I860e7bb5424ba42e44e1eca4852407d0508387a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/613c670467610f4dcb8581bae39f181a3a5b7fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/613c670467610f4dcb8581bae39f181a3a5b7fea"}]},"branch":"refs/heads/master"},"bbddf3e9f3f579f5ea7059e839be0e7f1f6202d9":{"kind":"REWORK","_number":2,"created":"2024-10-04 13:17:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/931476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/931476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/931476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/931476/2"}}},"commit":{"parents":[{"commit":"7b4913e6f8ee64c462727ee495a3bc8522331f91","subject":"aarch64: Exclude syslinux from ironic conductor packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4913e6f8ee64c462727ee495a3bc8522331f91"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 11:08:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-04 13:17:23.000000000","tz":0},"subject":"CI: record kolla-build output to a file","message":"CI: record kolla-build output to a file\n\nChange-Id: I860e7bb5424ba42e44e1eca4852407d0508387a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbddf3e9f3f579f5ea7059e839be0e7f1f6202d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbddf3e9f3f579f5ea7059e839be0e7f1f6202d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
