)]}'
{"id":"openstack%2Fkolla~931628","triplet_id":"openstack%2Fkolla~master~I0ce9e907594fd160a08ad7929d5bc4bca9abf536","project":"openstack/kolla","branch":"master","topic":"kolla-2024.2-release","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-25 06:20:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-25 00:05:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I0ce9e907594fd160a08ad7929d5bc4bca9abf536","subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","status":"MERGED","created":"2024-10-07 09:37:38.000000000","updated":"2024-11-29 11:03:57.000000000","submitted":"2024-10-25 08:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931628-kolla-2024.2-release","meta_rev_id":"46678b7e0407238fc8e46e221b1ab51f04add3a1","_number":931628,"virtual_id_number":931628,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 08:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-25 06:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-25 00:05:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-25 06:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-25 06:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-10-25 06:32:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 09:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-07 09:55:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-07 12:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 08:09:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-10-25 06:06:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"ff53f6c014db97b637bd62bb354f3d7db763ee12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-07 09:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b74f75c3961bc383ebadb4b64de5a5e96bf30b51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 09:49:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9dda92e90bf471fbb44e2626d078862\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ca3fac05c76d4089ae2fc56900531faa : FAILURE in 3m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be898a27284243ae93edc50e54885107 : FAILURE in 2m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d708dd890c7941bfb7a77f93a3435b80 : FAILURE in 9m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b32bb055478541eeb4edeabe7645c1ec : FAILURE in 3m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b35caf7ac7b54c868e10d934bc9ef354 : FAILURE in 4m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8fb27565bc4448efa44ba47566e53238 : FAILURE in 6m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f0d9680ac2a84c5baaa930e87946fc14 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a86e43b853e0c0453eb1974e29594f957a41e87b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-07 09:50:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"019d4e47675450e9b28d8ff0ca443c979c24757c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:14:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7e1cc57390943f4b89b7d4c6687798a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b4eb40ebd1d649a3a9ee19b9070e6090 : FAILURE in 6m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/da153a4fff5a4cb1bf13ba671410233e : FAILURE in 6m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a0d34248f174b49bda8fa57cd12b696 : SUCCESS in 16m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9699f3434b57448590eb94c1a836104c : SUCCESS in 22m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e32493fb73a246939042df94481ab1e6 : SUCCESS in 1h 05m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ab3e87b8bf8b4f5bbf4693114109abb6 : SUCCESS in 1h 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171458650cb44ace9fd45a04d92e8bae : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1dd351f374c457e9a11628b8b7a13bb : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cf3cb0d8e8f4e148aa7df8566f96ca6 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2ac3d76a6cc4accb73928be3fdd983d : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe689896468848e0ba88adf6782786ba : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddb7eb9ff2da473499aab37df0a3a7a3 : SUCCESS in 3m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bb6da3493ddf4bf7a3c62234e4a5ffdb : FAILURE in 10m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/355824dcd0ce40c9874aeaf5d1160b63 : FAILURE in 11m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9265afd1f1f47e0b02160a2c221cdbb : FAILURE in 10m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b28ace90af2745e3be6d4a884738065f : FAILURE in 11m 28s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dce5e885f6bd42d7913bf5a9a91ed75b : FAILURE in 11m 26s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eb996d9d816d473c9a6abede568c3f32 : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1923ba58b1ae4c2280fdf7e6e0a7a8b9 : FAILURE in 11m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ced09539b66f419ab0fd4986fbbecc38 : FAILURE in 13m 57s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/af420ee6a8b749ef8d110cb0ce725913 : FAILURE in 11m 17s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/50de5509c73347fab8e12a558f2730bb : FAILURE in 11m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b5da41ecd79f4cab87db6183e5fe1d14 : FAILURE in 11m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/68c1c00aff4546af8b42dc9bada15d70 : FAILURE in 11m 21s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/21093791d98d4b40956cc0f55c301d3c : FAILURE in 11m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4b9709e651dd4c3daf6e2fa6209af56d : SUCCESS in 24m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2c9b4c88dbda4554943e434bacf56b33 : SUCCESS in 29m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/87d8c856f4124da9bb8282a23aa44fae : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/01a99e1987b9400fa7b5aa1339fbbc80 : SUCCESS in 1h 21m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4170dff277234a898b59010decdbfd27 : SUCCESS in 1h 31m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0466ca98acfe410ead4015aeaa61f858 : SUCCESS in 2h 18m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/84ffd2e1a3fa4d12aacf2168b3d2ad87 : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a517087113864405b5d412294fe60be6 : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bad91fe3c0444375b61de47e0bfe8caa : SUCCESS in 21m 10s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8f5b58b6b95747acad113afa6b62466b : SUCCESS in 45m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c03089696e6b4059b994965e166cc84b : SUCCESS in 58m 06s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/37736a4a8fe04639a0b39f992ead8407 : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b302c81238684b62a7f30143df50178d : SUCCESS in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57cc0a9f7270c3f27d77804c553161a0156f063d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:56:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fdc363f35e45139f58869f251d8605\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/20c31e3af6f64c428614d4e29eeaa1c1 : FAILURE in 14m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d1f0d9cefb514e2d8e5a25fc5d1d41cb : FAILURE in 1h 36m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51a11efd742b45319e6c5f3de3cf6f30 : SUCCESS in 2h 05m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d11384d43fb54149bdceecdc8f57def5 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f981be7621e4a8db01294d442f54751 : SUCCESS in 4m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fe8f69f47b4642509e366df1607acbc9 : FAILURE in 12m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8f607785208b41dd9f9eee769debab74 : TIMED_OUT in 3h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aaf972596ec07ec8399599148aacccd3e921ac9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 10:07:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"afffce23b2a5ebd400a46553bb86aeb7f39a2237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 10:35:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a149774cfb364e6cc6ad7f5f62af36675cb66668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 12:58:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccd3501495b14ea1b04c8377cf0c6110\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f467a87a5b2d4954944aabb7b47e0cf1 : SUCCESS in 23m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/af32f74aa48c4e39a1e7563ffa6789b3 : FAILURE in 6m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0d0d11e1746b4aa58f3d7e38efa7a8c6 : SUCCESS in 18m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b17ee0b3fc4c4326b01ccaedd12935af : SUCCESS in 25m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/852b3a857a8f412a971f0f6ddc4b8675 : SUCCESS in 1h 09m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/65e82c5616b74325ab1eabdc639f5f44 : FAILURE in 1h 39m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61e873a8a6e426180a5df05964bf8ed : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a5ed1a8bd0b4c889800e8a112d4ef0b : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0947e1bda27141b3bebd0333259d1bd0 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6a00c56da80409e918bbe6278139bd7 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee942e36f5184591b4d238f4c94752a0 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d603d1cfe86b485990a7114f751cd046 : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3fcae6ea27bc482aba95c273277cc7af : SUCCESS in 23m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a3bb9066b15c4a0287130fb53393b210 : SUCCESS in 33m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a7319bf9e534e7ca85f7ae5f6c6734b : SUCCESS in 1h 06m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/302d34d6fc38491a85aff70a91359d56 : SUCCESS in 1h 17m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bf51caeca6fd4089aaa47366a1e23442 : FAILURE in 1h 36m 11s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ba706460fcd5421da865e5c9b4e2ffbc : SUCCESS in 2h 20m 10s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/10f8e62514b048a484d966b5eefe4492 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/0eb149ee0df74151a447a0cec050cffe : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c1a217ada89249a3810401ec0648b430 : SUCCESS in 26m 38s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8afb9805af58447fbafa3fbfbd4f31b0 : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d8f6f7102d8c47cd8f30ed66388c4ec2 : SUCCESS in 1h 10m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2fa855dd0bc64f0dae3337b4a504f701 : SUCCESS in 1h 20m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0528c1bc1aa54348a4378227eeb49959 : SUCCESS in 35m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/89d0b5379ed44caab5ef4d866647902e : SUCCESS in 23m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dd6da55df0cb443eace6fdba7e22a989 : SUCCESS in 25m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c6b03274388a4ec9b1d95f9ee63443ce : SUCCESS in 1h 03m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0fd46773a28d422dbd721cf6e8ea7d3e : FAILURE in 1h 20m 27s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9807c8a44c9d453382f879fac037cf69 : FAILURE in 22m 06s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5426271195d84c5998d0dd035822c2b4 : SUCCESS in 2h 08m 59s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/eda336c391e8464a928913abfa2141e3 : SUCCESS in 1h 22m 12s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a113b7c84d5f4b56a9fdd6210be67fae : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/419ba1e4abce4cf090c3de91a30791a1 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/1dec0f5a589342aaad63eac4ac974817 : SUCCESS in 1h 01m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3ad83fd6b3cb4f9797a84e8b628e657a : FAILURE in 53m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/97131129884e4dbf955683e30637476c : SUCCESS in 1h 24m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/86c1d3a53d284400a30fcda7747627d2 : SUCCESS in 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9508092911d19956c71761a221aa0ccd116ec934","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 13:15:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5f56f29046aee6327c413264caf5d8221daae34d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 13:30:00.000000000","message":"Topic set to kolla-2024.2-release","accounts_in_message":[],"_revision_number":4},{"id":"953cdd488df6a57b9963bdb5fc4113eeae8dd348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 13:45:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9c40920ea249ff38c78de0873ff7c2fa6716094b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 13:45:49.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"0edb529121fd21a3eace31babcba46fca7058cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:25:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93e5969eb8484e99b7e78a75eadae342\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dab8080683aa402d8cb27febaeed24dc : SUCCESS in 37m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a78684716dbe4e3db07c006d0fb0a3bf : FAILURE in 6m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/367040e235be4bb9ad674c72f63467f3 : SUCCESS in 21m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d526f8162bdb4b35866c1187c7fe3d1a : SUCCESS in 31m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/72721ff0d6e142ff94aad307f981ccd6 : SUCCESS in 1h 01m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0a325b9c0980472abe88b699cee851b6 : SUCCESS in 1h 23m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adeb079b36546d2a157cca7188fb923 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/379f0ead8b534de8906ba1400cd6e8b4 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f71f5c0cce5d4864b0ddb54452dedece : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bc307ea8c5f408289c5919b43f0de35 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f337b8e8cdfa40b38679ae4a93e43830 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd5bd08805ca4e8db218090f5ebc381a : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/409ad82d7e094f3994df21851e222de0 : SUCCESS in 24m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0b7424abb454a8c933efa2e5691b45f : SUCCESS in 37m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/df642fcece88483ba1efde8d2dfb93c0 : SUCCESS in 1h 04m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/30e13e7a20864d3aa3415d630eb1c004 : SUCCESS in 1h 26m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/60c0a69489684589a5fb0c24c4c87a77 : FAILURE in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1249fddab7dc48aca425b9c3910cacc9 : SUCCESS in 2h 32m 55s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d3c66b756ff94ab9af173702a130c677 : FAILURE in 38m 32s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4ec1f69067ea44a09038d31d5b6522da : SUCCESS in 1h 18m 56s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/99ecbc0b2a1e4357b0936c369d36bb9d : SUCCESS in 26m 33s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/484d04a0760b439480b6ef6dcbba8360 : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7c7f97eedfac40f4bde050b0b2b3412e : SUCCESS in 1h 15m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c4dc048264bc409196756274cb0a2143 : SUCCESS in 1h 24m 11s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4ad054da57e64247bd79b3bdd5f19eea : SUCCESS in 34m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/97628bfd8caa4d9788c01082d1c0e7df : SUCCESS in 23m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d9e6da3aab144e52addabc9ac8ac32b4 : SUCCESS in 35m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7939b323771046d988099493898dcd85 : SUCCESS in 1h 06m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dbd6ff0935344ae899416c1d813cea2b : SUCCESS in 1h 23m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9a41771aea94484fbb037e08e3a32212 : FAILURE in 38m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6c61d9ac7a884832a27e9eda5425e791 : FAILURE in 2h 35m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3125fb200c294392b157cb8c1df78caa : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3d02e158c30548ed8b8fe3c9128310df : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/92b6dabb59394d369bf9f89594789d30 : FAILURE in 28m 58s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/606daa00d3de462dafbb24202f8bee6d : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ef06fce2db4e4387b894f7339c04057d : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/486225ea160147049016efdd87bdb1c3 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f91ea4350cda4a5ab3ba8a92bff9634b : SUCCESS in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"384eaff100b622a452eb57e05f441051dbee789f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:56:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a09f525a3848348d294184da60981b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c6357c4702d0415a9b266914b51e1095 : SUCCESS in 2h 44m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57a14f126a27408db133cc013b7b1e17 : TIMED_OUT in 3h 01m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36390462be7941508710467f8c9dfa07 : TIMED_OUT in 3h 02m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c391d3f70fe44037819755956bf95915 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58032712c1bf4df6b46553891275fe98 : SUCCESS in 5m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/329e2dc540a24a51bd7830b17cbbd7fe : SUCCESS in 2h 53m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1324cd75167240fe9ab3d099df754717 : TIMED_OUT in 3h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ec5863307a620393bd11bfc0e9f956368bacffe0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 17:43:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45ba3baebd84405aca48fabe2f6f1fe741d4428f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:21:34.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4659cf23ad744613b4049a40968099a7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/57222b3c9a22482891c4002688662ec1 : SUCCESS in 25m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2aed6bf9f0e3403b99d709bb69b5d294 : FAILURE in 6m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/745a00f47ac047d6b298c3c7e1d4c2f4 : SUCCESS in 22m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d700cd7e57ea467181f5fdcf540f0df9 : SUCCESS in 28m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c31dfcf010374b978d01ee090b14d1eb : SUCCESS in 55m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9c50d94495834098a5aba22d380bc6d7 : SUCCESS in 1h 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6e9b130bce4579a3e399f1acef0ad2 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8451667c7774eb6a081642cd38b80bd : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dde27a7f8cd4c5181f87f0ff383787e : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64ce99f7126d40c9958f105b0b706cfb : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6296fbb4ac3b46faaa4b0d8e7fc9a3c5 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c8fadcf51b34ca0ad5257ef360cfb93 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d9a4ef0c14254a389951d14411404c63 : SUCCESS in 27m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/13056c9ed6d94deab32e0a1862494f7e : SUCCESS in 33m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/536be91faf8f44ce9f977a0e5d651803 : SUCCESS in 1h 02m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ac8909a55ca04deea778bd7f67626699 : SUCCESS in 1h 18m 43s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f3df595754ca477499d0d5c54ec16762 : SUCCESS in 1h 43m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/17e8627c3cb44690be71f19bf73e3094 : SUCCESS in 2h 32m 32s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/22350d061537405d9dbb8531f9db3c8a : SUCCESS in 1h 13m 22s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/3c1b773bc1e64a179e4bc1f51f7b71e1 : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/feca4cc449a04300a9522e6b305e563b : SUCCESS in 26m 24s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e32c0c09cc5748d1a78742014721801e : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/aa0d3b23da7b431d8ac29a7192aa6bb2 : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a162a53a47fb4016a3e431975fffe3f3 : SUCCESS in 1h 09m 50s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d4eec3aa920143f9a83ae45eab2474fa : SUCCESS in 28m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/80dcdee3a49a47c580baa6d37037e4ae : SUCCESS in 18m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f5eb4272df254e5986fcf45034ea8562 : POST_FAILURE in 23m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ea949f82dab4eb28b62850ff1dd61ec : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/173bf12efecb49189cd5b27cec449be7 : SUCCESS in 1h 10m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5711719de3244f9aa9f02074df39be75 : FAILURE in 22m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e6e7eb60e6e24cb9a633a4d4486adfc4 : SUCCESS in 2h 16m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/35d1805d5e784cc7b69ec06577c2f2b1 : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/4ddc512fa48f4db7adb4aed21b2c9863 : SUCCESS in 1h 21m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c06d99acf3114a4191762910c6d0a290 : SUCCESS in 22m 11s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d8656d4d26b14d0cbe7d7181e238cb5e : SUCCESS in 59m 47s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dea261242d584508855e19f0d3d3a316 : FAILURE in 48m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9ce3fba4747142918fb8ca2e6c83e57c : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/de63c8469f2d47d1b419fd2e59f2cfd7 : SUCCESS in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d8aa23ce3ad035a8b88ed0925c7b0e0ad0d5daf4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-24 20:40:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"14d522fc269038cab5021100cf41583fe2785f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:53:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fc93cc96424710a747ad7c0aa936f0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dceefa30416a4400956a7832484eebd6 : SUCCESS in 2h 47m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14e37455224447899bdb2051d92ca805 : TIMED_OUT in 3h 01m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79029fb4739644c997835aea6c93e3ee : SUCCESS in 2h 31m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f9433f767f649ad811a1d38be1d8bc1 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/535be40dbb324ad2817e7fc97f1bdcd1 : SUCCESS in 5m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4c7211f8e6f344fab77da0a5b7ac4e0f : SUCCESS in 1h 53m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0acdbac30b56459b9c09683d81724d5c : TIMED_OUT in 3h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"84e694bd8c94c7707adb094e7f8dd23ae5bf7d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 23:49:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173e5d6582b74e2e89583303457e994e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8017ef0e0ac24d218ee342940e192bfa : SUCCESS in 21m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/099bf83c03bd4fa8867f50f09c7ef1e4 : SUCCESS in 1h 11m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3760dc712015459aaa0f4ce47d0048cd : SUCCESS in 21m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c17060f82f0f4b1baf800fec7c11ba62 : SUCCESS in 27m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47bcc9e2f72346eaaf221444a5d4962d : SUCCESS in 1h 07m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d232b5dafeea4dbe9b94cbfcaaf8c32e : SUCCESS in 1h 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7627f8d0fd4d9aa0774134386433d7 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce5128f6c194d1a8051d20203213afe : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07797f8ed133482789f19ede95945057 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a98c75b00cc49d78a880b90a8fac50f : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e4be9842f84fff86170ef8157d5359 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b6f9ed8fd8a47f5afb6442a0341327f : SUCCESS in 3m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f0cd0d2cde7f4f35b9cf20b363c6c8db : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9d226538b594400c9a398378b9cfe70d : SUCCESS in 35m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e913b9051b0e4a3cb7c569d93fe4a881 : SUCCESS in 58m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a9222d430fac442991fbc3ce9f637205 : SUCCESS in 1h 18m 17s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a16a542224644d699ddc5a0c5ed34dca : TIMED_OUT in 3h 03m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cb2f2bd18e48469dab5adf73ad5cad90 : SUCCESS in 2h 10m 48s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/da9f94eec94b412f8371fb12f6cc0fdc : SUCCESS in 57m 21s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/cfdc62d55eb146ddb51f188ce973aee2 : SUCCESS in 1h 15m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4243f0ff0fb3430e9b8aa4b231c6ef58 : SUCCESS in 25m 21s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9d58e90be96c468d86169791d4cbe650 : SUCCESS in 48m 53s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7606c2991fb44e8b88f243aa75d603c9 : SUCCESS in 1h 07m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4b27367933574d929b3b38c69dd3fe18 : FAILURE in 59m 34s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2420d930ac2948f7ab6f57dffdc58243 : SUCCESS in 33m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ef7027ab8c3e4f2aa85120112fc19a0e : SUCCESS in 21m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/60416c001ad141ab9c1bb217a69f4232 : SUCCESS in 26m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ea551fb95a04683bb765d84a35c9800 : SUCCESS in 1h 00m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cc699280f2574e158b0d0b14e33f9fc7 : SUCCESS in 1h 16m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/aa4f002c94994428ba939af133f6d071 : FAILURE in 22m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7fe2d69771654ba48000563471922581 : SUCCESS in 2h 23m 20s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e03c0e405dd749388a51bfb9970cb2f5 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/4c966dd7ed5a427d8a16427e939bd680 : SUCCESS in 1h 16m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a02c3f5264304226b87402788edb5a5d : SUCCESS in 21m 14s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/dd921c0623db42859d418ffe99534f27 : SUCCESS in 46m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/282635be0ebb4dda9e13ae273184e277 : SUCCESS in 59m 12s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4a66b306da94474ba8e2423ad52d6fb1 : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ed835b7df772490983032529a8aa8783 : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31489dbd08fc56e2723117e10c8783b575068837","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-25 00:05:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b4f3852e97446aaf7a58fad0f804888cfba3b91c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-25 06:06:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"0d82f687c6b63eec2b0bcc9b88d89a8058ebb0a3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-25 06:20:48.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2eafb5abf51df5f0e00e910126d2869ede28666d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-25 06:32:56.000000000","message":"Patch Set 6: -Workflow Backport-Candidate-2","accounts_in_message":[],"_revision_number":6},{"id":"1e2b43e9166fde1b3c42cd3e8ed9a9c5b802975c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-25 06:33:00.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"886c0e2182a58fedd5d87421ae32b52a25e09ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 06:33:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8db25c11141bdbd8a972dd5a6e6da27914efa4db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:12:05.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0042cc263a87406b96aeb5f50bca2e71\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dd95c3527b9d487ca163c0f79d6bd03e : SUCCESS in 17m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fab98e13921246b5ae074bf451c79abe : SUCCESS in 23m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1ff43ebbf31485fa8926fb0edeb383c : SUCCESS in 1h 02m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/34369b9033db406bb322b9510ab9a3e0 : SUCCESS in 1h 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e952072f694048d5a922ad4b43ee524b : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbdb0176d8744d1a3f25eccf690d543 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e11236f531e14375834a085164d320d5 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046d2cacfe2c46169e75469c52a0fd71 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd67b31cab814b73b811da41bd5cf964 : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36b770a1a69d4ac1be6bf0f000f66e9e : SUCCESS in 28m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ddf10cae8264e84b1d6afeb2cb7cf2a : SUCCESS in 32m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1c21d1cb8c76479b88aae0acc5072e96 : SUCCESS in 1h 02m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/df64048fb744430cb9416aed7a0c9f62 : SUCCESS in 1h 26m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c9d0f112993d4db091f3c16f222a9aa4 : SUCCESS in 22m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/db41888e1e9940bdbf28fd0e7fc52517 : SUCCESS in 28m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f8726bb536744cd9b680f7ccad19c78 : SUCCESS in 1h 00m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4cb13e919994b86b30c1cd9f857bbad : SUCCESS in 1h 27m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ea6c8d5a792f7ca9e23c52e5ebc40af38bdd3bfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:12:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3f099642ae85e5ea80c43d99a05f82380a60b7f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:13:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee69ce4002be4b518ba90a25ae1bd42b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42df77302dd455dbdb3683126502de1 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106f7797971442c0a8c86f471abff62f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6},{"id":"46678b7e0407238fc8e46e221b1ab51f04add3a1","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 11:03:57.000000000","message":"Created a revert of this change as Ie2a747223b33039684650bede8863af0c8906d2f","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"db0f87d3c8e1e7eccdeb1ea46216d530711680d3","revisions":{"71783aff3b179fa224e70e157f906673f8eaae83":{"kind":"REWORK","_number":1,"created":"2024-10-07 09:37:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/1"}}},"commit":{"parents":[{"commit":"5f07585e0bf5959d4acd18929613d2d939318581","subject":"Merge \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Dalmatian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f07585e0bf5959d4acd18929613d2d939318581"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:35.000000000","tz":120},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71783aff3b179fa224e70e157f906673f8eaae83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71783aff3b179fa224e70e157f906673f8eaae83"}]},"branch":"refs/heads/master"},"12eadbac676d7a14700e8668c9e134e030bb6c8a":{"kind":"REWORK","_number":2,"created":"2024-10-07 09:50:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/2"}}},"commit":{"parents":[{"commit":"5f07585e0bf5959d4acd18929613d2d939318581","subject":"Merge \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Dalmatian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f07585e0bf5959d4acd18929613d2d939318581"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:50:55.000000000","tz":0},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12eadbac676d7a14700e8668c9e134e030bb6c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12eadbac676d7a14700e8668c9e134e030bb6c8a"}]},"branch":"refs/heads/master"},"28f7e887a8fd92e6d1d30781540afb86618e9fa6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-24 10:07:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/3"}}},"commit":{"parents":[{"commit":"71d7edd61512cdd56f1cd2c04beec83c5ab72e23","subject":"Merge \"Fix missing tss user in Ubuntu/Debian nova containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71d7edd61512cdd56f1cd2c04beec83c5ab72e23"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-24 10:07:16.000000000","tz":0},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28f7e887a8fd92e6d1d30781540afb86618e9fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28f7e887a8fd92e6d1d30781540afb86618e9fa6"}]},"branch":"refs/heads/master","description":"Rebase"},"367e00465cae79c63e16f37434194036625623af":{"kind":"REWORK","_number":4,"created":"2024-10-24 10:35:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/4"}}},"commit":{"parents":[{"commit":"71d7edd61512cdd56f1cd2c04beec83c5ab72e23","subject":"Merge \"Fix missing tss user in Ubuntu/Debian nova containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71d7edd61512cdd56f1cd2c04beec83c5ab72e23"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-24 10:35:24.000000000","tz":0},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/367e00465cae79c63e16f37434194036625623af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/367e00465cae79c63e16f37434194036625623af"}]},"branch":"refs/heads/master"},"d599e0954ef0ef6135ffaa0e0a441ac786150739":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-24 13:45:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/5"}}},"commit":{"parents":[{"commit":"830500bb55f9d666bd11790c42eb47a2afd914c9","subject":"Merge \"etcd: Bump version to v3.5.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830500bb55f9d666bd11790c42eb47a2afd914c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-24 13:45:17.000000000","tz":0},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d599e0954ef0ef6135ffaa0e0a441ac786150739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d599e0954ef0ef6135ffaa0e0a441ac786150739"}]},"branch":"refs/heads/master","description":"Rebase"},"db0f87d3c8e1e7eccdeb1ea46216d530711680d3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-10-24 13:45:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/931628/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/931628/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/931628/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/931628/6"}}},"commit":{"parents":[{"commit":"830500bb55f9d666bd11790c42eb47a2afd914c9","subject":"Merge \"etcd: Bump version to v3.5.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830500bb55f9d666bd11790c42eb47a2afd914c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-07 09:37:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-24 13:45:49.000000000","tz":0},"subject":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)","message":"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Dalmatian)\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/933312\n\nChange-Id: I0ce9e907594fd160a08ad7929d5bc4bca9abf536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db0f87d3c8e1e7eccdeb1ea46216d530711680d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db0f87d3c8e1e7eccdeb1ea46216d530711680d3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
