)]}'
{"id":"openstack%2Fkolla~932174","triplet_id":"openstack%2Fkolla~master~I02f3adf9825636e2b81c50543d6f95ea986019b6","project":"openstack/kolla","branch":"master","topic":"add_valkey","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-10-29 15:28:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-30 06:05:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-29 16:09:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-10-30 09:05:33.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-30 09:05:33.000000000","reason":"Change was submitted"},"36702":{"account":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"last_update":"2025-10-30 09:05:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02f3adf9825636e2b81c50543d6f95ea986019b6","subject":"Add Valkey","status":"MERGED","created":"2024-10-11 12:14:58.000000000","updated":"2025-10-30 09:08:01.000000000","submitted":"2025-10-30 09:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932174-add_valkey","meta_rev_id":"862c82e061cdd9e86089d573582e78829c717b02","_number":932174,"virtual_id_number":932174,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 09:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-29 17:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-29 16:09:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-10-29 15:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-30 08:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-29 17:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-30 06:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"reviewers":{"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-11 12:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-11 12:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 11:09:07.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-11-28 08:27:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-01-10 14:12:14.000000000","updated_by":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2025-01-10 14:26:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-24 14:47:05.000000000","updated_by":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"reviewer":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"state":"REVIEWER"},{"updated":"2025-10-22 13:22:56.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"CC"},{"updated":"2025-10-29 15:24:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"d64dccf8d8e6114a38f0ff581b948f7a56e70057","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-11 12:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b3f11ca76279856e2ac25123302b24f75dc6b85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 12:27:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/391d9d10624c485abcc282158e488898\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d8386c9c808c428cabddfba961370da6 : FAILURE in 4m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d597a4386ab46898fbfc9681acea8ed : POST_FAILURE in 2m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8693d99ea1d1468c8fa2e07a2216e15f : FAILURE in 10m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5228222265d44529a0146252f7f795b6 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/327b87f9efd44465810a2db3e1bb6370 : SUCCESS in 4m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f77f4afa39f845bfb17e282fee0c2a0b : FAILURE in 7m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/80a33563c533473dbf9229797355bb60 : POST_FAILURE in 2m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dddcecb52d9bbfa08a99f2d4ffaa5c71377fd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 12:29:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0714938e6bcd4c7faea1e8246e3abff3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/56f2a4b5966f4ddcaf9f93cf679550ba : FAILURE in 4m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f1bbf8435cd944c5bb847ea251329cc9 : FAILURE in 5m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8775491735224e9f9d41c545b832c8c7 : FAILURE in 2m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e79b52eb930f49d29f0c786de8659bcc : POST_FAILURE in 2m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/49f408ffde1444af8f7c0ae506570609 : FAILURE in 7m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b4014e149aa54170b798c54a1a757f25 : POST_FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833d3ad2d22e4f7aa4644b910a494920 : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c2258ec0fc4379b71514a0bd85734c : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d221e2bfdea441c98335af227a8f995 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1911946e608c42b3a1a7f1878546c5b7 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18da96866b304e2db08de6bb067b891f : POST_FAILURE in 5m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/16f7540af5264991a8605ad3a44f54be : FAILURE in 8m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ef02aa9127124b57ae23fad3c03efe52 : FAILURE in 6m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/34e7aa5eb2c44b4d993c2c50fd15dc7e : FAILURE in 9m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c62db2023cc3426592c9bcddfc0aef59 : POST_FAILURE in 9m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb51721941f04f1f8c60af7f314638dc : POST_FAILURE in 2m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bc207b1fe2fd44838ac7446c050487e1 : FAILURE in 2m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eba794bc614e42788ecc73646cade85c : FAILURE in 6m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/032af7be255249d2ae5071cf17b517f7 : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2db871e4cc795faa725616369f8dcf27b4f2e3b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-11 13:03:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"87c6743e800c10757879046f4d9cf50c24ec4e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 13:21:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e221f7b691cc4902ac9a149b280ea687\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a632044c83ff4e0fb30bbc4a67354e53 : FAILURE in 4m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba2aedf2183d49ae8ee1a7702e2b3f7d : FAILURE in 2m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/673b4dc2f45042e8a1fc6d740113dc43 : FAILURE in 9m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dbe1b700f3f4117a98332f1216a14d2 : FAILURE in 3m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3532048f21784be5b94e039c213ad937 : FAILURE in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5fcc92f9f8664388b1ae08bcd6a784fa : FAILURE in 8m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/24d3a53c214348e0972521b4fc34ba0f : FAILURE in 2m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49b69b934b26208dcbadf9e5fda00ba6afc56a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 13:41:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf6c672b1f3b4b9cb16a087cec1354a3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e9c5db8eafb0410db6b29fd291e1e553 : FAILURE in 3m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c75517655aec4af78a7834829d81e853 : FAILURE in 5m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dd3ae68b710740d481025e0700120227 : FAILURE in 2m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/afcc32cadd5f4ce4bbc63e8ea5c06d66 : FAILURE in 2m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c8459861bded487997f9a94fa171fc6e : FAILURE in 6m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d47f4eecd9f74f799bc20ad05272a2f5 : FAILURE in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2c91746d5a4d5eb8ed64daef88cf2a : FAILURE in 2m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fa6783ec57147efb9af8dd8e147df3c : FAILURE in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe008f7cd9764a1ba07840096276155c : FAILURE in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d2f5aa4d89e40ec8075bc713ab8c97d : FAILURE in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946f948ac973452e8480b0464bb2eb93 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7d881c410dbc429bbeeadcb0b47bf8a7 : FAILURE in 7m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2105d2a7d1a1472e98a78cfc48b9a97f : FAILURE in 7m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed657059e5ba4ab6bedf21542d1735a5 : FAILURE in 10m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d02ce7caa9534850b488d2d532165b2f : FAILURE in 8m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/87b1345462804c539dc7f5f3b334fb37 : FAILURE in 2m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1339496964e8483d9e9b7c6a8c911cb2 : FAILURE in 2m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1edab2e3fe644cbead7b6e3cb0dcb53c : FAILURE in 6m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/55ea69d8d4d14c65b76b5eb8812e0643 : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c096acc4c99775e0ef94333863acbb82d1728d68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-11 13:43:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f114f7fb540f0bde5234ccdf33004ea65bfec6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 13:57:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628d1650f61e4f7a93c98db0e51c5a70\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e4f5278bc84c43f5aae6f88316f75771 : FAILURE in 4m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34e89650fe154c54bb9c8b1f2358ff25 : FAILURE in 3m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/35471686a8ce490b9c55d39a25b9f3a0 : FAILURE in 9m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0632a482c1c64c39a068dee73b4c6ddd : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19a65db5de844a62b9d9e3738b8086ba : SUCCESS in 4m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/12450b3e1a6140cf9f5a94c7bf0a16b6 : FAILURE in 7m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ce065991b95a40ae929c61553133bcb6 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72bcacb7834d7e6c0b70ceaa2e137e0b5da73dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 14:00:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc1dd23a55ff47e48eee00b95c8ec8cb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4490c5138e4445019c497c0ba41cb7ad : FAILURE in 3m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4e27b15a51144b54a3b14f5188f46c6c : FAILURE in 6m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7c0c256dc564b5fbcc7313b5935904d : FAILURE in 3m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0f44f9bf8bde4760906a1a2e17ba2149 : FAILURE in 3m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/814fc5f0477d4a7caa65535fea3f8697 : FAILURE in 6m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0f7675b87ec74377938dbff1e87c057e : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372256143f284d4c8ccec3dafeee9026 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4cbdebaf03f4dc68515fcd4a6d32d57 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7ee861707b543198b7d56ba48b98dee : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecc97a7dd9584cce8885f14737c5ecb6 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f5cb469307468fa7a28e529f9b39d4 : SUCCESS in 4m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/89ce40885710443ab1fe4f2ea22029fe : FAILURE in 7m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/598f789876014d60a4cc47bb35e666a9 : FAILURE in 7m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cd45f02afd2b43fc8464c1396e28e4ff : FAILURE in 8m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/14a283e6e2ad474e9d35df266a8a50af : FAILURE in 9m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/24db83e0b7234773b6305c99c1eaae90 : FAILURE in 3m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1e97666965ca4679a00e0d921ac39fa8 : FAILURE in 3m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/03f374dbdc0a4cdbba156135526178cc : FAILURE in 7m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a5d6434e0d9043f2a0cece046e98e36d : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":3},{"id":"151b399aca91615231ffa63c1b5b556f01a0bcff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-11 15:20:57.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"834a6f5e80850cc966416e124797607a40b12b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:30:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/561bd13db8454da982076d4db6f10931\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6495b9e3499241129bb28d37cbb8da30 : FAILURE in 25m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/345d37657a4740a4b93c68c357929f22 : FAILURE in 5m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b032fbb9b6b2476c98115b5a5ebd1bfd : FAILURE in 19m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3a7b79b374a24eaf820da29b963493e1 : FAILURE in 29m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/78a0837c5a5547989dfbd52d85b7be9c : SUCCESS in 1h 05m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ce4b7f2b7fe4700b53fb41f65b58f7c : SUCCESS in 1h 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd9fcdca59be4e328ff88f088a834fb3 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dede3182f98c431eb857a6328670b0b2 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1ed58fe2b4a47e5a34f31d47f9e8edb : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0edd46455a5545e8b47ca31a5170b272 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbe020b0431439584268d55d11826aa : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/858d44aee0b0486a9888ecd238fab1ec : FAILURE in 28m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed5e2f061f8543a2a5f94f84f14b48ee : FAILURE in 34m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/956a68de23d04a6d99aa89b4c11ecdf7 : SUCCESS in 56m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7e1a1012f76b48ff80f180fa0ef1d9d3 : SUCCESS in 1h 33m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ac21384e546d4d9aa9e515a185a21c04 : SUCCESS in 22m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b7cd5ef59fd24a56a60dc995b3f15a7e : SUCCESS in 26m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bf64f4d319da4dd1abbd31b06ebbabff : SUCCESS in 54m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ea05c1267bad4d269e7fad0556932cc3 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":4},{"id":"12f237c8c0b2af98ee7ac7af2b6fdfdcb8325fda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 18:27:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc29265ef49447938c0254059853f1b4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b72f883a5a444d63bbab9b29d0cbe48b : TIMED_OUT in 3h 01m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a9839447e6ab441aa79cbd16f86d6585 : FAILURE in 1h 36m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f6bc14ed76774999b5d6ca41142ae157 : SUCCESS in 1h 49m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4126742e088f4c75922ef284b1202e6a : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/307cbd0d110d45c583ae931b3a0f0cb8 : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/381e3b7363e345bfb18a7d766a13a35a : TIMED_OUT in 3h 02m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f822052249ac46048cf0daf168335930 : TIMED_OUT in 3h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ae4cda21d34b300603b78ff2a6214edf9e8c6eda","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-14 20:55:04.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"67705acc6b7eab73a336a62eab59d48368169694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 22:07:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/573109f0214549c8ac9a7b18e54b3ffb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/219da1b0132145ccb5599ee63b2ccb43 : SUCCESS in 26m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2e18976923734474a5054ba394b2f510 : FAILURE in 10m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d08eeec16e9400685a09ae11cb8d182 : FAILURE in 20m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/43ded657753c47868f92e883f139d3a2 : FAILURE in 29m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dfa651fef7d144e5993b455bb7449967 : SUCCESS in 57m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/08d45d110c384d74983efc314a069708 : SUCCESS in 1h 07m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb61cb078b441a6a6ba287711c583b1 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e61f9dd2234e7698dfd870e90b28b6 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ec18aa019bf4bb2aa31a35b9a886449 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90c3b784373644fda4c90b020f5bf1b9 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296a5499f22a430ba60154a66b795763 : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e970e9072392472ab48718dc2e3eba6f : SUCCESS in 25m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/50d1fd0fa91b4e2a9785bfc38fdc81ec : SUCCESS in 31m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c101c2fc65de46c3b3d4bb2f6db413dc : SUCCESS in 1h 01m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/97ef18fda5444209b63b6d723cd4cbd4 : SUCCESS in 1h 10m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a50ed84fbfe6480e8ed4e7f9cfc75231 : SUCCESS in 20m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bdc8c7417b7f446690987890456d835e : SUCCESS in 24m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a932c24cb98945e8a0497da063aa3c36 : SUCCESS in 48m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/69a274d72d984a87a8a0bd262315f8e1 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":5},{"id":"644265d3029acbffa1e51dd00ddec07183a14c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 23:57:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cef7b15b3c2044e7bdbf45dfd1af8229\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9b4f165e2256433c8e1a0c3f30531396 : SUCCESS in 1h 42m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e993f0100d4842d9b3702a2315132f10 : FAILURE in 1h 54m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c10c9bd93fe24129b0b37f9880f7c42b : SUCCESS in 2h 00m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16efbdf3611c437cab17f4bb5fc3eac1 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f4908542b5944bb9299ffe88691681e : SUCCESS in 9m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ef11b96457f8484ea6afa323a7cf2454 : SUCCESS in 2h 56m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/688993aa0fe94617ae57afe81c059363 : FAILURE in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"99a18f407107127b69a875ab0a1fcd6851c3afd2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-25 07:54:41.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74ced27c96172812b48671ee766f743cb6fdc09e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-04 11:09:07.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"57305507d9bcc5f7d868705755e356a7cf7c645a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-04 11:09:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"80edb6afea5856b3fdd4190fa5cfa6a67aac4b32","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-01-10 14:12:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"061ae5815059d567d5f1088c45b318e7d0c2b551","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-23 17:06:19.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1e6756dc7639bc512e88da9f58166ce813920ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 20:07:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/552e88090f75471fa7dba6feb38f91aa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/59976ac2fa484b9ba773e7fe849ca57f : SUCCESS in 30m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/97eedc5b0e37481f85fff57a524c9195 : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1463a0c57e54365b013dd0aedf25127 : FAILURE in 21m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/baf53158b9d7486280ab81d81ced5aa6 : FAILURE in 27m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aae2672c484546a4a12df6fd3f3ae8a2 : SUCCESS in 1h 04m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4e2ad5ac64dc468aad291457ad550424 : SUCCESS in 1h 42m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ab4b89063d488586baa36e8f0f4b82 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4faa590cc914dce8d7083e5e14daac1 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e767bbbeb6e41adb2d9bec01c3fbd6c : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb36637227ff4e808d7c19252632b957 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526cd2dcb06f4602bd264f1d0cdc09ab : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bc7fd33b565a4b8c902ed6ec61f503d1 : SUCCESS in 24m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/24a723b8a866450d93c44355771ec0aa : SUCCESS in 38m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b0f516bd96a148e3a961d40aba8c9784 : SUCCESS in 1h 13m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7a77a3e1e4a74af18cdbb0d8b74f3ddc : SUCCESS in 1h 37m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9e9a77d457b44755832a41b1561bde09 : SUCCESS in 24m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab999dc8fb2d447cbd2be0f244278ac1 : SUCCESS in 25m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81ec86b558c94f20a78a7bb3f5b43bc1 : SUCCESS in 1h 08m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/443f1cc3dd7e4267b1b8d9b763ac1ccc : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":6},{"id":"4431612a9774256fb8f4029c8bc9c3097e47f26c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 20:48:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3357d651dc4a5fa0485f8247c6febb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cfb1875981c5417bbad399e607746648 : FAILURE in 2h 14m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f2cbc859bf34f6b9982a3cd7263379b : FAILURE in 1h 42m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c63d9f95dde04acdb44224988421ba3a : SUCCESS in 2h 29m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d6b2fb8c1214dcc8f0f2b36b92935a2 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13362f29c94e49fe9eadc0fc1f5f5b68 : SUCCESS in 5m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/df837996c61e4070a412769254a9cc0b : TIMED_OUT in 3h 02m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ba621c069364a7f839e18e022c27eeb : TIMED_OUT in 3h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3019e01edf6d0cedf40ff445dddd26547a5e1dc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-27 13:03:34.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_22629\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":7},{"id":"4d45589a53f63c6073b15f30b8bb345f0dddad4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:00:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50041afb13ba405997a00f90cafd29fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/930399f5825f4684b36bd6dbcee842ff : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0bf4bb5e670b4ffb93d1399bf3af42fd : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b4c4d6edcba487e8bac97e107ff11e1 : SUCCESS in 20m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/094629fa48b947ef86d11dd04c32ae42 : SUCCESS in 26m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c0bc8612da6f4caaa7f543149e6e3a74 : SUCCESS in 1h 05m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5d6c2bd294d6424f81005c18973ca7a0 : SUCCESS in 47m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84486b3ec8dc49d1a5f3bac59c664994 : SUCCESS in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f42853177dc3468d8647743f71487fc6 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d84165bb2d3e469d94f6b0b462f9b2c3 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/078a4dab5a2b48a1a407c6c5b6f02cc8 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd4ccab6d2c42c1ae99463ebf6a5730 : SUCCESS in 6m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af8fb6a698c540128c5b5b63bd3666b8 : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/877cb132ee17446c9c063168897fa86a : SUCCESS in 23m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/305bdcf3de2b44cdb638850a29f16de7 : SUCCESS in 1h 19m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/30ea1ada96334cca8179c8877343dd75 : SUCCESS in 42m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/85f09a5f34aa4134996a774f80997a56 : SUCCESS in 1h 32m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/61fc09679dcb4957ae22b33cf9da5a52 : SUCCESS in 21m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b97936e0e83e4837ad5f882995774f45 : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d9d34b412b704f86b61a336591d30ff5 : SUCCESS in 31m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ffa28cb32b094dff89afbf288bcf1bb0 : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":7},{"id":"df76c8d904bc55d95134a2c11a9a02e7eabe0e96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:30:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd874960ba1f47efb51919cd4bd0a2ae\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ddb0fc81dfcd4a1b93423ea38290090d : SUCCESS in 39m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c553e0c7198c4a6eaef0ff60e2c85536 : FAILURE in 31m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d22c6946ba1b4b3fb63bd917ee75010d : SUCCESS in 1h 00m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a46278180fb64a699b59feefa9a2678c : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18232f6a9cf44731a05979f3052f7d54 : SUCCESS in 4m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1652bfd771a843a6b5ce818b8502e9b1 : SUCCESS in 30m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d2d16807ba064e34b96cc777413525ff : SUCCESS in 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1f9b5a52e8aa233fdd016faec6c1fbafc2a347f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 09:59:25.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d72b280a3cd901db971f913e053d2ce17c49e7d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 09:59:29.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/users.py\n","accounts_in_message":[],"_revision_number":9},{"id":"db4809037f0cb7b2f762fd2692f1e5220216c704","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 10:00:37.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"fba5c192fd60c16a143c760be50589e7d99a9d45","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 10:02:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7d0bf8a7bde994726e051d93f4b01392ea24add6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 10:02:55.000000000","message":"Topic set to kolla_valkey","accounts_in_message":[],"_revision_number":10},{"id":"a33eaf2dcae76ab6d0e78a34a1a861156635587e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 10:21:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd0c40365bab41c4bfe2ba3d6d5608ae\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/df792620d49a4651aa32e61175c001b0 : FAILURE in 5m 29s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/fd7d625f5c774d2ab055b344cc264eb8 : FAILURE in 11m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/30385d9ff44f4c50a4141357a4bf5a25 : FAILURE in 5m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b58a1a1513094c20ad9784ead3bf968a : FAILURE in 11m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6dac61ffd8c8427cb85491ad010cd502 : FAILURE in 7m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4e49d7aae3404442a2066665db6383b6 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f312e6bdf446aebd0b2d1c1035cd2c : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ff28346e8af42228b96137a20b21f7a : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d422797ac9b41ee86db529594ceb8f4 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2607a49831e74af689e7cd5e3f79efc5 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1f17fbdd31a42fdb7ae3914c063008e : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72165ad6491e434194168f083af072d9 : SUCCESS in 6m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfe8788aec98475c8a0dc5534be80e49 : SUCCESS in 2m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a0de4bd5b2e497dbe00ced1366f22d7 : FAILURE in 7m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dd41bf01c8e4460b8d67168a7c969335 : FAILURE in 8m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0eba39cdaf1745dfa4f01a44a2e0277a : FAILURE in 11m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ad79567b0dd845ce93b81aafbb0eb5ac : FAILURE in 9m 03s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/59823db93c0845f2b90400ea30bc1143 : FAILURE in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"db34978f19f60d85c63dff575d9eb7ed71106651","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 10:42:37.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"cd1cc0711d3d3b0ef1be49f5f2d33a61d24f8c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:56:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0caadc74b9a84da59125c2b4c2092d1d\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/b011b6b7be6d473cb48385b4ea8abff1 : SUCCESS in 14m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e8c6d656c7f2439abcd330f2de1e189c : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9f6cc917059d4fd883097a063e9417ec : SUCCESS in 9m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b745d3ee31db4ae8b15bef7511fe9d72 : SUCCESS in 29m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6724af5d453546afa6883e58ee12a71c : SUCCESS in 1h 12m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cb33b198976f4320b70fac5e81445156 : SUCCESS in 48m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6dd6d8bf624df5b274698ace5f663c : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39395f6a330a4e3890a63e672cde07eb : SUCCESS in 1m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfcb47ab4e1142deb867a67e07452d04 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3512d194843f4fc5994e01bee2346633 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6664cb57e54288824d69d107847feb : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d50b46ae3b41f9a885c4c1da695277 : SUCCESS in 3m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/99ab57c47a3b4bbc8d5724e0868b81f4 : SUCCESS in 1m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d979c1895c244342b63065ea5416af57 : SUCCESS in 24m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f5364662f562406c8c36d0f57deec7aa : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/65e3ac17be9a46b7aba61e23b182c39d : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13767c36e72a486ab3f9e168af1b0c41 : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2c0b7d4066eb477b94f77e1067455ba0 : SUCCESS in 1h 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f31cca00ab3d181001adb29c2f877e481fb297e4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 12:15:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"9f353047a2444b55ad424b230bf78693ce8540b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 12:16:02.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e0ff3b1e9bf8287f543575a7e805062f09c1b790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 14:28:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed25e6a7ea5e4d17b54b9c81921baa32\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/27e3ff30dc8e4443bacf0d7a65a2f6c4 : SUCCESS in 15m 43s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/cdaf0c2eb77242049b195dec95053cc4 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2a9175594af48918d5fc65d1906f4ac : SUCCESS in 18m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/12937802191549d3897a1c77c53554e7 : SUCCESS in 28m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/20fa2a70baea4a73aaead13acadbd487 : SUCCESS in 1h 06m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4016aee01c5749218b9d770d8f92e83f : FAILURE in 1h 48m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f1ee30a94e4246af58beb54cc0a5b0 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0079c549ca3a47e98b15cb6bfd6526c8 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48c97551409c4895a867a56e2a0703ff : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd7d6bd02d4c49f79c8cf07e3fbb7c2f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f10cab280ea43ecbd47805c89b7426e : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d93b25a41b4c76a8474a9baf42c7ff : SUCCESS in 7m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d735f18f681442d0a57622edce0ece1a : SUCCESS in 4m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5a6fd845ac9f468a8ac10b8b5b2fd6ae : SUCCESS in 21m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0f6b735ff38e4193945591f2865bf697 : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c76346fdf0b419f9f8dc6e153eaa250 : SUCCESS in 1h 26m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ff9e20bcf6a2423b8b9a5e274235a841 : SUCCESS in 1h 44m 48s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/15e115bd6b9948c390fc4447e68b62b6 : SUCCESS in 2h 07m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fc312b153139522e2c817e687bb29b23d3677ecc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 22:07:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e708614f779488fbfd80dfffe9295b4\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6b2b66d9c69b41b48577a45a2e43c981 : SUCCESS in 51m 18s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/843ec4b8b8894d9fb239775ec57be2b1 : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0668f5ec62b24e90bf791ea359cf03b0 : FAILURE in 1h 24m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8428ccd0f0684e5d86fead3f483f61da : SUCCESS in 1h 43m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c791c2332884d9aba91f391d748edfd : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/896f0bff680f4cf4bfee27f0e60e5e76 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31fa228ef1c2438a83c7bd040e409a74 : SUCCESS in 9m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a4db266b7ba04d4aad4b0180b9023578 : SUCCESS in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ffd30f661e3bacb7818fad6fcb2ae1234622fbee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 09:51:11.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"28c72c77552e389f635280644a94766cf19d355e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 09:53:00.000000000","message":"Topic kolla_valkey removed","accounts_in_message":[],"_revision_number":13},{"id":"84964c00ef85a62d9ec6d6aaff9a496dc7cf4637","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 09:53:45.000000000","message":"Topic set to add_valkey","accounts_in_message":[],"_revision_number":13},{"id":"7d524298380100b95b4f26ad14dd05341bd134e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 11:04:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ae02201ea84fea9c7fdde0b0dce8eb\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7690a2962ef047f5b5e52d35c581ada6 : SUCCESS in 20m 30s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e2bcccbb8626429b8b6b128e72f8a40b : SUCCESS in 54m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5684733af5dc4e878df0deccc1a8cb8b : SUCCESS in 19m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e5732a691c4d451da53af69529fb9c7b : SUCCESS in 31m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b1742e7430584dc8b45f0b62b8c0aeed : SUCCESS in 49m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/890f00ed49c942b2ae6eab7a28766506 : SUCCESS in 1h 05m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b46c76534e4672b612bedbe0a24f3b : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/413a2158963e40a397c9b125fb6dd54d : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e12c1e3ef124ab798a55ee8f160abce : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab43409941fb4936840cfef8d37463b3 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8552424a51d24a0fb5f8b2dd9d15e04c : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23441ebaeb544e009fd38198313153f8 : SUCCESS in 3m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eafc9522e3564cde91cb9b402bf70ac9 : SUCCESS in 2m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3e489f4ef24e4ca09d0c499f738ed06a : SUCCESS in 20m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0f13242319df48e88959d17b25e745c4 : SUCCESS in 17m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c8a763f0d25a44b98512dcaddacf416e : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ae203d248a1c4d81938951c0c098079b : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":13},{"id":"470951d49a125080cebd1bb0a484144b568af591","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 11:48:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecf13effdb54e2ba6b0080ec636b368\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/03576eeaa80842ac85ca9e55d9cd5f78 : SUCCESS in 29m 43s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/162aaa31968f4358ace95d286997eed4 : SUCCESS in 1h 55m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0d1bdb50d7264348acdda4249113f0ec : FAILURE in 31m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a0f879451da64b69a3fbfbc6004fe4d3 : SUCCESS in 58m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/430e0d1622a14a9aa48cc0c961f973d7 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4930b95e0f2e47f29ca66ce9848454cd : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/609be9e0f1a7401db64112799db9478c : SUCCESS in 5m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/be97c5a2ea60466a868bfe88ef469152 : SUCCESS in 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fca7cb63be14b8c0c5c99ec4c58c7c2700e07c9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 14:25:02.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7f398eaaa19a0eec12c677814e9f56ff75556e53","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 14:28:52.000000000","message":"Topic kolla_valkey removed","accounts_in_message":[],"_revision_number":14},{"id":"f68e4674fa20acc4f9414cac22348d1f7ee2ada9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-09-08 14:29:06.000000000","message":"Topic set to add_valkey","accounts_in_message":[],"_revision_number":14},{"id":"fef75e2e64c08817cb000fa1a1110a4136b6ecfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:39:04.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36045639651a4bcba621607c40e7b4c7\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d869df32b6a44121994b4289c9713147 : SUCCESS in 18m 45s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/baaa30b9fd6d4862a2943bb87d970d5e : SUCCESS in 37m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/03b5bae46019478f99cdfa1551e05ca4 : SUCCESS in 20m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9530020c58f747dcb9891c05664302f0 : SUCCESS in 23m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fc3f666e215b4845bcdb252a2e41568b : SUCCESS in 44m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f260f84f3d964eaab642c68018039f5e : SUCCESS in 1h 06m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478d5be3cd324923b849d4417ccd9cb1 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf26a395854e4a3ba732fbd72a089af2 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d4992384fde4d7c80ca008ba243ea3c : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66daf2fe59374b0ba9db5d167be7ad43 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/735fbe8590d842a686693bfe4812de6e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d970576b37264c70b707f0087f75d1a8 : SUCCESS in 7m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1953f4263a7e47fdb691eecb0caf6e63 : SUCCESS in 2m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a87dd624c04a40c49b2c7b64cd33c4df : SUCCESS in 19m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/05c42da760a24a8aa1adc340d4c8c75f : SUCCESS in 24m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e6c90b3715b43e2a044d46d2ddba9c2 : SUCCESS in 49m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d2ac94f567b84bb48fc2fa5efa8685b4 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":14},{"id":"7c77f517f3140bd572f0e56d95a6dfc0770f084e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 17:23:58.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84b3ee737b54600969a63a0223a7ac5\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/29afcd3b515945168a7d59ddfa99de97 : SUCCESS in 25m 00s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/be5eeef4a87748a0ad95f44f3a6d9ba0 : SUCCESS in 1h 53m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4595a6ad51044c75b39248277f862bb4 : FAILURE in 1h 09m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d2212e842884fdeb6d5f43794f3e2d6 : FAILURE in 2h 35m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89362ac518bd48deb8931ceae9195800 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4998ff4f6e22460a9b803758f6f7ea17 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb4014b3234e47f7935de7f5602b40f4 : SUCCESS in 14m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f43a43e494464f31b3749285ccb892ae : SUCCESS in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"364406f9014e755490eda363957c0ccd001c9400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-10-07 10:28:21.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"44f78cc90b8fed805dfd1829571dd48c581b5b7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 11:08:34.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162d504c7b1b40418b65d25716717512\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b5daa193f0f843029320f72c7c5c1835 : SUCCESS in 37m 31s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/10d04aa18bfc4ab1a7e089ad55ef6f38 : FAILURE in 27m 57s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b9a50c52f8b94e11b7592df971b1e822 : SUCCESS in 27m 39s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d9be0ab4d85f42579803e52da6ebc3c8 : FAILURE in 38m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e1645a885884dfda288bd9dde70a206 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a0f255123064ce5957dffd6fb1faa93 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6915795dac5b447b814f88ed2cbdd12c : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c4f5a01831c8989f8e2633240ba9b72791cd75fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 11:22:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f902cde06d4d16ab774f258c214932\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/109ea04f4aaa468193f863b2d73c91b1 : SUCCESS in 20m 05s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ca33b90014f423ab148f4551b2a8366 : SUCCESS in 22m 33s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/948124768bfd4644b982b5416dcd9a08 : SUCCESS in 15m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7ae15c1ad0e9435aaf3b23d16b119973 : SUCCESS in 15m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e9087b09a87c4f7591d147df376fcb0e : SUCCESS in 52m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf90e9e90930464098c535749d90abce : SUCCESS in 21m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4ca899879144f2e9c4b527ec9643611 : SUCCESS in 12m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4b51f14ecc8e4b978cbdc068bb8355f6 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdd93291d94b424b90e1947d6bb0a6eb : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/660f61051863494e96fdbadea12494cd : SUCCESS in 1m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82a172ad25f54bcb92ccdd25d936f97e : SUCCESS in 1m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99729a7184294f94a73bb3cc485d6f98 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a596a888b48a408da8768c1320493bca : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e817d49809e42f5bf36001f647cfaf9 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":15},{"id":"d7359775f2d08ee1b99b3e0d23b4fc6b0948dab1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-29 15:24:02.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"37057d208efeb03012146bfad7962e3b8ad3424d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-29 15:27:53.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"8ecb9b29ef159f074b11e4fcab358b34d9aff3a5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-29 15:28:15.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4b15ad7927017af0c8bf2bbc01fa67529b17e05a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-29 15:32:29.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"0116b03fcd348dc29a161ee0d13bf205f4f3893a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-29 15:33:35.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"1cbefebb7cd2f5bcba8996693d88808a7088a9c9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-29 16:09:48.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"785475fc6c4316016ea2bbdeb4af25bf904ba55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 16:30:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/beda2b5dd9e249ba80d91b096979c136\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e449f4d02cf142eb92c3537ce96e37ad : SUCCESS in 28m 28s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c910fc5bc0a44089b1fadcff3b0e204 : SUCCESS in 24m 43s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/6a468a9490df40ae8032ae17eb6547b2 : FAILURE in 43m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/477814f782be4d98bb4849400c31dbd6 : FAILURE in 37m 55s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/242cb0b5ff514e0a97ad81eb66376ddd : FAILURE in 56m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fc2133b12ca4cc8ab6d904bc38cca69 : FAILURE in 26m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a80c2401df54bc9b86c58097588ce11 : FAILURE in 45m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/78fe376d645b4fb287cbcb54622ca5b9 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fff4a7cb72414fb81e41d8c867467b : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d8f4e397b7c438dabf2f63beab97c4c : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6b126f67fd34424a2080fb27af2ab6d : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e17ade74e324f69bfe5b928a22f3185 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef09b2914ae4418aafdc0896d8ac4725 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":17},{"id":"fb1bf5b40ff24e4d4887fada22cbf012ccb92e04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 17:33:31.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f66618c16884ed2863d7b1e9fbbcb46\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9ef5b1c551c64ecd84679d3551b7a10c : FAILURE in 29m 09s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/e12d2e39e6de4fe999ea73fb81d56190 : FAILURE in 41m 41s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0ea9f447362b49868ccb8de7a61b43e8 : FAILURE in 42m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8c6f276c8a5541bb8779239645d00fb3 : FAILURE in 49m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5155bbb1f42e4bd0b80f782139a83ac2 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3826fbd3cecf47bfac01baa4cb3816f8 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/661df3cf069b4030a9f79e243abf37bb : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"55fe18bdd54f987b0cbf52afbbd3c5493e391c95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-29 17:42:03.000000000","message":"Patch Set 17: Backport-Candidate-2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e29147667dbeaca9749cc720caabdca01d9abc7a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-29 17:42:10.000000000","message":"Removed Backport-Candidate-2 by \u003cGERRIT_ACCOUNT_22629\u003e\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":17},{"id":"e1c7aa88c617f496bf1389bb7a62d2c89c168d2b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-29 17:42:24.000000000","message":"Patch Set 17: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a3acfe91a620a35759373e0abd7a0cfbb17e616b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 18:42:19.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8513d7e688dd4af391b42ce8a2e82ee1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d2f0dc0c50ba475ca90aa5ac43687415 : FAILURE in 23m 40s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce74d33d65ef43b29549082363bb9a43 : FAILURE in 26m 45s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/4ceebdbd7ccd4a0987933f63cac7768e : FAILURE in 51m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3d17d144f50c4da994870e05ef4aed59 : FAILURE in 49m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a29303ce4daf40ec9a2328592766df8f : FAILURE in 59m 30s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb59c514343a4c5394ee3253fb1ae38c : FAILURE in 52m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7598a98ce0044e4eaab775e2af7ebb55 : FAILURE in 27m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/60949865fbfd461e8b55d115554e77ec : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83fb8f9718840048da5eab0932dcfdb : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da86f2d6474f4a59b5a473c7727f383c : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eee7f62c9b541b7bada7188cf85dfe3 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8db8d55baa68488eb008b2018b04b536 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921d9fecff3d4969a97a1a0a05c66dff : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":17},{"id":"f2715c19040417b7320cfe9223b1e166240c9d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 20:26:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d032335e43f2439e87606ae8c1d3cc8f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/45eb196be5974dd2b53c946c2a0e924d : FAILURE in 44m 16s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/724aae6602a24ee7b5404751c75a5961 : FAILURE in 36m 47s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/cbf170a173bd44188e9ab52cdc85f4ba : NODE_FAILURE Node(set) request a8b4d37da6514ebab42f216b2348a916 failed in 0s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e03b29eb58874f1d9c1d8e19a44cee81 : SUCCESS in 43m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b5be5293f4f416189ef5f98c24cab52 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0ebe29292d54deba1e49115d4a8327f : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8264de851eca46709bfb347b3f64b06e : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"eca8b71fd712523eeae025d5af9c524395c82be1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 06:05:24.000000000","message":"Patch Set 17: Review-Priority+2","accounts_in_message":[],"_revision_number":17},{"id":"d89ac5db37c298f8389636cc305c304247067433","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 06:08:31.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"33211b0120b88176657a2e37f8638e7b7fcdb9cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 06:43:14.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0759514f315047a0a7dd4d146f81d9a8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/47bc36de0fe34fa180ab7bea639cd6d4 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/fbd4f7a0c8f54da1910b1d828d6c50b5 : FAILURE in 32m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/131e8780280247a9b4a643352db9fbf4 : SUCCESS in 31m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/543db982057a4191b5a7b434247dd7a6 : SUCCESS in 32m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e6f43383af3489dbd526f50d4a4d67e : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afcc699c708946cd8f7c1ba2c5f116b1 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77bb2908d8b7455989ec021bc9daad96 : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"378bd83d42d2fe81647eb9019e3677b66c822526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 07:17:37.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34df5fc5f9647b082610f4fd6e2a090\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ec111b4519424c4cbae40d500d74d9ee : SUCCESS in 21m 41s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/236ad17dc6fe47baa80b1ab64932d687 : SUCCESS in 19m 37s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/fc97760cf9ed4a789e564d67b9147be6 : SUCCESS in 25m 22s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b93a300ab6924296ba086e27c9f31a44 : SUCCESS in 22m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/376702371e0e4a47b4fe8e7d2bf7dcba : SUCCESS in 1h 04m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acd5c508262042f8b83e04aec40176b3 : SUCCESS in 23m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23fed94ac2f34df08902728fa8c09f2f : SUCCESS in 34m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bc06932c3bb4e34bb4807eed1d57f19 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8097f0fcc6524ead8dd9764288c94469 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1adbee617c2c4ac98fb865c9f15f9235 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6224ab8c21042199223186c18c1933e : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ed60a77d694046babe19043e8357eb : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9630c67fdc45be98dffdb1b3464fe1 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":17},{"id":"c6726f559d26ea9b32f29c6312363257fcd0bc84","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 08:05:01.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"739296c6d573ac834b661844c73d020b74f47066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 08:05:11.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"00156a035bde157f233434b9f5e867af5d83065d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:05:32.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c52a533be2044cbab5ab8dfeca8ed4df\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/ad155281d70b4cab94eaad59cc54c9ce : SUCCESS in 15m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/905b39e03d154c49a7e76cbcec093aa3 : SUCCESS in 27m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e9276957a95b49f6a9ab95f3f8cd2f45 : SUCCESS in 24m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e2800d12c66243a39dd07126e1625f61 : SUCCESS in 58m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/471c3688ec774a73aa07aefcde787bbe : SUCCESS in 22m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a20f99883b0c488f8fe9650df2872409 : SUCCESS in 32m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7bca55dbdfe64959bcb52c6a107d3b88 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f949ed2bd4b482fb699f3dda64409f3 : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/316e8e5c289244a78089622522bcb315 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2aa54f31475a4751b5653ec037b30dd8 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbf9940bcf64fff9f9502b254b2739e : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":17},{"id":"070d6c46cea5a8dddecf2e6affc02833631a0b9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:05:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"862c82e061cdd9e86089d573582e78829c717b02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:08:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c671a49e9340e4b707f74adb8812a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a87d1ac0284a2caab2588f5ddee94c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"af9dfb0dba11f0cb025a806792c3b7dcb71dc39c","revisions":{"de595be9aff3d905a1986801630d3fd80871f8b8":{"kind":"REWORK","_number":1,"created":"2024-10-11 12:14:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/1"}}},"commit":{"parents":[{"commit":"068790cddcad68f070b557b339158d75047192f3","subject":"Merge \"Bump proxysql to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/068790cddcad68f070b557b339158d75047192f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:14:56.000000000","tz":120},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de595be9aff3d905a1986801630d3fd80871f8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de595be9aff3d905a1986801630d3fd80871f8b8"}]},"branch":"refs/heads/master"},"973d2a5b2f09d158b42543ed71b7692b52e815fc":{"kind":"REWORK","_number":2,"created":"2024-10-11 13:03:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/2"}}},"commit":{"parents":[{"commit":"068790cddcad68f070b557b339158d75047192f3","subject":"Merge \"Bump proxysql to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/068790cddcad68f070b557b339158d75047192f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 13:03:36.000000000","tz":120},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/973d2a5b2f09d158b42543ed71b7692b52e815fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/973d2a5b2f09d158b42543ed71b7692b52e815fc"}]},"branch":"refs/heads/master"},"a3426e8d673a6d4ec247594e9f698f2547e39da2":{"kind":"REWORK","_number":3,"created":"2024-10-11 13:43:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/3"}}},"commit":{"parents":[{"commit":"068790cddcad68f070b557b339158d75047192f3","subject":"Merge \"Bump proxysql to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/068790cddcad68f070b557b339158d75047192f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 13:43:45.000000000","tz":120},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3426e8d673a6d4ec247594e9f698f2547e39da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3426e8d673a6d4ec247594e9f698f2547e39da2"}]},"branch":"refs/heads/master"},"ae23b93589951fb9c13ef62d26175a8e9537c59e":{"kind":"REWORK","_number":4,"created":"2024-10-11 15:20:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/4"}}},"commit":{"parents":[{"commit":"068790cddcad68f070b557b339158d75047192f3","subject":"Merge \"Bump proxysql to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/068790cddcad68f070b557b339158d75047192f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 15:20:57.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae23b93589951fb9c13ef62d26175a8e9537c59e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae23b93589951fb9c13ef62d26175a8e9537c59e"}]},"branch":"refs/heads/master"},"4972f8d97f845df34aaa3f7649305bf5689dadfd":{"kind":"REWORK","_number":5,"created":"2024-10-14 20:55:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/5"}}},"commit":{"parents":[{"commit":"068790cddcad68f070b557b339158d75047192f3","subject":"Merge \"Bump proxysql to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/068790cddcad68f070b557b339158d75047192f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-14 20:55:04.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4972f8d97f845df34aaa3f7649305bf5689dadfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4972f8d97f845df34aaa3f7649305bf5689dadfd"}]},"branch":"refs/heads/master"},"e5e657ef6305ca25897d4c2dbd20505d4e01a16c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-23 17:06:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/6"}}},"commit":{"parents":[{"commit":"e70c58e35d0c6c1dbeca09c07190347bd5db461a","subject":"Merge \"Update Let\u0027s Encrypt lego version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70c58e35d0c6c1dbeca09c07190347bd5db461a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-23 17:06:19.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e657ef6305ca25897d4c2dbd20505d4e01a16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e657ef6305ca25897d4c2dbd20505d4e01a16c"}]},"branch":"refs/heads/master","description":"Rebase"},"d92cdb3935512ab92875c784d40ff532300c46de":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-03-27 13:03:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/74/932174/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/7"}}},"commit":{"parents":[{"commit":"0b3945ad1baba0ae7a1b113c60b66abb40ff52c1","subject":"Merge \"Deprecate bifrost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b3945ad1baba0ae7a1b113c60b66abb40ff52c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-27 13:03:34.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d92cdb3935512ab92875c784d40ff532300c46de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d92cdb3935512ab92875c784d40ff532300c46de"}]},"branch":"refs/heads/master","description":"Rebase"},"eadf8502006efcfe2204209a50fc259968267f79":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-08-07 09:59:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/8"}}},"commit":{"parents":[{"commit":"0b3945ad1baba0ae7a1b113c60b66abb40ff52c1","subject":"Merge \"Deprecate bifrost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b3945ad1baba0ae7a1b113c60b66abb40ff52c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 09:59:24.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eadf8502006efcfe2204209a50fc259968267f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eadf8502006efcfe2204209a50fc259968267f79"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2d32a5e0ee2777628847b75a7579d7855ce1edf3":{"kind":"REWORK","_number":9,"created":"2025-08-07 09:59:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/9"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 09:59:29.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d32a5e0ee2777628847b75a7579d7855ce1edf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d32a5e0ee2777628847b75a7579d7855ce1edf3"}]},"branch":"refs/heads/master","description":"Rebase"},"1768646454da150462433b0bd397e925fd9cffc0":{"kind":"REWORK","_number":10,"created":"2025-08-07 10:00:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/10"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 10:00:37.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1768646454da150462433b0bd397e925fd9cffc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1768646454da150462433b0bd397e925fd9cffc0"}]},"branch":"refs/heads/master"},"822ece38153e64eed3a94fb6498478ca00b6ed85":{"kind":"REWORK","_number":11,"created":"2025-08-07 10:42:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/11"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 10:42:36.000000000","tz":0},"subject":"WIP: Add Valkey","message":"WIP: Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822ece38153e64eed3a94fb6498478ca00b6ed85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822ece38153e64eed3a94fb6498478ca00b6ed85"}]},"branch":"refs/heads/master"},"f5f3db7701c5f5a2541dee19cb82d7573d230e0a":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-08-07 12:16:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/932174/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/12"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 12:16:01.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f3db7701c5f5a2541dee19cb82d7573d230e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f3db7701c5f5a2541dee19cb82d7573d230e0a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4afe06160f1585259441c632c5c097c23468dad1":{"kind":"REWORK","_number":13,"created":"2025-09-08 09:51:11.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/74/932174/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/13"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-09-08 09:51:09.000000000","tz":120},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4afe06160f1585259441c632c5c097c23468dad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4afe06160f1585259441c632c5c097c23468dad1"}]},"branch":"refs/heads/master"},"94bdf0b988eefdd36eacf181548b7097f8e2dcf1":{"kind":"REWORK","_number":14,"created":"2025-09-08 14:25:02.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/74/932174/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/14"}}},"commit":{"parents":[{"commit":"8dbb71b70ce32b86c3a220439739f522605251f8","subject":"Merge \"ironic: Stop copying rootwrap since Ironic dropped it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dbb71b70ce32b86c3a220439739f522605251f8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-09-08 14:24:52.000000000","tz":120},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94bdf0b988eefdd36eacf181548b7097f8e2dcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94bdf0b988eefdd36eacf181548b7097f8e2dcf1"}]},"branch":"refs/heads/master"},"89e73c36fa619124c01e21928cd7820c9edc66e4":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-10-07 10:28:21.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/74/932174/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/15"}}},"commit":{"parents":[{"commit":"34923a33ec4edd234c43babf1e2a118b86fa8b83","subject":"Merge \"ironic: Remove ipxe.efi handling for EL10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34923a33ec4edd234c43babf1e2a118b86fa8b83"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-10-07 10:28:21.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89e73c36fa619124c01e21928cd7820c9edc66e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89e73c36fa619124c01e21928cd7820c9edc66e4"}]},"branch":"refs/heads/master","description":"Rebase"},"8116444bb799b30001afdc45b43813dfcd09e51a":{"kind":"REWORK","_number":16,"created":"2025-10-29 15:27:53.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/74/932174/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/16"}}},"commit":{"parents":[{"commit":"34923a33ec4edd234c43babf1e2a118b86fa8b83","subject":"Merge \"ironic: Remove ipxe.efi handling for EL10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34923a33ec4edd234c43babf1e2a118b86fa8b83"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-29 15:27:53.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8116444bb799b30001afdc45b43813dfcd09e51a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8116444bb799b30001afdc45b43813dfcd09e51a"}]},"branch":"refs/heads/master"},"af9dfb0dba11f0cb025a806792c3b7dcb71dc39c":{"kind":"REWORK","_number":17,"created":"2025-10-29 15:32:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/74/932174/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/932174/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/932174/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/932174/17"}}},"commit":{"parents":[{"commit":"34923a33ec4edd234c43babf1e2a118b86fa8b83","subject":"Merge \"ironic: Remove ipxe.efi handling for EL10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34923a33ec4edd234c43babf1e2a118b86fa8b83"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-29 15:32:24.000000000","tz":60},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af9dfb0dba11f0cb025a806792c3b7dcb71dc39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af9dfb0dba11f0cb025a806792c3b7dcb71dc39c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
