)]}'
{"id":"openstack%2Fkolla~932845","triplet_id":"openstack%2Fkolla~master~I4ae349dc0e9c606566388113ac53de42d5e54d0e","project":"openstack/kolla","branch":"master","topic":"add-relay","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-28 08:51:04.000000000","reason":"Change was abandoned"},"30463":{"account":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"last_update":"2025-04-28 08:51:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ae349dc0e9c606566388113ac53de42d5e54d0e","subject":"Add OVN relay image","status":"ABANDONED","created":"2024-10-21 08:36:32.000000000","updated":"2025-04-28 08:51:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"f367c789a9759030fb32dec25d82bb24fda73054","_number":932845,"virtual_id_number":932845,"owner":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-21 10:18:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-10-21 10:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 10:29:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-22 08:36:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-12-02 13:23:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"41004f318c111526c7f7d915a7627ee1ca54764b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2024-10-21 08:36:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5244a72370ecdb3bfbffe1adc015a927159daa95","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2024-10-21 08:38:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2f073aa08bef4d511bfb4cec05052f2be437cd89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 10:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b672cf685924ffaa9f2db02c798a6a8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/be053954663f4450b7a626c9dce62a8c : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d208681f405541628232b51276970200 : FAILURE in 6m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e36ac578a3b4866b2cb4ef5c89877c4 : SUCCESS in 14m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/587a2fd9fa3f422c9cd58543e69c0b6b : SUCCESS in 29m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a7daec5e07354c69987ba4962435dcd5 : SUCCESS in 47m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1efc464bc6e8449b9b9ed2fc4d72e165 : SUCCESS in 1h 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0686a0c9b6854aa4a10be70bb13aa9fb : FAILURE in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5431bdc63041434194b8447b27853eef : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d1b95768bb94baab7b9f60c2c96101c : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fad629522af481694110747c1085957 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071b651c157f4095b5220183931fe143 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2b16c427f8a345e28525e948aedb2167 : SUCCESS in 26m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b967a6fea0544bc885a0f0e08ce68bf7 : SUCCESS in 37m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ddfdcfc313c440c69a99e355767c8c91 : SUCCESS in 53m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3cec3e8361074087a7bf70473e663795 : SUCCESS in 1h 34m 59s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dabe4610c3d14b70b822d296b354fc12 : SUCCESS in 53m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/44bbc20567534973abeff39442a65b60 : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/61702efe7011465a9a6c5ee8a68a80e3 : SUCCESS in 22m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/34532b8ffbd340f0b27c99f88bb1da4c : SUCCESS in 25m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b82a9e82e18d4769a3fa69245727c612 : SUCCESS in 1h 01m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4955115fc6614cd4b47beeeee17de41e : SUCCESS in 1h 17m 35s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e3f82df5259f4a3cad3a1f2f4f79b8b9 : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/101a1f21e1b044b3be3ccaedf6b9b108 : SUCCESS in 1h 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"672560029aebdbff5912acf65607ce674d3f90a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 11:47:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668d8ca496e84984852f251cd969c643\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0ccc4f70b3e94fb18a7e87ab4c1c1eb2 : SUCCESS in 2h 45m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb46da397fbc4a32a9da77730fec0673 : FAILURE in 1h 25m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0fd25a1b032e45b8a050fab1d581c1af : SUCCESS in 1h 26m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/479cce86779f4b29beae1f38362d22b8 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e5f8c9ea32c40e7b76c6536c9b95b6f : SUCCESS in 5m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ed669c9386e8478ab5c68e45a5633f08 : TIMED_OUT in 3h 01m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f257957212e842b8a2d5e337d1eaa59b : TIMED_OUT in 3h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7fbc9fb87da093aa1816043ca84cc82f4adbc07","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-22 08:36:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f367c789a9759030fb32dec25d82bb24fda73054","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 08:51:04.000000000","message":"Abandoned\n\nAn alternative version of this changed has been merged.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc5826f1898484535865d76ce7814a7f746590bb","revisions":{"cc5826f1898484535865d76ce7814a7f746590bb":{"kind":"REWORK","_number":1,"created":"2024-10-21 08:36:32.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/932845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/932845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/932845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/932845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/932845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/932845/1"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","date":"2024-07-23 13:54:12.000000000","tz":0},"committer":{"name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","date":"2024-10-21 08:35:58.000000000","tz":120},"subject":"Add OVN relay image","message":"Add OVN relay image\n\nOVN relay for it\u0027s connection configuration needs\nconfig file which is supported since OVS v3.3.0\nbut ovn-ctl still does not support it in latest\nrelease but main branch.\n\nWe are building ovn relay image with newer ovn-ctl.\n\nChange-Id: I4ae349dc0e9c606566388113ac53de42d5e54d0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5826f1898484535865d76ce7814a7f746590bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5826f1898484535865d76ce7814a7f746590bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
