)]}'
{"id":"openstack%2Fkolla~934313","triplet_id":"openstack%2Fkolla~stable%2F2024.2~If91e0330dc149143c82d2183b8ddf6fa9f68d80e","project":"openstack/kolla","branch":"stable/2024.2","topic":"fix-config-json-handling-stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-07 11:58:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"If91e0330dc149143c82d2183b8ddf6fa9f68d80e","subject":"Fix handling configs in base image","status":"MERGED","created":"2024-11-07 09:29:58.000000000","updated":"2024-11-07 13:43:33.000000000","submitted":"2024-11-07 13:42:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934313-fix-config-json-handling-stable/2024.2","cherry_pick_of_change":915440,"cherry_pick_of_patch_set":24,"meta_rev_id":"b7a53f2efbace20dce6ad73637f46914d7f37cd1","_number":934313,"virtual_id_number":934313,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-07 13:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-11-07 11:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-07 10:52:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-11-07 09:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-11-07 11:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-07 09:32:14.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-07 09:29:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-11-07 09:32:14.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-11-07 09:32:46.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-11-07 10:52:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-07 12:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f270315651859244f3d5412dcab4f6dee50a4a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-07 09:29:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b464291ef07dace0c2c538f548b5e309b633b1de","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 09:32:14.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"12b916fa1e371d4984bd9a7a52f990bea28b7d9d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 09:32:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b092a263fd90d2b6e8c7e8d1962e1f693b09c2e3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-07 10:52:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"78215eb136726acadd6af6e81bf0f7249723db75","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-07 11:58:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c98892ca2dc3fb622f3d27662a5faef75e9b9571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bc37e57145441b4b5428706aa778da1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bbd7d23a6023405f99bc8b5813c038d5 : SUCCESS in 19m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8f25b5ec77934614b8625bc04dc07ea7 : SUCCESS in 58m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/78cf20efeb39498bbcc403a7907ddbed : SUCCESS in 20m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/021057f57c3c4f9a80418334634d9a7e : SUCCESS in 25m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ab1d44bc22ff446c8695684a247969e8 : SUCCESS in 1h 01m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/14e036df8fb1407ba4b20e530666c807 : SUCCESS in 1h 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f4dc802b874b118435524a79611c8d : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad3b996aaa8a4b7ba939bb00d9700642 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f817b47f1e0a4a1bab16ea3e2393fc7f : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da70588ac7904513abb29f92785f78e9 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3159061d1d254eceb4386ebd04604772 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3776ce583764fb2ba3aeed8f817111d : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6839c4c982ff48a19aa3aa6eadd5f60a : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b06aa740099415381f12584838386d5 : SUCCESS in 20m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/44b1caba484e489b9cfdb474af03c27a : SUCCESS in 37m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3778ffd3d7334611bbd944f82d666841 : SUCCESS in 59m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a684b39143334327b3c64a194f501f4f : SUCCESS in 1h 37m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d93920bc99584ae59cc347cb67d65bf8 : SUCCESS in 1h 47m 43s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9dc1cde8e5543de9526c854ccd0c08b : SUCCESS in 2h 14m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d23ff9f43c7440139f164bc4c11406b9 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/cdb5283d151642769874e30be05e38cf : SUCCESS in 1h 24m 13s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1cb733e13abf44f9ba468e2137ebc046 : SUCCESS in 26m 19s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/3330622e5d614e06916e2d184bbfa301 : SUCCESS in 59m 28s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/69d1e13f457f483a8604b913b8ea7927 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cdcd9718cc424c489213466a01891b86 : SUCCESS in 1h 18m 14s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a7cc0b944c04e5caa1d6ec6b56a429a : SUCCESS in 29m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f29cfa7d7ef24bc5b7058c8f15b34da5 : SUCCESS in 23m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4da1205fe4d14d97b7475b12d8ff1fc9 : SUCCESS in 36m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee86435c5c624b9691853a42b0a590b6 : SUCCESS in 1h 00m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/01ad1854b9ae41d09914d95c831728c5 : SUCCESS in 1h 30m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b8ac551311f74452aec43d32e968c63c : SUCCESS in 2h 11m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b3bc6135be66412b931eae21803fdfc2 : SUCCESS in 2h 12m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2d69641b0b774f6bb1f733c68370097a : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f50ad84d345b46f5a6494c9486860101 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9e4421af5da348e6a1cbec829db2953b : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c3a1b42d6b3a42aaa3dc0b717d05ea93 : SUCCESS in 1h 31m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c1c19cf3eff34c5cbef659e8b5c6f06c : FAILURE in 1h 25m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/73f480d1d65d4591a7008a5787cac66b : SUCCESS in 1h 31m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c3af6c204ade4f1dbad82bb31a93a8bd : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8b3b5d2af45e34b14c639685f3fac881b85473c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:16:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e01517aacae18d8fe8d353879c9a70477b93019a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6789dab4aa244a7395110ae4435227a7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad39532418034114873c6851c22a23a8 : SUCCESS in 1h 53m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f2e9a20a392c437a91171050bf647059 : TIMED_OUT in 3h 01m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9cea7c8a469340b89f501647d15c3a4b : SUCCESS in 2h 21m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0371aa8dea71463ab8c1688ae59d27fb : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b45e873eaf6431785b406617ff331ea : SUCCESS in 4m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/86a0554590ee4516b0a5cce688978ee2 : SUCCESS in 2h 12m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d51818d18eb041d19d30b9382f13c136 : TIMED_OUT in 3h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ca23e7dc4e5410fe45e1b4b1556bb063eff0144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 13:42:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5349b434d72d450b9043aac6d4fe083b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b1f4176db9441a780025f070cb2233a : SUCCESS in 15m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f70def43decd4bcfbc907792d5578cb4 : SUCCESS in 28m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d122b1c414684f32bb660870c813dac2 : SUCCESS in 1h 00m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/91bb7e1e6baa47bbb6ff7bacb2ae6aa7 : SUCCESS in 1h 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786313a931db4a2591c2967fafb05582 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26de73614a0d4c5cb717188c735b1013 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ac75ddeab644b898cde93b1437ccf7d : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50111e515bee4f5c935eb2784afd2e60 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f478d3861ac49dd9e4a8c9beb2c8116 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/29e74aed12f54a35b71f4ceb533dfe22 : SUCCESS in 33m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f6f5414f671480e8107b50ec733badb : SUCCESS in 37m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07cb53fafd7549cf907b345e773c23a2 : SUCCESS in 1h 01m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d4671484c5dc46e4b70e968d12ffc07d : SUCCESS in 1h 15m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7cde13e4091a4243b2c117f1568d107d : SUCCESS in 19m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4c39b56a25b74d28a413ea48d6ade528 : SUCCESS in 24m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/01b600d700704262918389a4a483d8ee : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9b97f27e10d04be19e77340ef823f902 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"009b65ded97d319f7e444cf85b0e9671ca051e72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 13:42:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7a53f2efbace20dce6ad73637f46914d7f37cd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 13:43:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/469917a2985f403d8d00910f2b800817\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9720325750b24d4890033918786d4cc5 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728d2ee38c354116a6b0ba4f86303085 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e49ba16fcc930ac54a610f7007ac40d5580ce9fc","revisions":{"e49ba16fcc930ac54a610f7007ac40d5580ce9fc":{"kind":"REWORK","_number":1,"created":"2024-11-07 09:29:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/934313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/934313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/934313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/934313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/934313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/934313/1"}}},"commit":{"parents":[{"commit":"622a90c164ec63548ba56db0deb7570f2f104179","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/622a90c164ec63548ba56db0deb7570f2f104179"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-07 09:29:58.000000000","tz":0},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n(cherry picked from commit 36c12676ff185f524a7f4d2f2f4c70b5d60462b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e49ba16fcc930ac54a610f7007ac40d5580ce9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e49ba16fcc930ac54a610f7007ac40d5580ce9fc"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
