)]}'
{"id":"openstack%2Fkolla~934315","triplet_id":"openstack%2Fkolla~stable%2F2023.2~If91e0330dc149143c82d2183b8ddf6fa9f68d80e","project":"openstack/kolla","branch":"stable/2023.2","topic":"fix-config-json-handling-stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-08 18:36:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If91e0330dc149143c82d2183b8ddf6fa9f68d80e","subject":"Fix handling configs in base image","status":"MERGED","created":"2024-11-07 09:44:49.000000000","updated":"2024-11-08 20:15:39.000000000","submitted":"2024-11-08 18:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934315-fix-config-json-handling-stable/2023.2","cherry_pick_of_change":915440,"cherry_pick_of_patch_set":24,"meta_rev_id":"e3cd24ec039e10302db17517c03afd179997d360","_number":934315,"virtual_id_number":934315,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 18:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-11-07 10:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-07 15:01:51.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-11-07 09:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-07 15:01:51.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-07 09:44:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-11-07 09:51:25.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-11-07 10:52:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-07 12:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-07 12:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 15:01:51.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"8c60398fb53ea3611542d892328599628920faf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-07 09:44:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d161fcbcda08c3fabcb3e8bffa1937964a99c057","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 09:51:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b88f85cf25871a73a444e920a512271d8ae75d45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-07 10:52:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d8928f45284b13275631b3a565d5ea9e81b18806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6777b29cc04d21ac163a0c421e6802\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d2d96814ab3b40beb3a5cf0e233933ea : TIMED_OUT in 3h 04m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c78c70767cb4d3e95d841865e9f6b5d : SUCCESS in 2h 36m 39s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb540e0f50374e3bbbe2e6b6be7773d6 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d45cedcae7e4f4fb170e611420d8ada : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1db21cb14a74c2396a139050bcca88b : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0000c6fa45bd4034b0be9d6aac787c3b : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64166a90b53f2d43ebd9f05a3650cfd8d227597a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3e13202c2449a9b5e07e5d1e356064\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/683f204cdf2f4b3e8d1624dacdd4209f : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b4192e4478fe4fee9e54d5cbe0205300 : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0f3eaca4777498c935ae7867e929e82 : SUCCESS in 18m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1d3f160f7dd94d30be109f1df507155a : SUCCESS in 28m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/da81511c0a4c4a77a41098a9716f350f : SUCCESS in 1h 25m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/82058c740c6c4f6aa0aff8f21a3014aa : SUCCESS in 1h 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b763cb35fd924f1abb428cc541b57823 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/002af08f5b60481bae5c28acab715380 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4515a19358f49b8adfd49aafbfbc4ca : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24b0c7c6b9fd4e1aa7e627a841279bbf : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f62392d1fc247f59a41bfb8c9df0b57 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bdae9a6c27546858df73e6555135455 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ade2fbc2c947f38eeac47bfa59c276 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e48c74bdce144959491b16d33339cf0 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4771a6d539c54c28af68d4e113f91438 : SUCCESS in 24m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c20caeae14234dcc96891198dcbf729a : SUCCESS in 37m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5151574ef4c948949e617e67b675ba0c : SUCCESS in 56m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/107ea33a22f84d88a42db1559236f12b : SUCCESS in 1h 21m 35s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/10482aedb83e42399352130a68587ad4 : FAILURE in 1h 13m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d76eaee445014af4a5c7fd333a4200c5 : SUCCESS in 2h 25m 46s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3b76f98aa4334e1aadb470f3131b77b3 : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8b60b69fc35447d8b5aa50053d3bbe7c : SUCCESS in 1h 17m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3bf520f1ea474f02a774723d3eb606e0 : SUCCESS in 29m 00s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/638bd8f31bd746ef93428a252b04a1f4 : SUCCESS in 55m 28s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2f2c173d629e4973ae2bb6415e59042a : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d9a883f876ad41bd94ea6d72a119c489 : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8aa7ffaf03e2447e9a77cb73f499564b : SUCCESS in 32m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98b85f9f0aa946c2b75a34d5ec9391e9 : SUCCESS in 21m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/835f4a611493466f859f8aaa52809789 : SUCCESS in 1h 12m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/231ba5c62ba948e5be500d27e3747c86 : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e4122e39fe334ceabc0bd8675b6d1130 : SUCCESS in 1h 42m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f4e5e869e3c6454680dceed8f22e0170 : SUCCESS in 1h 31m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/952bb9a7f1f647e5850760adf1b75671 : TIMED_OUT in 3h 03m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/91e2b53c54444b07a12b20f1d4e14ba7 : SUCCESS in 48m 24s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d0b734575cc045ff9ef798a50bd58260 : SUCCESS in 1h 13m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/64547fd9edae4d57b535a95e6c49e76c : SUCCESS in 42m 52s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a2b5250df62a4c48a467715d1262bed2 : SUCCESS in 1h 30m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c1c4f82de3ae4d33acb3678eedf66a82 : SUCCESS in 45m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ec0c757ed9a54aa3a19aedb7a4250a6e : SUCCESS in 1h 21m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7b30321bebbd4963b103ebca7510097f : SUCCESS in 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6e489566daef3f25cc6ca6d384779dba4f3da19","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 15:01:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80826a43146c0da0faf1308c7612383aa6335c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 15:02:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92f830675705439a370ed390d989bf1ea8cc79f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 17:04:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb9ab62d5c3f43fcb13c79f3e80c6687\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2b1137e195984a3cab02593250886b1f : SUCCESS in 17m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d6936bdaf82f40c4affa4dc00f1247c5 : SUCCESS in 27m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1df4b22619a64e4fbd69c6f79a67fabf : SUCCESS in 50m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5adb95c127ee4045949429d8bbe8e276 : POST_FAILURE in 1h 43m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78b93353a064b1d90bfeb8164a0f96a : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd5f5d9bf420496da6b2946f731590f8 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b02b01d574594a83950dc8aa41ec14ea : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe7f075950254e11a2c4e0e527b56a9e : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27b8bbd67fa4c99a91c6a75fd8d28ed : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a4a3504c6f24f66a329a477ad7a2e72 : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e4d57697d63448d88b51af35be3b5345 : SUCCESS in 27m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b663d509e77a46b0a4d19093cb218d53 : SUCCESS in 37m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c032e012eb2049b98714d8566b5c1f68 : SUCCESS in 1h 10m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/849814ef7c9e43068db29a5c715cfccb : SUCCESS in 1h 23m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a9604c581f1e4e24b843576aff60d858 : SUCCESS in 21m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d00c092a61a447a29ee2a76d83bbca55 : SUCCESS in 1h 12m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8c828ca50824452fa2d2f69af7727954 : POST_FAILURE in 58m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/66d15d9bfcec46eea33b5009f0816047 : POST_FAILURE in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1b56e46ffc253ef4ad6f13fb92ff2a584a4de399","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 13:49:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"754e1fa0cb12744a3d01c67073ad9d5fb6fa1a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:44:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8867501b0d4bb7991922f00212d254\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0bc3af0de189474dbc34ad25e6de685b : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/91826855df1a448bbd1a3078dbe55dd0 : SUCCESS in 1h 06m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5199897ebbac4b89a5ba47483edb5497 : SUCCESS in 19m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/882efc6a79c44b11b275ebc33e4510ef : SUCCESS in 26m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dda65cb0ce054d32956814ac78123ce6 : SUCCESS in 1h 04m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/89570005cc0f444b8e1d6dd940b38cc5 : SUCCESS in 1h 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0627ec195547c99e4b9fca43d518b9 : SUCCESS in 2m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e104bdf433fc44a2b149b8e15eae8010 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1f4c1741b3444a9683a0a27404b7db : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e36529b705864dceac87ce00e971a0be : SUCCESS in 2m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa0f39bf729f440bb7ebb1ae1de3b90f : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6810928a5688487cb1425115b35a639e : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c2ef7eae9a490e8dad6643328086a6 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e10b5de21f6c4a6e880013cc514a8ee7 : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/673f35e25d844644965866a47396830a : SUCCESS in 27m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/471ce875f96949b281a886db2140e902 : SUCCESS in 26m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f7efe9504f840fe906a0cca43371739 : SUCCESS in 1h 09m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/be429eda28f646b88038c5f153d2bd7d : SUCCESS in 1h 36m 18s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/13f7ccd2a6b44c77ae5d1855e63e214f : SUCCESS in 1h 45m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0892ec71463f46c8b18824ba4f3431bf : FAILURE in 2h 42m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c6dcd27b6704485cbc982efb4ccc499b : FAILURE in 50m 19s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d589139f61854b75b553993b6b2bc841 : SUCCESS in 1h 30m 04s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0168a50f95da41f99d537eb8ba6b3e04 : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b30469d4797648f38de4c456fe871246 : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/30f40edab2d14e1a8c73d7e79718caaf : SUCCESS in 1h 07m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/75f272a2b6854945b89042e5967fd9a4 : SUCCESS in 1h 28m 30s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f74aaf8c60284cd788d85e7385b5c7d4 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1242974e10e341a982c61441c3182e02 : SUCCESS in 15m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/18d67eede28144c88a93328cbc886d1e : SUCCESS in 1h 08m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3bc3036c040d4cc4802bbcc159834a4e : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c46a19fa167548a8a92419a59d389246 : SUCCESS in 1h 28m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d1169b8936f34439b28c5253796e5d54 : SUCCESS in 1h 24m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6352a70fcbca419fb86e2c38cf113122 : SUCCESS in 2h 14m 30s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/83deb28db35b4017a5ea830ed97e276a : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/eae1129d7656413f89d9b791f6653c66 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0d997993625e4f19909e6b046b4e6b74 : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b3a1aabc4b3648d985c1f4aa145d5264 : SUCCESS in 49m 18s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/320128fe0c39424fac22b5835673875d : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ae6db139216849b1ac628ec44d3f73e5 : SUCCESS in 1h 14m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c90b725055c04357977b9033dceba1ac : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7285c90639ead54f17e8b5033b3923af3b4ee7c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:45:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c57cacda9c692f0fae9e566d471e98932d4c615d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:36:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b3579ec61f4928aaf034c2463668c2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e460363a601c49aca9637795dc38eb07 : SUCCESS in 17m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b8383015342d4a24b65a9dbe1202602f : SUCCESS in 33m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/84c4de1909bd4e07b02f45747d577886 : SUCCESS in 55m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d971f8ccf55b40e098d5501efb527bbb : SUCCESS in 1h 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935e2d15837a41558cc9413ccf4ee94d : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f2e8e8e236648bfa1915f0ca018bfd9 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78b8813c12ec48cc96f97156c74ac7b2 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b29193c09f34c199786f48d627825af : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daddbef87a3c4b23bb60eb3d67ebbc38 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52826c2253394593ad22c2fc60763715 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13c1165da6374405a453a6de91724387 : SUCCESS in 27m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b6d1bfc82af04d6198d5fd5968e0f10a : SUCCESS in 38m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6d182668a1bd40cd9095b0c86b6f0c5e : SUCCESS in 1h 06m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/abfcebd5692d489eac2d3d04a41d1e52 : SUCCESS in 1h 44m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdd325c9f11a423598150aa1800112a7 : SUCCESS in 16m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ae27272618e846fe9dfe9255bc5083a5 : SUCCESS in 1h 10m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a1ed0207ab0a4daaa0e84e295e5ba373 : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ccaccb27145d4cf4953b89a85f2948b1 : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0479d5c95f517405424ab2c036df849c1ec26f18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:36:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d1f3fb5c792989db8cb925926679cb3ce0e5e09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:37:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1067213ac8674d15975d9bf016f7f8de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bc579a74bf47ee8546cd665b5791b3 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fc49ae2a15d4469806af3b04576c173 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"e3cd24ec039e10302db17517c03afd179997d360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 20:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65112aecb64c43d7845dc98fe2d01f44\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90a88d66fe224f01b44e2ac070914e62 : TIMED_OUT in 3h 04m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/76790d572b074afd85c94af8864b66d4 : SUCCESS in 2h 25m 09s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b21c6b973bd4c57b62412c67f86590b : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/720a8c58c9aa468db21e46a875f68568 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6b04b2d1637647ee8ea4fd47334de485 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d64e5dd699bf4ca08f2b587f57c95fb6 : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3b7db854201f07c81591a74527a026ca573ba6c","revisions":{"b3b7db854201f07c81591a74527a026ca573ba6c":{"kind":"REWORK","_number":1,"created":"2024-11-07 09:44:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/934315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/934315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/934315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/934315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/934315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/934315/1"}}},"commit":{"parents":[{"commit":"5a2189c4e65a956400f4dbf02891999001414382","subject":"Fix missing tss user in Ubuntu/Debian nova containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a2189c4e65a956400f4dbf02891999001414382"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-07 09:44:49.000000000","tz":0},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n(cherry picked from commit 36c12676ff185f524a7f4d2f2f4c70b5d60462b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3b7db854201f07c81591a74527a026ca573ba6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3b7db854201f07c81591a74527a026ca573ba6c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
