)]}'
{"id":"openstack%2Fkolla~934346","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ibf7e40e8b059d733e114963022df06180249c650","project":"openstack/kolla","branch":"stable/2024.1","topic":"mariadb-proxysql-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-08 13:47:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-08 14:56:08.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"Ibf7e40e8b059d733e114963022df06180249c650","subject":"Add way to backup from MariaDB replica server","status":"MERGED","created":"2024-11-07 16:31:07.000000000","updated":"2024-11-08 16:37:58.000000000","submitted":"2024-11-08 16:36:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934346-mariadb-proxysql-stable/2024.1","cherry_pick_of_change":928956,"cherry_pick_of_patch_set":17,"meta_rev_id":"4d01e1911a76741030f47213b0c72fce24e60b62","_number":934346,"virtual_id_number":934346,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 16:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-11-08 13:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-08 13:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-08 14:56:08.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-08 14:56:08.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-07 16:31:27.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-07 16:31:07.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-11-07 16:31:27.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-11-07 18:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-08 13:50:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"73fec34c8d9389f55ac959899f2eaf5ea071b3c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 16:31:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5ca82040055d9fc542cb4a504fe3dfe9f828fb3a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 16:31:27.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"924bb36df0f1264d63e3c00850f320a2868c3f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 18:49:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e61632abdc340a2826bbc18d75abd90\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b947b9c7bd104c8facb9c697ebb07c07 : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2682300f31c142efad5cf267d2f3e734 : SUCCESS in 1h 06m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f877df7fda0a4ab19e556568a658b07b : SUCCESS in 18m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/15409a012d144f0189a432eae5769610 : POST_FAILURE in 32m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15b07ee752254b42b52c75aa42f92ac0 : SUCCESS in 56m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0bdf8b3526894c6992e0de0b62316d3c : SUCCESS in 1h 26m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b3039d572d46c9b8150afcc9b12184 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d15c51883b1f4606940765d8bd1f04e1 : POST_FAILURE in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5eff47986c6b4396a15562cc2e66110f : POST_FAILURE in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9e9802175c41e59255d4124f3e8274 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25dbc4f9286b4d8da88fb7f8892bcc30 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ee06302fe94e1dbef2019693e8eb1d : SUCCESS in 3m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/307e55d8e0b540dcaa36bd78adb856f1 : SUCCESS in 27m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/69bb8c5aa919434981ffbe144ba478b7 : POST_FAILURE in 29m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/958a31c32e64468fb7978c15834fe6fb : SUCCESS in 1h 06m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f0f82db5cb614b4d898da16bb038cea6 : SUCCESS in 1h 32m 46s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f89b07ba9d9749218edb9b6d13f12e87 : SUCCESS in 1h 48m 25s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5f813e8c3ee243c095b9bbc47c80d911 : FAILURE in 1h 21m 01s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/277f031c79c2452b95afa6b034e47c07 : SUCCESS in 28m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71bcdacc195c47219516dd93fb417474 : SUCCESS in 18m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/97a69cb5c4764c599c45cd18d928c063 : SUCCESS in 58m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9105775d7b5144cd9baf02dd18e897bb : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5f5c9cb2e21a4c909ff097814941422f : SUCCESS in 1h 13m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/755de75b4c1c4c6d8bb8d5299d60dc1f : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3041e2662cbf43dd9f8482e78f0667a6 : FAILURE in 12m 07s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/649543a5e1f04e4d89cee26ece23dbe0 : POST_FAILURE in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86e8b1e2b28390951ddd2ce53e02a4285c6a7645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 00:59:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050391d5e922411286674cf0b0c9d2ad\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02da11ebeaa64a9ebab8be2135a5e65e : TIMED_OUT in 3h 03m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3586a5364757444390d564e42915a247 : SUCCESS in 2h 00m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7454fea9291f47e5a931de1e5ff22026 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbdb95c5b318494a9b7ef857036b4846 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7451ce914a3a3007850da02cf59ca712c3e45b19","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 09:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c75a7b5b3ab1fae4f0ccabc7da1dc1162390d261","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 12:10:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3b9c97732a4450a254fc1e4f8389c9\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ad750523d494cda9e8884046f9a9662 : TIMED_OUT in 3h 04m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a8353ab77de41c7a0f86e29ec2c942b : SUCCESS in 2h 11m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc589016d07844e9951f3e511580d887 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d06b2789594f429a96ccf50d7670b88b : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"072192092a81a132dce5e09a4bd465c9c8a4a6bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 12:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3088538eff9c469fb43977d983f466f7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/25e79479f7e74a22aaffcf8722a6b6e9 : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d7954df3d0e343d8ad0150e947c55d9a : SUCCESS in 57m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/91ae6c6a47c946a88096391f939cd173 : SUCCESS in 24m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c66d6ecc47824af4a41500d0f0e3e4ad : SUCCESS in 29m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d1349ad9826f4286a15e5e9b4e90b528 : SUCCESS in 1h 11m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8c2342117bf847ac9ed97edf2ef6dd2c : SUCCESS in 1h 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b2ce832b7047b28fb7bb30d452b12c : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b81dce25b4b4eb7a7f08a0afb67d99a : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5deb0442bcb44c0bbe457fd008626c95 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3205ef5cd034154ad8b5b4a8e3da419 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e8258cfb3894bd9aac5f8b38ac80c85 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b6c089d5274743a56961bc53f0309a : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/24c9fcd452894523b64fc11e7890b1e6 : SUCCESS in 24m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ca95584ae7964b4a82a366d846cc89c5 : SUCCESS in 29m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a53881d57bb4b85a7a82caaaef14435 : SUCCESS in 1h 11m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/94562febcf0f4bcf82aea3da4366a8ec : SUCCESS in 1h 23m 19s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a139e19bd3f44145ad8dfe988060f8cc : SUCCESS in 1h 50m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f6303862395e457584c4c0cacd0a3278 : FAILURE in 1h 29m 23s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5a71a9f8e2c34d00a350591764372767 : SUCCESS in 27m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e631ff085ead4fdf84c9166feef5f2fa : SUCCESS in 16m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/95451a15457248b584265190bba9f3bb : SUCCESS in 49m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b3098497da75493b9c4cd6af39619ef9 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d5e23a525d074033bcb5d5971c794ba2 : SUCCESS in 1h 22m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ea40994be2ba431485a82ff835fdfe70 : SUCCESS in 1h 41m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c3fd5fddbbc74a768ef3ce85c30eaece : SUCCESS in 2h 30m 33s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/98c55c16ea03412aa715c5d1f981a8a0 : SUCCESS in 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7bf197349e2208802cc30327a957bea5bd995f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-08 13:47:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"34725d4c37e8c68d662287674800bc82d7f5e737","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-08 13:50:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"837533c92aca0ded1214fddc7f681ddf46d9b8e4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 14:56:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04076b23895cdf26a9223b83b48e7172700f9c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 14:56:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"076ab0f1a2ceaf741ed1a179febe70866db17caa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:36:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59fae7d4e7a47d1896e3ced0079b3e9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c96fb621bc3d492fa5a9c7eda73c4085 : SUCCESS in 25m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/91317d0c93dd49779ce59359cc6a21a0 : SUCCESS in 30m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/badc5acaaad94949ac554783acaa95e4 : SUCCESS in 1h 12m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ed339c1a1e3b45978bc5ecb5499f3c39 : SUCCESS in 1h 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa14fb7bb0e48b3ab044597b8271d01 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f85628a42242a29aedd639aee83c87 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cfb4324ca0f4b56945dd935a142df89 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1531e8d87ec245e0a08a6739ce540269 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0554e40bafd5415cba850599644ad938 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0da4f0a5c96d497088127ad505fdfd6a : SUCCESS in 28m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/15ef887bdffd4bf38e7f340bc1f46e05 : SUCCESS in 34m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ca472b44eaa3424a9d2524e168b75c3f : SUCCESS in 1h 02m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/30411a8499fa4b0f9fc95afea83602c4 : SUCCESS in 1h 34m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/65f552cd38a548bea2203b61f2983945 : SUCCESS in 21m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6103308db2ff41f3b9e01dfd83178807 : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c760d54c891e4c6f89ba152b84f1e516 : SUCCESS in 52m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/14e09add82db4052ae5cfe460edb7b10 : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"105419cab6fc703d0b42609831100eb479d39889","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:36:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d01e1911a76741030f47213b0c72fce24e60b62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:37:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc6f8e0ceb44e92811819a88bb8b5c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1b2693a3cb471bad056c2071f007da : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/337b3475201c491ab414b0d4f5354301 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68e96e72f412afe793e103874fe67e32079ea670","revisions":{"68e96e72f412afe793e103874fe67e32079ea670":{"kind":"REWORK","_number":1,"created":"2024-11-07 16:31:07.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/46/934346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/934346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/934346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/934346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/934346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/934346/1"}}},"commit":{"parents":[{"commit":"648be98c9b396797a25980e8c111063e716b3846","subject":"Fix handling configs in base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/648be98c9b396797a25980e8c111063e716b3846"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-11 17:20:54.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-11-07 16:31:07.000000000","tz":0},"subject":"Add way to backup from MariaDB replica server","message":"Add way to backup from MariaDB replica server\n\nProxySQL reports its version to MySQL clients,\ncausing the backup process to fail when ProxySQL is used.\n\nThis patch adds a new script that handles backups from\nthe replica server directly, without modifying the original\nbackup script. The new script retrieves the server address\nvia the load balancer and connects directly to the replica\nserver to execute the backup.\n\nMoreover, this reduces the risk that the backup will be inconsistent\nor that it will block OpenStack MySQL clients, because they are\nconnected to a different server, and no one is connected to the replica,\nor that it will overload the load balancer (such as ProxySQL or HAProxy)\nwhich listens on the VIP.\n\nCloses-Bug: #2080818\nChange-Id: Ibf7e40e8b059d733e114963022df06180249c650\n(cherry picked from commit ec4c4310dd870ff5bbdfd9762a0beb8389009b9f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e96e72f412afe793e103874fe67e32079ea670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e96e72f412afe793e103874fe67e32079ea670"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
