)]}'
{"id":"openstack%2Fkolla~934735","triplet_id":"openstack%2Fkolla~master~I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c","project":"openstack/kolla","branch":"master","topic":"fix-bifrost-upper-constraints","attention_set":{},"removed_from_attention_set":{"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-02-20 16:15:19.000000000","reason":"\u003cGERRIT_ACCOUNT_35010\u003e replied on the change","reason_account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-27 11:25:01.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-02-07 22:13:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-03-27 13:12:48.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-02-05 15:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c","subject":"Change Bifrost Python upper-constraints source","status":"MERGED","created":"2024-11-12 10:30:17.000000000","updated":"2025-03-27 13:13:57.000000000","submitted":"2025-03-27 13:12:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934735-fix-bifrost-upper-constraints","meta_rev_id":"92869022744a40da2bd047f78cd910914a09b69c","_number":934735,"virtual_id_number":934735,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 13:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"all":[{"value":1,"date":"2025-02-20 16:15:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":2,"date":"2025-03-27 11:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-07 22:13:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":1,"date":"2025-02-05 15:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-03-27 11:25:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":-1,"date":"2025-03-27 11:25:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2025-02-07 22:13:09.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-02-05 15:37:21.000000000","_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-12 10:33:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-11-12 11:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 16:53:34.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2024-11-14 20:15:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-18 17:38:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-11-19 18:38:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-05 15:36:22.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2025-02-05 15:37:59.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-10 09:49:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"8d73f99e2a48f647124626d54a3cd8bc9a8cca8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-12 10:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0fbed13e17ff63fadc886deaaf47a74b40e59fd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-12 10:33:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54b100c2bd0865a08601908b1b4ddf76f96ce947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 11:59:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1b41499e7e34ffdae3bb2ec039e4603\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/21494a66b2f24338a2076edaed8da1e9 : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/52dd75259f6b42809c4053544b4bd4b2 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aa5b1508509642ab912892c3ad542550 : SUCCESS in 22m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/61c57781db4e40ab8dbebc00f9ea299d : SUCCESS in 28m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b74b0d729e07451eb6d6826f1b0f8309 : SUCCESS in 55m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/193939d24d5b487791273bc369a50e90 : SUCCESS in 1h 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a63d596e5d400a8a19db06a7b9a757 : FAILURE in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0de3bd3d565b46f48e66679fa83fe518 : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79cb3b1a2490446b897c79f198525206 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05f83cf80d894c3b9c4938734acf1343 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6485efb7c5064386ba432b44dec0f8b8 : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66383a67b7b643519c33be11def034a6 : SUCCESS in 3m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3a2c15a58a3a46ba8efcaaea28e309e8 : SUCCESS in 27m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb9faaa84fb240d786c4adc124b0a93a : SUCCESS in 34m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/43cee5b5af324306845ddf594326cd2e : SUCCESS in 1h 05m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/04cc1dab737942e39bebf92ba2aa9181 : SUCCESS in 1h 21m 49s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/68121d4f19cb4d07ada7c8355d6cf047 : SUCCESS in 32m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ee59bd64e8d4ff6b244f66609a6943c : SUCCESS in 22m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/454b8e922b144712b6c69536b4083821 : SUCCESS in 19m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ab64164cd90a47afb92545564ee53135 : SUCCESS in 1h 07m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/414406bc9db74c779eabfc7b31b07b12 : SUCCESS in 1h 14m 11s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/81706f74f59d40aeb6cecadbdcc4334c : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2d9f481670ed3dcb95d0ffed35d04f53276029a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-12 13:24:46.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2f449e99575d8249ff044bb2f05d4085e40b62c9","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-12 13:25:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78b0d01578170a8e6bb1fd385e400399395e499d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-12 13:26:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1f82f85b37f2566e21dd97b1ec805ddbe7a6b3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 15:08:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20871cee5f2644fea85f3dddc1c60b7c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/da560fd2208b4c898ba81e3b2d232a66 : SUCCESS in 23m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3fa4cd36dc054148b9b97f9bf96f8dd8 : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3c6c114ca9094457a1aa31fc78bfd35e : SUCCESS in 18m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1b74d8a7edb541e29477bd11f12dfac1 : SUCCESS in 31m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/68e087d7c86b4efa848bcb2215ab70bf : SUCCESS in 59m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fa6ca75b8988479696e216d3b4f783c0 : SUCCESS in 1h 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17d33e2249e44d194c93d6a6d0f58d4 : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce55fc06036f4d66b35b9e9a2e3a2bc5 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f99e74b72e9c4b689444594d8d41ec65 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed29ad2123e408ca7576a9967fb10df : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89c91641aa94441a64fc9cae30b8a97 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0541fc3845944578627eb2964af0db4 : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/517654f5e3644afd93a8235e1bf61b0d : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/82ae176e7c5e40a2b790e3a01121458e : SUCCESS in 40m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2495703f87ee442882176bb4f3069394 : SUCCESS in 1h 05m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5e27ba10991f4817963df412830f58c8 : SUCCESS in 1h 23m 10s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/75f8f26526e046468330bb91ab1b6c57 : SUCCESS in 29m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c954b575921644bcb5aa0889b6c506bf : SUCCESS in 24m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/26c3b23c718d4074b32806dcf0e66f77 : SUCCESS in 29m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d6016a8107ee48cc8aa3374681f45a1b : SUCCESS in 48m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/04d0a47dde7d49788a7142e2496ac316 : SUCCESS in 1h 18m 02s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/514478caf215428b9aee9d00c8b44544 : SUCCESS in 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"066c0f563197356fd34e480ef86fbe9750dc827e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-12 15:28:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"097d695b0cd40072ded640f7b42f33a3867dabd5","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-11-12 16:53:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"97c4caf891c1bdb127008a073628ac2cf7b1740a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 16:57:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f55abe8cbc042e29f95075edafb31d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2a2eb67431a94fa1a258513dc5a5c2d1 : SUCCESS in 32m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/29cf4db9d80d4ffd8afcde071b77ee1e : SUCCESS in 1h 05m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25bee516a973400e826a1a69faa29590 : SUCCESS in 17m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6968ceb8936e47a6bd7af5b845b8b5fb : SUCCESS in 29m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c4ed7b86d1c147ed99cce18588e2b1ef : SUCCESS in 1h 03m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/36d3672852eb4cdb9d29d98b3b41836d : SUCCESS in 1h 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45acd9be7ae48a793b580ca7b1adba6 : SUCCESS in 2m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d0f39329b74571908e5109b1dc61f5 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c5fab5e51bb452e8b3b49f6bd2eaa55 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d88fc08eae2d4220b2133cac14982544 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea87caba57aa4c44904b9b62c1c37e49 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb04a70b9dcb4f31a0aeb876d72e9cb0 : SUCCESS in 3m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9489e1ad64ee49bf992074b402ccf8e0 : SUCCESS in 34m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/002a0b93177c4df9a74188bcda675ca2 : SUCCESS in 35m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5027e1185e2442249a73cde6e5a05ad9 : SUCCESS in 1h 01m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3104b677b0f246f79d620bf891ac3fcb : SUCCESS in 1h 20m 51s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/27c2d0173f2146058853d14aa07dab08 : SUCCESS in 36m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/280fde878adb4d9ca61f0aa0440560b0 : SUCCESS in 21m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f2aa9cb995e7466690de548e064e9015 : SUCCESS in 27m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e5da9a185e9e4c07a6756620d2ec5285 : SUCCESS in 59m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e404a3f67023489e82cb163462a58cb0 : SUCCESS in 1h 22m 05s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/89979da2c7064baaa66c3c13183f2a9e : SUCCESS in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8330db7caef762247f5e60b6415c0ffe10171144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:30:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69cb944b8d14358b47bbc766902cb1b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/22f37ca0d9624103920e55a7d03dff07 : TIMED_OUT in 3h 02m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9656453f772c4607a58872efca136316 : FAILURE in 1h 33m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/640711c558e44fed911316a11ad287aa : SUCCESS in 1h 42m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f81804bcf204691a4373caef289b1ad : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/270b0ba7b24542b9a993145974d59295 : SUCCESS in 4m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a2530a6c7be84d429eadb93d8c965ad6 : POST_FAILURE in 2h 11m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2ca4cc125e2e4685aca9a432ff854878 : TIMED_OUT in 3h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f54fd1145e7a967718907d5bd01d45fe112d5359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-14 11:40:19.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f831af893babd1e145f1b1e1b19fb32e0a46b8f8","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-11-14 11:43:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"45b7f9d08be28fccaf43932d2c970d40fd371c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 13:16:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6298665783b34564ba5d30337f2216de\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/46948838b9e7410b8621b31d54fd201c : SUCCESS in 19m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3636baa8aa3f4667b2a532ac076511e8 : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad1f6fd923c2432ea5fb3cb68d4440b9 : SUCCESS in 19m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bb58a3dae46645eabc23548bd22fb880 : SUCCESS in 24m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87e295e10a464f449b3c688ee0e9b623 : SUCCESS in 1h 05m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d09510d84bce49ca97c0f125059888ed : SUCCESS in 1h 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ac4eefd50e4b849e0686342e636b69 : SUCCESS in 1m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/697b462f7c6146009dc918fd381fadd6 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c22c043efcae4fd7ab05be06ebac1004 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/491e55c14ee24e7bb0924db045e20e1b : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ad68183a804a2b8fc52f8c7d0a3eb6 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f8924ed1e4e45ffae992dc192db8b9c : SUCCESS in 2m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cfa3e9726b7346599c3416b7c9215e23 : SUCCESS in 21m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8ed02614f98849c09476b5fdafb26912 : SUCCESS in 36m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce49fabce06b49d69d6a12dbf545a547 : SUCCESS in 1h 00m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/854a08a0f3a34d4596a8979f6232409b : SUCCESS in 1h 28m 52s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/2cdaed03acb54b9fa16195fe2c0606cb : SUCCESS in 34m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/635c4a7c92d04a58ac42a9f54598e0da : SUCCESS in 22m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/59362a05f2cf4f3d94ef4be78d00e75e : SUCCESS in 31m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/80aa340a7d474896b4eb61edee9b9161 : SUCCESS in 1h 04m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c0387c20c3b9487186fa3a4dce11f5d4 : SUCCESS in 1h 20m 28s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/0c09f871823047b698d596242c3d3d46 : SUCCESS in 28m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7bc3416ea3394d973c2350c8dc5fc8b3d48b292d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:50:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/082a9e19a1b74e55a228908d029b6637\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b01208f6bcd145c5aedf51e74a05a871 : TIMED_OUT in 3h 02m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fd928ad2a4e54b9bb5fdd64e0a5083e2 : FAILURE in 58m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/95cfd5bcc941409fae4d869cb0e7204b : SUCCESS in 1h 42m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f4a03b0754249728330db09a5d71466 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e9facb6c2f6469ab617df3977dc535e : SUCCESS in 4m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a86532a5be0249559241c9d7afb3891e : TIMED_OUT in 3h 03m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e099e01254ad43b19895b612904d30f0 : FAILURE in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"527ffcd5896eeec05ba92030925e9b8706804349","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-14 20:15:16.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90fd85abfd6153d2b77ed8cfc9e5f257071862f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 10:53:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"33c57d8b39be58137164e6cd446d613a7b45f073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 10:54:24.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"88437f73cd4b6f75c977af8767af136a618476ea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-15 12:55:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ec76f34f4bd42380a39c6607d8d3c6fb8c5582f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 13:33:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d68efe4bec10453c8075699343c5e664\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/996216a543f14dd6bb9175d95209495f : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/01f82f8688c54e4aa678fe4016449630 : SUCCESS in 59m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c10d60a9130b40d4bb2cdd5f2ddad281 : SUCCESS in 19m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f23db474bbdf4504b7a594ff995321ce : SUCCESS in 27m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8c785edbd81748e48ebb5b6c8ef0aa5a : SUCCESS in 1h 06m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9072d85a24994cf691f5bc88d8289cba : SUCCESS in 1h 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2387f1df204640e58053abf4d207c5e7 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cab7e1c42e2d47afbad1af3dbbe76063 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/733f9f1bb5594dd395bcf5429c5b63a4 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9043416a614540a85b00e8c7ad9076 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36731e71d9f245e7b2b6475abb3f8f2c : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74bc70f3867d4aeea60512089abc048a : SUCCESS in 5m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/832ee5a2a2764ff0ae151dcc5cfed8e2 : SUCCESS in 25m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/55d03387342f4f8b9eb7c442b264dea5 : SUCCESS in 32m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa2bf13f50e048d48bc29a886e8e8ddd : SUCCESS in 1h 15m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8a1630879f704a7e9b1e453364aac9fb : SUCCESS in 1h 25m 18s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/235bdaa11a6f4442b57ab9d09e904de9 : FAILURE in 1h 16m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4e52a0c3735d410faf6bca22a8633779 : FAILURE in 1h 28m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6a2da83dd0d34e77ad15abe32477f99c : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/402ec1fa920f4214babec73bc2247f4f : SUCCESS in 1h 16m 01s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8368219041b1441b99555833bc6c25e3 : SUCCESS in 25m 39s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/69ea6891d0b24be5bab32190a9f4f6ff : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/75b085c076db4162ac2609dc068ea4b8 : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1132226399654f0b92a3b7afcae8a407 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/29a5176b130b42c788b034a8bb36151c : SUCCESS in 36m 44s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/a1bde95f28b449baa67edf1baffbb39a : SUCCESS in 32m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25f3b5564c7e4c9f921a134452fd0859 : SUCCESS in 23m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7e640800d1a44b439eefb019299e0730 : SUCCESS in 32m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c5dd7edc5b5242209ffbc74bb645d740 : SUCCESS in 1h 01m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3a3401d9fe5643af84b4d1969d8171a0 : SUCCESS in 1h 19m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/519031362b0242f0b4f80259a353b802 : SUCCESS in 1h 34m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4666899d0b54463bb55101d89dce54c0 : SUCCESS in 2h 24m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0a7a11e40831448a826142b50a0e7774 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d5e9d3584e3f4a7f8169fbe01aa817a0 : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4d12b473e0c549f3b09fde89feb8c7f7 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3a1092c050284714b093af0afa9ba7a0 : SUCCESS in 54m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d4e7234ed19f43818a338bd0df051ea9 : SUCCESS in 1h 11m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ee49760a4f8744908a0b235cba2a3b6c : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b0f3424b618d43859161a2b876813120 : SUCCESS in 31m 35s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/ccd7bc451ab7411dbabf02717143fc9d : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0749922e40c625dd119f85947ac978cf6cfbf510","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 13:46:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"16d74b67f6633da8d2d039b7e6f499ad3ca9224d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-15 13:59:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d70f42b6d624ebac2fa702d9225dfdf589f1972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 14:01:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c578262a73e437fa0e61b0836af5e9c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/08d1146aa596465f9d1554ca8293d7b4 : TIMED_OUT in 3h 01m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/926fe4a87fc84a7dbc9eb359ee4ea3d1 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/663a5e3e8cc44ca9a612575b3b643b67 : SUCCESS in 1h 12m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb18cdb824984aa494940b98ea93742b : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a69b249e9cd42e98cef3d85b634f99e : SUCCESS in 3m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/febdbd2897894d8f9884e423b0d33fe5 : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4d110b51b36241ceac4dc3826e8a5291 : TIMED_OUT in 3h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b93ea713722fc3ba81db52f5e0f0d0c8e0d24da8","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 14:15:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c64697d78eb076b5db1bfb163351fd7ac6e811e5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-15 14:22:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4a379bec16f35c6e5cb9f840a4665d0555092ff8","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 14:28:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6652f66ffa1fc9f14ab97cfcd1bb346f4a7104fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 14:36:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a913f04b4f670b764a176e0ec1d211a5f24c6f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-15 14:37:52.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"839648096d3f9c2029c346ab7b0ee3028d616ced","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-15 17:08:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5d22385c10b6aa3cb45cc4c08b75c89c246ea2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:18:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95d27ce9bb3458f93e9d01e864e1309\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e2779007cb3646c8998cb5a4498c2ad1 : SUCCESS in 26m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c7574588b4a9457d89d9741182f6206d : SUCCESS in 58m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2803c90f4e2e4f8dbfe7b96a5b9b711a : SUCCESS in 18m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2d2b46f23764b49a6f50bf3e9e3b967 : SUCCESS in 23m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/56cf0662a96e45f1813b7e83897a1eaa : SUCCESS in 1h 05m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2edb9921b9df41b18cb0abda23421d4b : SUCCESS in 1h 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4883ddd281a649a4a2e75034751f390b : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec27ad376d74476590e24d33d4d70095 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/753011d57df84385a5bcf47c2561cbf8 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33acc1217e9f4d1e9c6d0bd01bf668f5 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e84db31df2347148949e10ac6573753 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7477b676b50540618ba67c1fbd053fc3 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0a0ba1a108db4e418d887358abc25d13 : SUCCESS in 23m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/31d4726248ca4456b6f53b2a7cb4e65e : SUCCESS in 31m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8458299e80b340128eadb0f1138246da : SUCCESS in 1h 06m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9bb975ff98bd4f4db5767ef10248f149 : SUCCESS in 1h 23m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ab67ee038f4a4c9998e3bdf34c2e4694 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/38fa2a79dea44c3181e6448f840da863 : SUCCESS in 2h 33m 55s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/644a6a49c2344e47898cd1ae9123f458 : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e7c0d8d0f20742cd9134703b965433f2 : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3689816e540f46a1a1d6f3bbad5951ef : SUCCESS in 23m 59s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a7c0bb4c87ed470aa3c7be477607dbd7 : SUCCESS in 1h 00m 41s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bedf3c326170445f9dc412b916936aff : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4675606842ec45dd806d89b2d88cbfb0 : SUCCESS in 1h 15m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5484208c14374547beabcd6eeb46b35f : SUCCESS in 33m 05s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/e606b73a8153445fb53b0999e0c04e01 : SUCCESS in 35m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6f3f46a065a145108db321357f455e7b : SUCCESS in 28m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4e5b94078e6c44568f75d6796c4bd4c4 : SUCCESS in 30m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c956b0f1bef64446a67036d2a6cc242e : SUCCESS in 1h 06m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f9dca902ebeb4a608c1d3377f53b0e68 : SUCCESS in 1h 19m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0a2728ae9a704918bb26997aa1051254 : SUCCESS in 1h 40m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/63757842a0474954b0f3775d57667183 : SUCCESS in 2h 11m 20s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9aaf6ae68052496ba54b3ec8ec0d7b2d : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8362d026f7b64fb1aa03e3c2e3862f53 : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9253d0e057924ce091e8b754cd7f8e71 : SUCCESS in 25m 11s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e790f45a302840e6aea1bd95a366caa1 : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/db6ac4096aa34167ad94eb7e56fec4bf : SUCCESS in 1h 09m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/51c05970e68b469c816ca6a7d3031461 : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a4b9a93af18045a097301864060d080c : SUCCESS in 27m 26s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/4f090c688ae94210bc48a6749678fe67 : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0024ab2ab0eac49f5cedad3dd874acaf06eba9c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:48:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/090c97c2f5644c4dba133b2ca15c770e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/76a89e2bb29c4c95bd9a0437c2ac1750 : SUCCESS in 2h 26m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bcb1f674c8bb4fcc93e1882be2bd2643 : TIMED_OUT in 3h 02m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b6b3164faebf46f68c1f57a3e79ec911 : SUCCESS in 1h 48m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d43bbc8f5a7b4358a78d0d9aa4596452 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0478237c6d164bdcaa1ee46feb9d1758 : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95e6d265005540078fa7e41b6e86c9ea : SUCCESS in 1h 53m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/96f0b2e0a5c047a791ddfe2ac06df48b : TIMED_OUT in 3h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d9d6221183e57915f7e4da9063fc91b85150578f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-18 08:25:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5cec6d5587fa42e949f09cd1a0ad3e8794fff38f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-18 10:56:27.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b57bd36c5058d749c977d852937d4056e5b28ab3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-18 11:02:06.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cf08324cec33f874e2ffed0f992300adb9c8ad74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 13:07:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/563dbdbfa9a448889ce8e627da2b3dbc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6756330a99f54a6e93dc1b59b8609435 : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f2d6ae8a866a4a39b19a3cbec9197314 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8bb3f427c104e16b52eee70d96123e2 : SUCCESS in 23m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/67a2899c33844f1c8f4adceb82d89980 : SUCCESS in 32m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/00aba2daace74652ba14b7fb5a18da3d : FAILURE in 7m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f60eb78317464757a5f62e3333e40493 : SUCCESS in 1h 44m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da2ef22b3ca4057abab57b6d678d8e0 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1efba85b674aacbf1b9ac03bf2ff6b : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95e45656853b4afba5bcfeaf8da3735d : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e336806525d421fafc2106de71e348b : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3682c255be97444f9276a4a2708740fe : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be19090140f49cfbe1151d050e71116 : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/022b7d8680cd43b3a238aefab1b82d0c : SUCCESS in 22m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4af24bbd50354dbdb706b4b1417111c0 : SUCCESS in 31m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/662d010925d24ad997ebd7ef1a7f1821 : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6304f3a56a384d67a398860873a23b96 : SUCCESS in 1h 24m 43s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/31644bead3704c809cca6e7f9146d8e2 : FAILURE in 11m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/efe13a9c57f5421d8b3f79b5f84ba81d : FAILURE in 1h 46m 03s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f999f44132da4855bb7aa1449364a3cb : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/bb3e9cc5f62046b2b35e54c9798a69c9 : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8e0f788f949843fe899489c0bcc03466 : SUCCESS in 24m 41s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/763dbeab61e54793a09ab2bb3189d5c2 : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/452c00ba29194899a923289e1f2e8277 : SUCCESS in 1h 17m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cc9adec3670d455b8e01730d402de6de : SUCCESS in 1h 18m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c6016a8230df4125ad85326cc6f0888b : SUCCESS in 30m 24s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/724e2a8dd7a84df9b3104fd18c138b15 : SUCCESS in 37m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9fe616176c7e4ee0abbd015fc53b817c : SUCCESS in 22m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/398da8e2a22a4691a5459d4552e3b5a2 : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/84ab51bd1d494788b6b1229a50574763 : SUCCESS in 58m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/067a5e60237e4645bbda2c664115ec98 : FAILURE in 1h 30m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d342b151b0e347d0984f581801152f54 : SUCCESS in 1h 41m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/faa0a3aa144245149a99e2567f7f5a26 : FAILURE in 2h 02m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/15be54da50674e81a10f4beeb8c65964 : SUCCESS in 1h 21m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2b8c05b59c084a96a4db51738747bba0 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f546c5929a09429e88434dba309665ae : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/771f204a776c40ddac72ff12ad3f10c8 : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/510a17f514f1491b80d24a961ce81571 : FAILURE in 11m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/02d7bd2c3ecf47878b3a7d1cf412199b : FAILURE in 10m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/869ad35ca3d741b4bf1951e320f751cb : SUCCESS in 27m 46s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/769eedfa114b4e7d9372e86adbdd883b : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"69439d69ca690124f7cf02a1531d5f70375289f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 14:21:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8c0fed74bd44a0a506d971bbab670c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6254a91f4fdc4179b983f77e8bdc40fa : TIMED_OUT in 3h 02m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1631dc20303342d2a8f9d98e05bef242 : FAILURE in 1h 53m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb3e141642354305baad1826197971e1 : SUCCESS in 1h 23m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86246ef986114c10ad73118cf4c1437b : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/941e0abe885441c880468bcc27919f5b : SUCCESS in 4m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/982b0f5a1b98486195974df213efdc8e : TIMED_OUT in 3h 02m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5019dbdbcf7840908b0da1f56ef88a8e : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"297f33fc905285667c9b0d1ed420aea404b1b0b9","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-11-19 13:41:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3b656d025ca34e18445bad9ccebd817427b1339e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:07:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241e950c7a3b4fefb0d36e69d60e0caf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/89400136ffa94c74b287cdfa5980146d : SUCCESS in 2h 09m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/efc4edc25d824d8692ddb6244a5913d2 : FAILURE in 9m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1bf2cb83f80430aa16eadc8854909a7 : TIMED_OUT in 3h 01m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a93e83a1d58546dfb55b06d7ea120f88 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/231f82014e2c455c848aefab77812a18 : SUCCESS in 6m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c535af8491114690968df342da0a63be : SUCCESS in 1h 48m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d9b8d2dc97084167a0d8e1f7068641e8 : TIMED_OUT in 3h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7270e8c39d6ac33fa0a5d49158b16e31005a83f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:26:38.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc40c0e83618495fb6c8bb0968da1633\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/58ca2583e70447f3916f8a6cd082178a : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/25544c56ae8045d9b9894247b2e0a9dc : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ac588c191023423c91cbde1e51788fed : SUCCESS in 17m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3cb74314d0c642cf854ff40710a59dd6 : SUCCESS in 33m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/80dc84c5c6f74b19b5bc12d1cc60515a : SUCCESS in 1h 00m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0dc0fe1d6cbd43ef99ca2297240282d8 : FAILURE in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da28744f95164a36884ea64e4a44541e : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce914adb22024be1a290252fb8cba0cc : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1623e72c0efc4b7e85a972f405589e6e : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad8f0d1e747849e78e1912f886c9f7d5 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ae200ec89d4d56b2e610995860f300 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec7f91f220e24d889b46d9ebca65c77d : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e5e111477e048449472660e435788f4 : SUCCESS in 31m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e2556473332647ec89198cf12aa78ffb : SUCCESS in 34m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1e9945de16f84015b96c8db13863cdd4 : SUCCESS in 1h 07m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ae88d07c185a4915b97e1b15e67ad79b : SUCCESS in 1h 27m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/04ead1eda9154e4e80d74e4809dfed43 : SUCCESS in 1h 51m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f760b73aada347869a5d9247b60d7f03 : FAILURE in 2h 48m 28s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b688df38dd3c42dcbd22618e62df8678 : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e4b9551311884953b2d0a9986ad2db40 : SUCCESS in 1h 22m 06s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/27be6a10a9904679b4a939f94bbcf5dd : SUCCESS in 29m 51s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/50b6c2b92b524343bda1211baa3a4280 : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/43dcdf0f06d946eb97444fe595d085c6 : FAILURE in 12m 22s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cb5cf264393f4229a1e80b9d2ecdf154 : SUCCESS in 1h 20m 55s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c8df2475e87f44988c6bc6b79d0a159b : SUCCESS in 39m 49s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/44a52a1b1a9d454c8e780a472213f3e8 : SUCCESS in 35m 37s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1a64ce2bf5744057bae6cc3757225879 : SUCCESS in 20m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/69005f9a56c74dcd8f506f644af2aea6 : SUCCESS in 26m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3d741593f08f4f17a2c2e49652457720 : SUCCESS in 57m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e497a79aca5a4a7599c2bafb067024d5 : SUCCESS in 1h 13m 22s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/75f801ab2859451b9fed803080d9f09b : SUCCESS in 1h 33m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aa3bef93253e46d5b5dbe4c4a1ce6395 : FAILURE in 10m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/32a4b3f6b09649f4955b4a83e11d01cd : SUCCESS in 1h 24m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/58364e1d6c16479ebaa7292bd3f13ed8 : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cc19f541ec46437bb9d5de15c2d931c8 : SUCCESS in 23m 16s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/cb2293c934f643c5b91dea9cc44d8476 : SUCCESS in 1h 04m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f6e9c559c8414861b6f06d35b0ee4344 : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/430bf3e588514c04a90f9bc851f3a3c5 : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d0a6629aec7b4d56ad4cde808f626a2e : FAILURE in 10m 19s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/a3c3d66281be4a45bd0061f0f817ae83 : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"89f93e0db0792270daa378de4305cbe9393a2f97","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-19 18:38:43.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b4d7b225df7a99e508cbe29534b1c0d2642a8ad9","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-12-05 11:23:09.000000000","message":"Patch Set 10: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"67f7b284fb89e8884882d1722550a1f380fd5936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-12-17 13:50:31.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"79219d9cf65810f434af0499422d9a58e3208d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:32:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/037a147a4d3a4abaaf29d805f63cb346\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5c843f049fc54528b1c89dcd6cfad631 : SUCCESS in 25m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/942e00c0523f4e02a19314462e3e4b4c : SUCCESS in 1h 14m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f2f28e4ec96b4cd7846b07d7f7a9b5ba : SUCCESS in 18m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/99524fc6d9c74a5587bf82ae77a70de9 : SUCCESS in 26m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0329a857cccd40fcb87b79542e7cea4a : SUCCESS in 1h 03m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/203cd46b79b94f06a9c3fcdc775cbfd3 : SUCCESS in 51m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a223b7bb52354bf9a6ce8a8cf3f8e2ee : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1664f6ccaa9d4df68f558c0540c2f85b : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5fbf01b66d44146af24763ad04152e4 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10b4ddc76bbe450f973e505235365df2 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66865b65f6447f09d34535899c3b421 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30a6ced9aedb4ddcbdefc568ce553140 : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1df5be24c5584d7d930cef3df788730b : SUCCESS in 30m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/491ea53e3b294357a58a97928ca7eebb : FAILURE in 35m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/77683de0c0794557adce51988309ff9a : SUCCESS in 1h 12m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/01a8846f667c4e57957af67e74da57ab : SUCCESS in 58m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7a2eb06a590148efa8485bf248fb17c7 : SUCCESS in 1h 51m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1924cfec0ca6468aa3f08369b2fa6f1a : FAILURE in 2h 10m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c814490b1e924c509bf643d178a8e562 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8f5777a8b7a64d958b8958a96c7e6947 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3158645f84c447a59ec30beacabda7e2 : FAILURE in 28m 22s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a3a3bdf5bbe24136be75d25068e0bbc9 : SUCCESS in 1h 02m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d5fe360b0cd74fbc82b51dda67122dcd : SUCCESS in 1h 11m 44s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d3c71f642c0f4dbb88f8711575a33eee : SUCCESS in 1h 22m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7cfeac06ac6249bf916dc96e578e9475 : SUCCESS in 33m 31s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/233dc7703f4143fdb50aa131b0ff3e8d : SUCCESS in 40m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ca55f447779c4204a14b3624603b99a3 : SUCCESS in 23m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d0060489b51b4a7ebb71d2e80ac3ed27 : SUCCESS in 1h 59m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a9b85b56bee94bfeb202df237ed4ea83 : SUCCESS in 1h 08m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/43960159ff474d16b2622c79dda2529c : FAILURE in 54m 49s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/113b74cd8d9c456780e55bbd1ee4047d : SUCCESS in 1h 40m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f864761efa0149ab8444474fd676bfe2 : FAILURE in 1h 07m 04s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a5e6f839ac4c4f909987e37ae3eaaa8d : FAILURE in 26m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8b519f0078d040708263e9926a342208 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d2ec4253ea324480b95083cf528c5512 : SUCCESS in 17m 19s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/eab67729ff744f58973d168a42a12fd2 : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1e94e5ba2cd142c0a59c59159df772fe : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cbd3735680274795a6a0f68fef77bb12 : SUCCESS in 1h 14m 25s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3f3b1218684f491d998d09c96b7e4190 : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/22881771d6b842d085c584c362bfda27 : FAILURE in 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6af9b5fbe38f05590e0bd236908eb93a31a2e8a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:10:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3fa13b4e78493da5b2955fff317fd9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d92be7016d614a3283947bd6113321ef : FAILURE in 2h 39m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3db420b94def480692f26b22cbd5ad65 : FAILURE in 1h 30m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/327f46a5616f4629b97bc4eedb3feb4f : SUCCESS in 2h 50m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f3962f4604a4bef9cf558b67eb3b69c : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2fa04ac1c2e40109071025cd96abf9b : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/30cc68bcb5ec4f3aa76d6d2832382511 : FAILURE in 2h 42m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7033bbd2daec44d59cf06c06f168e15e : FAILURE in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1ec6428a0a2a063941d138f1c28483cd5133be00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:11:37.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3fa13b4e78493da5b2955fff317fd9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d92be7016d614a3283947bd6113321ef : FAILURE in 2h 39m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3db420b94def480692f26b22cbd5ad65 : FAILURE in 1h 30m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/327f46a5616f4629b97bc4eedb3feb4f : SUCCESS in 2h 50m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f3962f4604a4bef9cf558b67eb3b69c : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2fa04ac1c2e40109071025cd96abf9b : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/30cc68bcb5ec4f3aa76d6d2832382511 : FAILURE in 2h 42m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7033bbd2daec44d59cf06c06f168e15e : FAILURE in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5cb43eec8e7987f16d0697e0214b6e7865bed9a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:28:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3fa13b4e78493da5b2955fff317fd9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d92be7016d614a3283947bd6113321ef : FAILURE in 2h 39m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3db420b94def480692f26b22cbd5ad65 : FAILURE in 1h 30m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/327f46a5616f4629b97bc4eedb3feb4f : SUCCESS in 2h 50m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f3962f4604a4bef9cf558b67eb3b69c : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2fa04ac1c2e40109071025cd96abf9b : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/30cc68bcb5ec4f3aa76d6d2832382511 : FAILURE in 2h 42m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7033bbd2daec44d59cf06c06f168e15e : FAILURE in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7474735530bc97457fc82b518ebc0e31f92dd29d","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-12-19 11:46:07.000000000","message":"Patch Set 11: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fd2338b0017ef06d7c31659ab13ec17934c7a86c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-19 14:50:08.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"113d4bddc92dfc6923f43fda935e6d658bd512ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 14:53:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d74a24eb0e5a487aae433cb60f857223\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/61d3ff8778e94ab28b818e0940fe554c : TIMED_OUT in 3h 01m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/511949f55b444a8eaf92294656e26d43 : FAILURE in 1h 58m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a364cb4e3214a63940764b2152911f7 : SUCCESS in 1h 47m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26326ef9a886493a958f48ccd8ffa5ab : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/762369731505434d933e058429c9fe00 : SUCCESS in 3m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/21fdebacf68a4efca6189fc089bf7cc3 : FAILURE in 3h 01m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b3e6a31ad51648cb82a226d3e881d34b : TIMED_OUT in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d3d53e4fc7d18df6dfd17a8f63f59dd9f18ca856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 15:04:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bbec48e68d44508a15e6be93dbf6c3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/09c6b372c9c74b9698e2526ace7a8127 : SUCCESS in 22m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0b7c579394444ffdbd514ca8046d830d : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07ba7f4eabb14fdf8c44fae21717e28b : SUCCESS in 24m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4e08437899b4988861c392162195261 : SUCCESS in 19m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bbe9ab1e406345a7895ec5401239ac16 : SUCCESS in 1h 02m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b963a546ba248efaee63e2087934f22 : SUCCESS in 1h 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b400f433894be19c926aa010dfd5af : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a2afbe96b3e4b2c89382754e574b0b7 : SUCCESS in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d5b52bcd3264de39fccbebb41e41d2e : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfb07666abe243669e7b3b1e56ee2ac5 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f61468ada714e06af56213cdb0bc7d4 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a3471e5306e4b309887345db2bd2ad6 : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3869f035661d4d17b548671713bc7374 : SUCCESS in 28m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1cfaa4dc053495093e294b6a680c090 : SUCCESS in 30m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/63591619dfc84f438fd65fa383316918 : SUCCESS in 1h 10m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8f8bac6474744e4381af692509774827 : SUCCESS in 1h 22m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/13e15f25f6404ff289443ff93d8d9443 : SUCCESS in 1h 52m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1259e7c5c2f346e487ed09004bb81615 : FAILURE in 2h 36m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f436fb585ce94118b5b5cbc39d0a5178 : SUCCESS in 1h 17m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/965f93227ed2470b912193ea50ee4eb6 : SUCCESS in 1h 24m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/af9b2fc048064f37bfd2630414c4c26a : SUCCESS in 18m 48s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b610598fc7814b0a86f24b267fe8579c : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0d560cc6d82b4720b2bd90b6d083282e : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8e54b9ad07f94c118590fdd085301d56 : SUCCESS in 1h 24m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/54fa1621d23e4d6ab5fdbe2b91e2e11f : SUCCESS in 31m 41s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/b7e8a7019182445d8b53f1d233e623c3 : SUCCESS in 36m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7803cd541c504a929fcbf9bd57d21c90 : SUCCESS in 14m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/173eadd1be75415b9e7add0ea3f7a62a : SUCCESS in 27m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/23419fb603ec4d1a9e813fe282371400 : SUCCESS in 1h 07m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0928a80a35b2450b9f08be29368601f1 : SUCCESS in 1h 22m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b4a5d0f491214ad4ab221812db07ac26 : SUCCESS in 1h 36m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f730566374a247cbbfb55b91d7cb0bed : SUCCESS in 1h 38m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/49453af74fb1428788bcead45296eaaf : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a886ea1872f3483c99fa6413ce9853ab : SUCCESS in 1h 35m 38s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a80ad331453b48819df807440f35bccc : SUCCESS in 24m 20s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/91b962bbcc7c4f4788dafd8567798f0f : SUCCESS in 1h 01m 09s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/78053e4ef513496fab15b898c261696d : FAILURE in 47m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d60ebaac5aa84eca8befb98f9289a79b : SUCCESS in 1h 18m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6214a80565ce46f3a43e91bc06d7a68e : SUCCESS in 16m 39s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/0f2c9b44495d4946ae95c7d9ebd242a6 : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d2e48a72cef7ff962576b3d0364a4ca8a8320416","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-01-16 13:31:44.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"b2eeaa9aac053bb3e38630a13d6f0b5bdc64a841","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-05 15:36:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6b3658249e8476f003317b941903243cfd0cf5e9","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-02-05 15:37:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2b1cddd5f81dd1b05216b3e2215b1a585de3df42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-02-05 15:37:21.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"b19df629d1b58bff8494f52012aa174fb9b779a4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-05 15:37:59.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"601078b99349300475d01ded722ffd5b865dfad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:23:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97aaee88834042679035364749620f0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2c8da38865374749b5b5da2a530497e1 : SUCCESS in 18m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c6143bf58d1e49319c288e2c91254353 : SUCCESS in 59m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9c40ec0bcb014246825a966bb1f3bc50 : SUCCESS in 15m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ed1aa13268f548b3bf5df0b121e406fc : SUCCESS in 29m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/379d8ca0c9e94c6480b195ce9583a4e0 : SUCCESS in 1h 00m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6f48f513f604882912f17bc8a83cdaf : SUCCESS in 1h 43m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e618fa03ea4ee6a439952aaf0fc935 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5e0bcbfe7154189966526e5ef00d395 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/793b6dfbdba24472b997313933e5cf71 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bd0154e61b9453e8853166324159b93 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a3c045b0c744e39b086f08ab2d8e33 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2216d801989e4461b39ce095c008e561 : SUCCESS in 3m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf27bd3d31de49d99c713b669959beca : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4593a8c2789e4753b28614695942f999 : SUCCESS in 31m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/873ad9cb73e445f3a909648e0ec216a8 : SUCCESS in 32m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3420722ce8e148ab98ac343807ad5894 : SUCCESS in 36m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c460a01c59ef43099ca64b8f3f31c1ef : SUCCESS in 1h 29m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/04dbe6230ba340a2b9f8df4334e9d088 : SUCCESS in 1h 52m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d44eaec10b644ba9bf7a4eab501f5eac : FAILURE in 1h 23m 23s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/db389e99b6d7464c8bc8e219b2ec88ed : SUCCESS in 1h 21m 46s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/434fd0ddd5954b7d9648eafc096c1d79 : SUCCESS in 17m 18s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8f3013d800614c87915730f161b9c548 : SUCCESS in 51m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/72185ff8057842f7a3df92d2d86eac87 : SUCCESS in 1h 16m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/59fef6972b8443898950012fafa0651a : SUCCESS in 1h 39m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b4032830ae2e45ea9d95529373141027 : SUCCESS in 32m 23s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/e42b5a6109c94667aa68fbdd459de1be : SUCCESS in 36m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fb47c6b33b44eabb2998b5bdf8c7c7e : SUCCESS in 22m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/143ecc727a134a718720511578872321 : SUCCESS in 29m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9877fe1b0eab4354bcb1168198bee066 : SUCCESS in 1h 12m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1636c18b4a37406982ca8932e142bd1f : SUCCESS in 1h 35m 29s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9fd15b6a40874f92a79ffd771fa7b43c : SUCCESS in 1h 54m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d36ea5a1782049e4a0cbb4a10f2a735c : SUCCESS in 2h 39m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a9ed97df271245eebc5d4e782eab0932 : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/15957ea4e94148c9a8155b9367485e02 : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/68548bd7b7264f6cb9d557d390b724bc : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1226e5db6f7441b4acdbb207078e0241 : FAILURE in 1h 27m 49s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/be335133124b469bb334442646451d1c : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/63c8b4c6fac14ee98e9ebccb005ad761 : SUCCESS in 31m 23s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/67d3c782c15b4ab69f92298697459eea : SUCCESS in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"badc39933f147f7c5cea519d7dd3f1faf66d08ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:45:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85b60154e28412f91657e4f9148a70d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/445a12f7e8c944a694855d01fa34725a : TIMED_OUT in 3h 01m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d79203e40882454ba44cc5a15d4798d4 : FAILURE in 1h 43m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/efbf12a2ab214820b81250f4b827ed4d : SUCCESS in 2h 28m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d52a1d8b01eb44148661384a3b6da517 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c313d570ad9a4f7f94bff768e0fd9632 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/613adb56afb049119afc519ebbd15b06 : TIMED_OUT in 3h 01m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6af5d303583e4d2290bc410366736983 : TIMED_OUT in 3h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1893b413d433ea8688cb83ad7488e8c5118ea6cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-07 22:13:09.000000000","message":"Patch Set 12: Backport-Candidate-1 Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"ff117f36f4d991e9e4cc42e90d7db2232742cbd5","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-02-20 16:15:19.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"abfa8f73527c05e8975435f5a9177c1abb27dd05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 11:25:01.000000000","message":"Patch Set 12: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"7d129d61e5c55995e6b58b23063473a33eeedc7c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 11:25:21.000000000","message":"Patch Set 12: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"aed9ed035d4e2bf610a7687206e08564800541f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:25:44.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"a19319bc9415e3bc21936341588114b5cc6e5fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:12:48.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8837a239dab141eca1c46bc9455636a4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/84c977433fd7417e891f75eda784bc4c : SUCCESS in 9m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92e61b391b984c6c8caa5bda76ea57d7 : SUCCESS in 24m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7742e073c6ca470c922a4d61e24d89ad : SUCCESS in 59m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a32ac21895e745adb4d6c87f1f27b8ba : SUCCESS in 1h 42m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2502d5202d154d12ae883315123ead6b : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b68c01718dc649f5a5be349846d9ed8b : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a70b0b480f44cbfb1018f12650b9967 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e363fb7af99a4b2fa896ad4edf421472 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a103cc4875c348159cebefb6da6f3dbe : SUCCESS in 4m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/87ea7131796a46248db7125c2ff0041c : SUCCESS in 2m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1b02e3dc4be43b3ab245c620b832e69 : SUCCESS in 24m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7fd671ebac974141b501e8deed42c904 : SUCCESS in 1h 02m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d4aa2f4b16504c07ab178dc1a3ebd3a8 : SUCCESS in 1h 09m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d5ecf76d3d9248e5b6887bd4117d2d43 : SUCCESS in 1h 05m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/965191b200d1476a835470fb5efceb29 : SUCCESS in 23m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ce86742a78304a37989aeb05b28d83f2 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/293c3328db774faea9a257305f23499d : SUCCESS in 1h 08m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1521615345554e309f9c85f662b7ea96 : SUCCESS in 1h 39m 06s","accounts_in_message":[],"_revision_number":12},{"id":"b7b74dc00ac2dc3e196f0c1808b5e16f3a3e2083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:12:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"92869022744a40da2bd047f78cd910914a09b69c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:13:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b210c9e99694d18a00c2f87a6f47979\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0665966ad8b9426ebcd80b2aa25eb1d8 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c2e3a1c97194bf286e7c554198f6650 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"4cab35b1a81055c3696e5fa42d794d6fc96076e6","revisions":{"08da2f89f0ced63e20667fec43815968f174c8cb":{"kind":"REWORK","_number":1,"created":"2024-11-12 10:30:17.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/1"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overrided\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml\n\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08da2f89f0ced63e20667fec43815968f174c8cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08da2f89f0ced63e20667fec43815968f174c8cb"}]},"branch":"refs/heads/master"},"73d98c7c80993a1087ef3ec5b151d746763e4da9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-12 13:24:46.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/2"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 13:23:46.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overrided\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73d98c7c80993a1087ef3ec5b151d746763e4da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73d98c7c80993a1087ef3ec5b151d746763e4da9"}]},"branch":"refs/heads/master"},"247426df26828d4de0c921c73206c6dd203e214b":{"kind":"REWORK","_number":3,"created":"2024-11-12 13:26:58.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/3"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 13:26:50.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overrided\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/247426df26828d4de0c921c73206c6dd203e214b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/247426df26828d4de0c921c73206c6dd203e214b"}]},"branch":"refs/heads/master"},"ffbadb1ebaf81c1238f464edc40e64ca162949da":{"kind":"REWORK","_number":4,"created":"2024-11-12 15:28:02.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/4"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 15:27:55.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overrided\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffbadb1ebaf81c1238f464edc40e64ca162949da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffbadb1ebaf81c1238f464edc40e64ca162949da"}]},"branch":"refs/heads/master"},"bf590a8a0b68b2b2e846d5c10ae6c684a8275e58":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-14 11:40:19.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/5"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-14 11:39:28.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf590a8a0b68b2b2e846d5c10ae6c684a8275e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf590a8a0b68b2b2e846d5c10ae6c684a8275e58"}]},"branch":"refs/heads/master"},"96461258a6fa8bf930cc924958710447cd7be48e":{"kind":"REWORK","_number":6,"created":"2024-11-15 10:53:28.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/6"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-15 10:53:16.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml.\n\nWith this change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nMoved UPPER_CONSTRATINS_FILE variable example to the header block of\nkolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nRemoved example UPPER_CONSTRAINTS_FILE in Dockerfile of base image.\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96461258a6fa8bf930cc924958710447cd7be48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96461258a6fa8bf930cc924958710447cd7be48e"}]},"branch":"refs/heads/master"},"9d0705b8d645643a545f6c35e526004e17a213b7":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-11-15 10:54:24.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/7"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-15 10:54:13.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml.\n\nWith this change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nMoved UPPER_CONSTRATINS_FILE variable example to the header block of\nkolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d0705b8d645643a545f6c35e526004e17a213b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d0705b8d645643a545f6c35e526004e17a213b7"}]},"branch":"refs/heads/master"},"44a2ca969701ac22c8230cab1b7006dc8b7d8a93":{"kind":"REWORK","_number":8,"created":"2024-11-15 14:36:28.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/8"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-15 14:36:09.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml.\n\nWith this change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nMoved UPPER_CONSTRATINS_FILE variable example to the header block of\nkolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44a2ca969701ac22c8230cab1b7006dc8b7d8a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44a2ca969701ac22c8230cab1b7006dc8b7d8a93"}]},"branch":"refs/heads/master"},"96f5ec23d4a5f06f695ac9b6fb148dffb50b95b1":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-11-15 14:37:52.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/9"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-15 14:37:03.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml.\n\nWith this change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nMoved UPPER_CONSTRATINS_FILE variable example to the\n``kolla_toolbox_pip_conf`` block of kolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96f5ec23d4a5f06f695ac9b6fb148dffb50b95b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96f5ec23d4a5f06f695ac9b6fb148dffb50b95b1"}]},"branch":"refs/heads/master"},"b0381c69991c92ac4a1d147fd504a530f8d08dac":{"kind":"REWORK","_number":10,"created":"2024-11-18 10:56:27.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/10"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-18 10:56:18.000000000","tz":0},"subject":"Fix Bifrost install playbook ignoring openstack-base upper-constraints","message":"Fix Bifrost install playbook ignoring openstack-base upper-constraints\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis fixes the issue by passing the location of upper-constraints\nfile to the variable ``upper_constraints_file`` of install.yml.\n\nWith this change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nMoved UPPER_CONSTRATINS_FILE variable example to the\n``kolla_toolbox_pip_conf`` block of kolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nCloses-Bug: #2087960\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0381c69991c92ac4a1d147fd504a530f8d08dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0381c69991c92ac4a1d147fd504a530f8d08dac"}]},"branch":"refs/heads/master"},"e7a9653a213ae3eeca476b94e7784e270a8cf279":{"kind":"REWORK","_number":11,"created":"2024-12-17 13:50:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/11"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-12-17 13:50:30.000000000","tz":0},"subject":"Change Bifrost Python upper-constraints source","message":"Change Bifrost Python upper-constraints source\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis change sets the source of Bifrost upper-constraints to use\nupper-constraints of openstack-base for consistency among all Kolla\nimages\n\nWith this, change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nTherefore, moved UPPER_CONSTRATINS_FILE variable example to the\n``kolla_toolbox_pip_conf`` block of kolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7a9653a213ae3eeca476b94e7784e270a8cf279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7a9653a213ae3eeca476b94e7784e270a8cf279"}]},"branch":"refs/heads/master"},"4cab35b1a81055c3696e5fa42d794d6fc96076e6":{"kind":"REWORK","_number":12,"created":"2025-02-05 15:37:21.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/35/934735/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/934735/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/934735/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/934735/12"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-11-12 10:19:54.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-02-05 15:36:52.000000000","tz":0},"subject":"Change Bifrost Python upper-constraints source","message":"Change Bifrost Python upper-constraints source\n\nUsers can override the upper-constraints.txt by setting the source of\nopenstack-base in kolla-build.conf.\n\nHowever, bifrost-base Dockerfile so far doesn\u0027t pass overridden\nupper-constraints file to Bifrost install playbook resulting the\nplaybook not acknowledging it and just use upstream one.\n\nThis change sets the source of Bifrost upper-constraints to use\nupper-constraints of openstack-base for consistency among all Kolla\nimages\n\nWith this, change the environment variable UPPER_CONSTRAINTS_FILE is\nonly affecting kolla-toolbox image.\nTherefore, moved UPPER_CONSTRATINS_FILE variable example to the\n``kolla_toolbox_pip_conf`` block of kolla-toolbox Dockerfile.\n\nAdditionally, updated \u0027Python packages build options\u0027 section of\ndocumentation to represent latest method of changing Python package\nrelated options (including this change).\n\nChange-Id: I0a9c782ca1177ef84e4323cb1e5bfac0fb3f383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cab35b1a81055c3696e5fa42d794d6fc96076e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cab35b1a81055c3696e5fa42d794d6fc96076e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
