)]}'
{"id":"openstack%2Fkolla~934952","triplet_id":"openstack%2Fkolla~master~Ifa92999672ed267ce877c492c9140fbee271b8b3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-31 23:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"37430":{"account":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"last_update":"2025-04-01 01:31:48.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-01 01:31:48.000000000","reason":"Change was submitted"},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2025-04-01 01:31:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa92999672ed267ce877c492c9140fbee271b8b3","subject":"Added libvirt dependency to nova-libvirt debian based containers","status":"MERGED","created":"2024-11-13 11:35:51.000000000","updated":"2025-04-01 01:32:49.000000000","submitted":"2025-04-01 01:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934952","meta_rev_id":"2f9f9892cd9d72cc30dddcf3c8ec580c5a736f52","_number":934952,"virtual_id_number":934952,"owner":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 01:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"all":[{"value":1,"date":"2025-03-13 14:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":2,"date":"2025-03-31 23:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-13 14:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":1,"date":"2025-03-31 23:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":1,"date":"2025-03-31 23:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 11:42:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-11-13 13:03:01.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"state":"REVIEWER"},{"updated":"2024-11-13 13:59:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-13 14:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-21 14:36:59.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"}],"messages":[{"id":"05d63ed432e3dc50c72b58f1788cf16aa8068f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"date":"2024-11-13 11:35:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5d81add1c9f36d3b917db4295bea67acf31de01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-13 11:42:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f7940652b2f9d07d1b12ec795902b9ff2b18496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"date":"2024-11-13 12:51:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5ca220b94b5816a673e183966f00c24def0d9e6d","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-11-13 13:03:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"becf78f8c20edd09415e423522a052f2c25c36ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-13 13:59:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e58953c05ee01295dc42d6872e0510abeb224068","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-13 14:00:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68e1e0cffad039f5deb8f5b4672c809527699487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 14:21:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4b6a1f8e11544b7aba4711af69272eb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a55785a4e5ce474bb6072b0a14d29c15 : SUCCESS in 22m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4c1edcff35094c36bbb6a298aa5ea97e : FAILURE in 7m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0df369546db24239995be3d69e365eb8 : SUCCESS in 21m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9d25e8c36ac44e4b909bbd2b89e7fe81 : FAILURE in 12m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5e401d91896f47718c447a9c7d2ea298 : FAILURE in 8m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/03634533682b4d4083806d188dab250d : SUCCESS in 1h 24m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecef29eee5f04fa08f96c127e6d8a16a : SUCCESS in 2m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88f50cda4a2c405facb159218be6b59e : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d70458feb7645019e4adfeb255f1878 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ace4fc3d9fd3473480f4a43c52d3f88d : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64cd4c2ca834dd1886fb55cff673c04 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4fad4a511b84670b2d3fde6a4f60f03 : SUCCESS in 24m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/31b97a4eddff4ca28d77b73b716f19ba : SUCCESS in 31m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/534a1d08a7ce40f4b14c812ec16333bf : SUCCESS in 1h 11m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7912ea965f994d01bcf5211949188ae7 : FAILURE in 8m 28s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/7346948727074bcdad28eba2ba922fc1 : FAILURE in 9m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/defe0f39b252495290b391a4b8917212 : SUCCESS in 27m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5423ba4a1b2b4e13b2b09bf5717351ba : SUCCESS in 35m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/49e85c03105a4f3fa4341dc2bae1012f : SUCCESS in 1h 12m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0348d4b1aa68425bb8f80422fdd96819 : SUCCESS in 1h 23m 33s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/e54c1eb56546442eb33cdd2e2facfecd : SUCCESS in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64c260b1e5a74c2c68dccefb9fd090c0ee3ba389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:07:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11cff2e811484a6a8cb874a755073576\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b886cd3e8b7b4d8ca6399990a26330d6 : TIMED_OUT in 3h 01m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1dcda5fb510540a2844c58ab2802f086 : FAILURE in 3m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dfce269a3cf8400baacd0f40e3bb2fac : SUCCESS in 1h 55m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6afe1f129c4479ab91a897c00e7d3a1 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d94c4e81766b4e11b7693d247d724c5c : SUCCESS in 3m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f3c188888b8c4150a2c9ba6165f210a7 : SUCCESS in 2h 11m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8fb1917c78c04400bcce7fee62c336f5 : TIMED_OUT in 3h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23729b18dd263bf0e587c80e327ff5dcb4e1b392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"date":"2024-11-14 10:38:17.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"486aebeed3b5261ca3b6776192c3b6ed69fa320a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:06:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7381077a21934f54869a95065bd01923\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/901396913a504a3480f0c9660f156953 : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b35a163d261145d297efbd297c373eea : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d774feca5ef54256917055d9b12239f2 : SUCCESS in 19m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3ebf3f9647904cd785c7326e90413800 : SUCCESS in 28m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ea120401d96843a083063d3a3e82e15d : SUCCESS in 1h 05m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ca10196741b34f368b938c71476e4a38 : FAILURE in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fee70984734568957434125815c9b8 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63c151e3a1e04e3e85eb703814394d92 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fbd2a60ae5c4933ab60fb4367d58298 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e9af584dd24595827bd85194d7f174 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202fd8ee888c4ef7bc8cc12f74c42b8e : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ee0b9496f5a74a2fb1fd2124bcdab170 : SUCCESS in 24m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d8123b0e6bb84b3e9aa8825e7032310e : SUCCESS in 40m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f636efd4e2dc45e2a9e997dad0993537 : SUCCESS in 58m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aafabd2bd1644ab9af0a364fe61382a0 : SUCCESS in 1h 25m 33s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/228f5971372d45fe9112835d651b256a : SUCCESS in 48m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2ca0a3d0d67048989aa73888e02adfe3 : SUCCESS in 23m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4dfa1ccb13614fa5ae53d19b93d2f797 : SUCCESS in 25m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c35503c0b6674193af73f5e05192f7fe : SUCCESS in 1h 01m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5638a0351bfb481eb969beb286192962 : SUCCESS in 1h 11m 28s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/8ae23ea0fab24e9fa9815165aac28351 : SUCCESS in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8165f2d753951f32335b6489755b066387c631b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 13:43:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed71b12de3f548d4a91bcd8e857cc3b6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dc3fbe49ee7449df8afacd1c814643a5 : SUCCESS in 2h 30m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ccf9de8a4e124793a9a756674a77a27d : SUCCESS in 2h 53m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/29a87898e3cf40a1a07d328ad723d56e : SUCCESS in 1h 33m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2de2e4b60fa4b23ab9b3ee4865161af : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60df1983ae494215bb7d9317090822f9 : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a03e02a6ddfb4484a8864ccceda45557 : SUCCESS in 2h 23m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bbd5da27a36e44c790f17363eaf5a39d : TIMED_OUT in 3h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"448cb607ea5422ea8906e15634cac0771247d8e6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-15 11:21:27.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ac5a91e8205980f8fe05b6b5f7e94e2bb95716b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-15 11:22:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"639066bbae434dd33d7026a41556255d5df203d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-15 12:39:20.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e893ac1c71bd77c44ec9cfa3910b9b7f9ebd056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 12:49:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0467cf715af541258fd1808421370f3e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6222f99eb62540e39a91905d4364d97f : SUCCESS in 21m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cfb7cff987ab45e6b348c075a9ee5c55 : SUCCESS in 1h 10m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b31dd3b3ef514cdda53a1ac87160fa4b : SUCCESS in 18m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c5c4e1baee734091a3f1a0d49b379e82 : SUCCESS in 29m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aaeb1adcc9064772baeda4b8611c7490 : FAILURE in 9m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ad8e1c6373e4623a0fa43e450e1985e : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345a15f846f74023b20b4204cd784b9f : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/668161e7ee8541c29a737c8b4dec659c : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fba397bbfc66479aa8c921ba3fa1e4d8 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbad038223b84cf79c4accdd416fbd61 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16daac048944b4087ac1d282f45e828 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9b56cdf1c6a148c1b9186eab9f02b1ad : SUCCESS in 29m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bb9854f48c784732ad7c2fee501b5813 : SUCCESS in 35m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c25139970f524c648173f36e2f59369e : SUCCESS in 1h 12m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d63e2e16ddf84f1ebfd88f330e207535 : SUCCESS in 1h 26m 39s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/d215cc0f8b2d494587c072c89eb9e562 : SUCCESS in 45m 37s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c93ac92450c430d8e9290935cf323b3 : SUCCESS in 23m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1763b09c1837479c928a3c57fe0ab051 : SUCCESS in 24m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8179c5f0d8414c9f8047978af857b88e : SUCCESS in 1h 00m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/401e0eee524640df90678c62358c05a2 : FAILURE in 9m 51s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/ec2a316f0b834411b3b75f0f26ebbcab : SUCCESS in 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1421eac9405dd43a7917a09ad8b6627937ece28f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 14:29:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0c4392fe7f4f33acf85ba6231e2e40\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5ad09657ac3d4a46a1be0221a7060a65 : TIMED_OUT in 3h 01m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30ce254d4d0645a681ba117e65470ebf : TIMED_OUT in 3h 01m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8a9be3760fb5487c9269c1557b0ca47b : SUCCESS in 1h 15m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b867bd9179de423faa9774054ae82abd : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/341abc606c9340eda8924450afe9af8d : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f1f19e9874394a2681d2b18a59524b54 : TIMED_OUT in 3h 02m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5f2314bac3b540208ca721d278586862 : TIMED_OUT in 3h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"045e1515e141ccfa8a22ff9d5d47c67c32e0d831","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"date":"2024-11-20 12:34:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"95e522e414e326a55bebc2000c720959ef59285c","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"date":"2024-11-20 12:36:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18cbfc7e2f264fed47ed3096a9456f4edf80ae66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:20:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217b1f97a60c437abfd1826acb87f61c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1d8f485a90244c889fe78fbc9dd21a7e : SUCCESS in 23m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d776dd5b333e4a6f8184015bca01c6b9 : SUCCESS in 1h 04m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/03819d2c8bfa459791ce782aa3d9afe5 : SUCCESS in 21m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84dd31b404b24e33a554c7025de452f3 : SUCCESS in 32m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23dfe82e11284af6baa5a2922f69ed82 : SUCCESS in 1h 04m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/21c2d7db44e84aff821e3b38b0430971 : SUCCESS in 1h 37m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e266122bdf43f1aaa5e54c73ec6b57 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc5b2a04a3ed437881919ede18c9f347 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ea99649872d4c3daa75c6ad54f10530 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e758fd971d8c4ea0b4401b9051e9a4d8 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61619d1c4f034b0184858e70cdbb0dd1 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/45a087caa38b4442a98e99192632f50a : SUCCESS in 33m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8d3798cf5da244caaa707fc624df964f : SUCCESS in 35m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7eec6820398a40dbbb1504d9eba5d632 : SUCCESS in 1h 06m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/48be28f7acc34cc59b343dda92e43bde : SUCCESS in 1h 27m 00s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/639573b76ec24384b4b42d7a89c8491a : SUCCESS in 45m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f0c44594d0d44b4b8b9cba2d720cf55 : SUCCESS in 29m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/61ce6d26db314c13a5fb08eb1184d384 : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/101e0f9b865c40efa5197d36c9276c2d : SUCCESS in 1h 08m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8daae7f617164193b4ec56475f740bd0 : SUCCESS in 1h 21m 20s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/d84dfbe0d2ab4a74a3796677f6416541 : SUCCESS in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c1ccd6f61ccf920637f381aef82c39d84b1878f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 15:45:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/456687629dc74702a7c00c69ceb5c06c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e19d5a4f667645a4bd3ee66569b30e96 : SUCCESS in 2h 24m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c22b9d071434458da2e47300e1b19b40 : FAILURE in 1h 25m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/870a44bc477c4b4e97962150d256f20b : SUCCESS in 1h 49m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1107cff0b087441e9a3414c2ecd6379d : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3c36f27905c4f5ba8113095cc35f288 : SUCCESS in 4m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/19101fd094d94692aa629fb16043de59 : TIMED_OUT in 3h 02m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e1c1f6861b224e768c4a7cc85a9b9473 : TIMED_OUT in 3h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c915bfdeff8bf0bee0bd921c489301e0a440c517","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2024-11-21 14:36:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b11c9e8cf9eb1cdacee297eff6da5a89ab3f69ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 20:25:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96223a2a5811c645f3d0cb21d836c82a4bb568f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 20:41:22.000000000","message":"Patch Set 3: -Workflow Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"9183bc0ee31d236d2e74bbebb1cf335d5112a286","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 20:41:25.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a3b9934686a06037ba575ce7b51e05d7cf937279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 20:41:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3de70280e8f49c84cf155ff6de7e324d69bfca3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:03:23.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/711e6c94020c4e7faee476954856de68\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b53c9101c2e4ce6ad2297a0e1ee7d2e : SUCCESS in 13m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d89f7a0e130b4e80a439e71c1a828177 : FAILURE in 14m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1168a4a423c548fa82df236ed0f1c5ef : SUCCESS in 1h 00m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9559536d86724c6fb2c823a1d5b4b928 : FAILURE in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a11186c51e4e1a842d629887f43290 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d61b88580e74a1e829abf8c013bd392 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9aa6c21e829341739077b98eecacb307 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ba6c7e763a4e559136ef74a079a3b0 : SUCCESS in 5m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec64bfcd58b74c7181ef3cf62a6a90b9 : SUCCESS in 30m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e864fa84a7d4039889ada9628005bdf : SUCCESS in 37m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/09765e1aa08243d2bbb33eb68d00bb91 : SUCCESS in 54m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ca8337bd7dfc4e3586429e493bd1652b : FAILURE in 10m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b1f1329849c84064afcfdb403a43e8cf : FAILURE in 4m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dc03b04f5e29443ebbfe15d5be0deb6b : SUCCESS in 28m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c73ea544a435461293c987c16f33140f : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8242d07bca1d4c399f87ac24a0da07f1 : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"758e407f089d7a4b67adaaead8a14b6698d3e717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-22 05:34:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37430\u003e\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"}],"_revision_number":4},{"id":"213e1b9a3217655961457be6d1cac7c074868cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 06:54:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d57282480de54ac2b8a4a4c87d0e07d1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2dea7f47e454429ea83f10832252d9af : SUCCESS in 25m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/55d457940a144d3cac4c50ab0be79544 : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/26d2d2f83d564acf9b780209630867a8 : FAILURE in 3m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f7e24bd2e2294083946a6d8a2a05bcb6 : FAILURE in 12m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dc93630b4fb646be99a7e59f573dc2fb : FAILURE in 12m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5bf62389e5544a8abcf8d6ec9c90c7e3 : SUCCESS in 1h 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e4ae6a5a494c8895fe0307fb4b76f4 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/835cf41c8ce042f2acfac6358d7c87f5 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe7a129de2184565bc99e2d3067a7c5f : SUCCESS in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225b887ce08b4c2eb4b04d8a8f7dad39 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1695dd4b9c2545808ddfa773310c8c3e : SUCCESS in 5m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d258a207710a4684b524813a26d89f72 : SUCCESS in 26m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/35b1b3e035954a00b54b793759edf86b : SUCCESS in 33m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/34f5dfc811e74cef877953f80293e815 : SUCCESS in 59m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fb58e4898102423e929f13499faf870a : SUCCESS in 1h 19m 18s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/558435268b154a2cbd66c764ca77ae71 : SUCCESS in 50m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/22e90003b6a74bc3b48c77fe7a35e056 : FAILURE in 4m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/50edb34a94c840559dda78040e8023d0 : SUCCESS in 29m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2b9356f026b04e398107967848e430d4 : SUCCESS in 1h 08m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3294c093bebf4887bb099388eaceceb4 : FAILURE in 8m 20s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/49d1087f6fd44517a4525aac51bf99c3 : FAILURE in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8fa03c1536789bab27e271de5d15ec218786c4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 08:39:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00883a9330ef4e82ac2107e35bc805ee\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/74650f884655477f9c42d570ca0e09d9 : SUCCESS in 1h 50m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a275bb16866b4e758feb0f48c6e0806f : FAILURE in 5m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/56a381e8090a48c1ad670de248384653 : TIMED_OUT in 3h 01m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe527a3d1daa466e81fe9091d69c02e0 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc9db058b28d474fb505d543b1ecb5f4 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d46c1208b08e49cb937c927f384f1121 : TIMED_OUT in 3h 02m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03c080e2ed7140038cda5ac74ac86743 : TIMED_OUT in 3h 01m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1a91a17247b117e7d6c8a57b4b88ee71f8b7990c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 14:19:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_37430\u003e\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"}],"_revision_number":5},{"id":"bbddf2aba3f138a0f5a51f7b3dab6f09e6bfd4d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 15:26:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cb198cbeb04fc399871db1ab626d9f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bab57882ba9b44a495e8c279b33a489f : FAILURE in 3m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/524a4fbf60a94c1ab804d374410b8dac : FAILURE in 3m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d61ad2fa664a4f66ade618a90468d58e : SUCCESS in 53m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2988d4ca1dc34a82b4448e48fc5879ba : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e2a0ac38d404375bac0f6eb9bbb9c2d : SUCCESS in 4m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4215c64f0d304aaebfe612487a39da7a : FAILURE in 4m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c0840e3e23a443c08068fb543fcbbcf5 : FAILURE in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd11130a06ad2678e6df7fd4666335b5c0137a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 16:00:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c2145de8f94c9787976e1c8cdd6554\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b5f2b360290f4761a4fcbd5c75234450 : SUCCESS in 30m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/44cc1d589c8e42de9227d54039ea1c4e : SUCCESS in 32m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef14f85438a84236b5f2ef70978d2ce9 : SUCCESS in 18m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f4404c2acd8d4ab98db0b956b812ef48 : SUCCESS in 29m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e84f43d3c04147a98d9ad8f89047cfaa : SUCCESS in 30m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7db8438d5a6b4e8c97c2324255a3cfa9 : SUCCESS in 1h 25m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee254395d0f449996905d0305d94710 : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e91e495d53476492e1fe50211299a5 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/733ae68f61af4cfb8b412976f14e671b : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dce41cdfa46a4edbbd4818004ede427d : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb506042fe048b4860d0a589110a1f2 : SUCCESS in 5m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9cda4b6ec364fc9ad7ae66dc3ee304f : SUCCESS in 1m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/19073f5a76bd43768b21f18e0e57555e : SUCCESS in 23m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b94aebef35941d5911cfdf2efded2a9 : SUCCESS in 1h 27m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82a8980e4ac141a7ac61b5d47e35da11 : SUCCESS in 37m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b89c1e986e4b4b94992f6be20226cdaa : SUCCESS in 1h 31m 36s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/72952690cd364e19aa1c697fa87cb81c : SUCCESS in 57m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5657b99eb46144a2aa2c4c6e984cbb1c : SUCCESS in 25m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/02431cb5520c425c97edeb0e01bdd194 : SUCCESS in 26m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8d7f1916560d484098db3d1d472e25ab : SUCCESS in 1h 10m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13a07f9df12e4f7c866acfb5b46582b8 : SUCCESS in 1h 38m 18s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/7786b098241b4245808b5bd39bcabeb7 : SUCCESS in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f592ef8dfcf5f3884b0c0036963d978faac942a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-31 23:48:32.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"63bddfd8ae9dc0b6854c45e621f931e988eb6941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 23:49:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"45ae3a8656554af4961b227a757d470ac72816a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:31:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2bcdb9fa62461eaf023ccbc0635789\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4d6b46b84ed042c297c637e021c46296 : SUCCESS in 23m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3cfdbd0559934aa683be8e88045e5028 : SUCCESS in 28m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59161ffb8faa420a92e2801d1824cc46 : SUCCESS in 1h 04m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cdfccbd68c404761a0a89ac4892467c2 : SUCCESS in 1h 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b6983c031c4d58968e2a00df68c1f8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8415af885e804afc8ebe2a5581adb3ba : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09aaa225f61e48b7bc1f1cedeb0081ab : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ae22dede04431a924542978ac57e57 : SUCCESS in 6m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a1914b9934b144a2a9aaa0385cf116f3 : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e0b5e4e97fae477a8ddb5b3a276879a8 : SUCCESS in 22m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fd3f8e0e52e5406e8a9aab54ed29edfc : SUCCESS in 1h 06m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8bb0083333cb4016bc39982dd46b4fe7 : SUCCESS in 52m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8ca9c870424b466eafc53cfb8c36397d : SUCCESS in 1h 38m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27c6314b63c0414b82353c72e4805626 : SUCCESS in 20m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a2d4e054b4a54e339ad7ef9408b4cc91 : SUCCESS in 28m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/62e5aa8b61254fe997b74ea52acd993b : SUCCESS in 1h 04m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77342e0fa4834c6d9aece8c1af6b014f : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":5},{"id":"a873efaa6df67e7c1003013e1eda9bc19aa86901","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:31:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2f9f9892cd9d72cc30dddcf3c8ec580c5a736f52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:32:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b32c52c17bd4ca4bb49e9f82abd39eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5291f7276492410b800f64948b445114 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"09f706bc2edda00ec05b67316f7f9e1a20995ca3","revisions":{"0ff788691779164dbdce82a182523e82331b752c":{"kind":"REWORK","_number":1,"created":"2024-11-13 11:35:51.000000000","uploader":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"ref":"refs/changes/52/934952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/934952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/934952/1"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:03:37.000000000","tz":60},"committer":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:34:56.000000000","tz":60},"subject":"Added libvirt dependency to nova-libvirt container","message":"Added libvirt dependency to nova-libvirt container\n\nfix: added mdevctl to nova-libvirt container, mdevctl is an optional dependency\nof libvirt needed in 2024.2 for nova vGPU feature.\n\nCloses-Bug: #2088055\nChange-Id: Ifa92999672ed267ce877c492c9140fbee271b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ff788691779164dbdce82a182523e82331b752c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ff788691779164dbdce82a182523e82331b752c"}]},"branch":"refs/heads/master"},"ebb9c9e09ca1ced16d4c5245cb0048ddb73242c7":{"kind":"REWORK","_number":2,"created":"2024-11-13 12:51:44.000000000","uploader":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"ref":"refs/changes/52/934952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/934952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/934952/2"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:03:37.000000000","tz":60},"committer":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 12:51:19.000000000","tz":60},"subject":"Added libvirt dependency to nova-libvirt container","message":"Added libvirt dependency to nova-libvirt container\n\nfix: added mdevctl to nova-libvirt container, mdevctl is an optional dependency\nof libvirt needed in 2024.2 for nova vGPU feature.\n\nCloses-Bug: #2088055\nChange-Id: Ifa92999672ed267ce877c492c9140fbee271b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebb9c9e09ca1ced16d4c5245cb0048ddb73242c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebb9c9e09ca1ced16d4c5245cb0048ddb73242c7"}]},"branch":"refs/heads/master"},"21ab1327c9fd5b2aea8455f382a3da81073f77fc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-14 10:38:17.000000000","uploader":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"ref":"refs/changes/52/934952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/934952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/934952/3"}}},"commit":{"parents":[{"commit":"68fa7d302095b5362f176b7e690e3cbaa516fae0","subject":"Merge \"CI: Add reno linting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fa7d302095b5362f176b7e690e3cbaa516fae0"}]}],"author":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:03:37.000000000","tz":60},"committer":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-14 10:37:46.000000000","tz":60},"subject":"Added libvirt dependency to nova-libvirt debian based containers","message":"Added libvirt dependency to nova-libvirt debian based containers\n\nfix: added mdevctl to nova-libvirt debian based containers, mdevctl is an\noptional dependency of libvirt needed in 2024.2 for nova vGPU feature.\n\nCloses-Bug: #2088055\nChange-Id: Ifa92999672ed267ce877c492c9140fbee271b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21ab1327c9fd5b2aea8455f382a3da81073f77fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21ab1327c9fd5b2aea8455f382a3da81073f77fc"}]},"branch":"refs/heads/master"},"bab8771b2158ed5f5324a5bac6bb790865332e44":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-22 05:34:33.000000000","uploader":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/934952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/934952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/934952/4"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:03:37.000000000","tz":60},"committer":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-22 05:34:33.000000000","tz":0},"subject":"Added libvirt dependency to nova-libvirt debian based containers","message":"Added libvirt dependency to nova-libvirt debian based containers\n\nfix: added mdevctl to nova-libvirt debian based containers, mdevctl is an\noptional dependency of libvirt needed in 2024.2 for nova vGPU feature.\n\nCloses-Bug: #2088055\nChange-Id: Ifa92999672ed267ce877c492c9140fbee271b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bab8771b2158ed5f5324a5bac6bb790865332e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bab8771b2158ed5f5324a5bac6bb790865332e44"}]},"branch":"refs/heads/master","description":"Rebase"},"09f706bc2edda00ec05b67316f7f9e1a20995ca3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-13 14:19:05.000000000","uploader":{"_account_id":37430,"name":"Mickael Razzouk","display_name":"Mickael Razzouk","email":"micka@ikmail.com","username":"moot"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/934952/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/934952/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/934952/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/934952/5"}}},"commit":{"parents":[{"commit":"9efdcc01ca5a66304568a6582a182e4171e3e625","subject":"[release] Use UCA Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]}],"author":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2024-11-13 11:03:37.000000000","tz":60},"committer":{"name":"Mickael Razzouk","email":"razzoukmicka@gmail.com","date":"2025-03-13 14:19:05.000000000","tz":0},"subject":"Added libvirt dependency to nova-libvirt debian based containers","message":"Added libvirt dependency to nova-libvirt debian based containers\n\nfix: added mdevctl to nova-libvirt debian based containers, mdevctl is an\noptional dependency of libvirt needed in 2024.2 for nova vGPU feature.\n\nCloses-Bug: #2088055\nChange-Id: Ifa92999672ed267ce877c492c9140fbee271b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f706bc2edda00ec05b67316f7f9e1a20995ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f706bc2edda00ec05b67316f7f9e1a20995ca3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
