)]}'
{"id":"openstack%2Fkolla~935974","triplet_id":"openstack%2Fkolla~master~I581ff79993d011b421d69295a6552cb14b8e2e66","project":"openstack/kolla","branch":"master","topic":"kolla_uwsgi","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-19 09:01:02.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-18 12:07:52.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I581ff79993d011b421d69295a6552cb14b8e2e66","subject":"Move uwsgi installation to openstack-base","status":"MERGED","created":"2024-11-21 19:35:59.000000000","updated":"2024-12-19 09:02:22.000000000","submitted":"2024-12-19 09:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935974-kolla_uwsgi","meta_rev_id":"860daf6f26046e51872818de188f3089e0b7a59a","_number":935974,"virtual_id_number":935974,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 09:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-19 05:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-18 12:07:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-12-19 07:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-12-19 07:23:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-12-19 05:59:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-19 05:59:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 21:04:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-21 22:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-19 05:59:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-12-19 07:23:51.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"950beef6f5c730aa2141ae12dac7b67598f271d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 19:35:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"401f41d4139a2ce669aa3fd4e6c08b13c227b187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:09:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06a25b5017c949be9c7fabb28b661e78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ce911a1823b0460fa1968b3e2d8b45ba : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0d89ac96fc2345a5b0778151b7546fbf : SUCCESS in 57m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/19a66755b8da45cda08216d11214e045 : SUCCESS in 23m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/76894b0862324998a53d37761f399293 : SUCCESS in 39m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0218378d10b04899a5a44849b41ed2af : SUCCESS in 1h 10m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5da407fbf07e49e1831be206c831a69b : SUCCESS in 1h 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44b2bc881ff4cdf98ddcbd04d2c3490 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ba2783cc5334f0b81fd3c148211ffb9 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60b365997dc54835993b132f9935cfde : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14607eb1313f497db3d35560ffd3b2c6 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411e63fa30ec4bdcb032afb4e5233624 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5411e2b1c8f0425192e39c3dbf16b2be : SUCCESS in 5m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dba52a8301ce42d69fbe5f8281ffcbff : SUCCESS in 20m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1818dff0c3a0486bb25a58d066b714cd : SUCCESS in 23m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e47b7119737a4372a6f774bf8d76f067 : FAILURE in 10m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/43910c796119449bab1bf09bf3420983 : SUCCESS in 1h 25m 37s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a5e4ee742f52483e93030eaf60dfebb2 : FAILURE in 8m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9da0ba4ee55a44acb0d3fdf0044dcf08 : FAILURE in 12m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1d3ce9e346f847ec808ae6a396eba148 : FAILURE in 11m 42s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e5f41fb7ed514fd3afd81c9e0cf8de8d : SUCCESS in 1h 15m 25s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0a2123e57ddb45c5864774a9f7074065 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f05205d8e02b4092b71b30f64571dfde : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4b543ee8e2834bd489d9d366b923fc83 : SUCCESS in 1h 11m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/50a26ac2c04b486bbe70d347841927fb : FAILURE in 5m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0fba5183491e4b27aaa8ef10cd36c615 : SUCCESS in 35m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b5c35b23b3af40cebd2b5c9934f3bd2d : SUCCESS in 1h 07m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/41ce7e8303094fe1b60ad070e2dd6014 : SUCCESS in 1h 21m 46s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b30a100dd1fc48a6b4518c6e9192308f : SUCCESS in 1h 51m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cfe57a92075a49b1bad8d46a2baa7734 : SUCCESS in 2h 25m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3acf67e83d4d4ee89b19a0631b92c8e1 : SUCCESS in 1h 22m 06s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/eee786af5c2d45949d3eacb463b46505 : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/050142ad12a941198e6aa3c0ef991bd3 : SUCCESS in 1h 01m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/65c3fa2c17034947a401c827fd0d13fc : SUCCESS in 1h 15m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b8a4691ef0f7455688123895873056c5 : SUCCESS in 1h 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd89b83cd95fedf871963c73ad3ad49d85a28472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f04d407b2d6044018e9b4a74ca2d8c79\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ef5b0be6bad842deb0050572cdb4467f : TIMED_OUT in 3h 02m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/911ee75ed26e4df18084abf108f833f6 : FAILURE in 56m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1fe9048ea76b4548a1705157b8157a24 : TIMED_OUT in 3h 02m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b782349b7ff4e0eaa05883c4231d8ed : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/773fbf6436cf4d78a5a7461abd57c7b0 : SUCCESS in 6m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4054afe7bdb5457a89f135f50f32bc0d : SUCCESS in 2h 43m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68d9ed9c8b24443aa4922f1128d73887 : TIMED_OUT in 3h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d307e83f7ca34e73da7f4634442b5adb12ca3d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-22 05:43:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de28e399aaead688f029dd0ac47845def155cd6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 08:42:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea2150f05db2433c878dc157f7ad06be\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e178e62da4804d3194ab558eddbb4350 : SUCCESS in 30m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f49f952632e8425891c1381dabd12d6e : SUCCESS in 1h 01m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/208113e99a3c425490eda96a2fa1eec7 : SUCCESS in 21m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ef7b2af778449939d98d4dbc3b08f4a : FAILURE in 12m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b8f564e3b31e4bcdbb3fc29a5b2dbd4e : FAILURE in 8m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e32615af8e83499798bdaca6f5005e2b : SUCCESS in 1h 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19068de8dbfc435c9ab2390b199120ae : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4052bd0d9d14c5eb9d92036322d9f06 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0bba66478df4f998979d8535b466a03 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44250ffad00492782ca38d26562f31d : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59114f07448a492f9ae3a3935e90db81 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09bfe1749525490488b9edb5fcf19d54 : SUCCESS in 2m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85eda6b46e0c4d52878d2958400fc65b : SUCCESS in 23m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/be70a53459ba4d91bce00d0e8f4dbe11 : SUCCESS in 30m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd5a04b916434b9c8136bd8c729d71d5 : SUCCESS in 1h 07m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f4f8c024ecb94138b62fd53883d4ac80 : SUCCESS in 1h 27m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/75597d3a8272407b9a8b209d0d0d74d9 : SUCCESS in 1h 46m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5f019f6d6f444e88a69236cb0745aa18 : SUCCESS in 2h 21m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6207d774b3e14166ac85346f2b42c380 : SUCCESS in 1h 15m 22s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5b7cec17e0684885ba32c9ee23eaea7a : SUCCESS in 1h 25m 05s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a70b1b7340674e53bd21924939914541 : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/483451a4b7954f8cac21aa748bb16e24 : SUCCESS in 1h 17m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/50f627fe3b6941e2b4024866ba01601f : SUCCESS in 1h 25m 24s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f9703b0301fa4080aa6587056d82c609 : SUCCESS in 33m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f4223a1c0c6048c9b5ea21b2e7b81f24 : SUCCESS in 35m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/165786c34ed54d4484cb2706462cf305 : SUCCESS in 1h 14m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b6f0dd4e0bfa4b4985046904efd835b2 : SUCCESS in 1h 28m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/25cd43e4d30e44e1a43f110708835651 : SUCCESS in 1h 53m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a33acf094c044f589660f8483432faeb : FAILURE in 2h 37m 06s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/75c99415500d4c6487e49bac943ee3a2 : FAILURE in 9m 16s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8048920a9c30462b89a0f3286aa23707 : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/970dc9f3d9cb4f5f8bf8de12d2d8b535 : FAILURE in 11m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/464c62e750a5436282a3ef61422c8515 : FAILURE in 11m 11s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a70f784ef0a44b748f3def707c5baa53 : FAILURE in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"310dd1b3b0d6cc72474d654cefa58505666313a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 09:05:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c9e13321c84ee287ee52e8bea2cdd7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f3e368db50774cf1a4599fa81637bdd4 : TIMED_OUT in 3h 02m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4fea31998c984c07936722b248d15ff8 : FAILURE in 4m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/71591f5e3d3b4a129e6ce94acfa6a613 : FAILURE in 2h 54m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4abbc9d4e2304f9c91ef2c32ac45f6fe : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ad7521b12bb44c19cc11ce64f129525 : SUCCESS in 5m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/09220645fce1435999a5bb8d6780d761 : TIMED_OUT in 3h 03m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4b02a7c8144b4edeb5232ea8e92003ae : FAILURE in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"230429120819d87fe9b13a858dcb4b2b0b9fb617","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-25 09:46:21.000000000","message":"Topic set to kolla-uwsgi","accounts_in_message":[],"_revision_number":1},{"id":"01d2759a6cc4c8b8da2f76010cbfdfc30e41c496","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 17:28:25.000000000","message":"Topic kolla-uwsgi removed","accounts_in_message":[],"_revision_number":1},{"id":"1d772d5e0b01f898831f74206d657b5ffb67be14","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 17:28:28.000000000","message":"Topic set to kolla_uwsgi","accounts_in_message":[],"_revision_number":1},{"id":"bc6b1917350ba47cdade7626a17e1f6ef1ee88d4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-05 14:41:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e14010a7866a331745a1554ef0b640ed0f2ebd16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:17:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a916522e76224090b5f6e4d7294e324a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/34ec5f1b29a1463aa6bf5528b1170674 : SUCCESS in 24m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/922020e087684420a1f51ffee2f147de : SUCCESS in 1h 15m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/72e7fe7e086749389773ce356428673c : SUCCESS in 30m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6cda4a0de75846f7b330ff1a00ffcf3c : SUCCESS in 34m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dbb1a9d3f43242a0af3d0d709acc5dd2 : SUCCESS in 1h 14m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8d1a1fa7bfa24a1a99defd8038ce8d38 : SUCCESS in 1h 34m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0563446db6644d3094703b74488e8cf2 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/392ece68bcfa469984667570ca522404 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f6009ede2404687be34cfebd85e3258 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070585bb008e443c855d4825ea854e9f : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85ffc6815214d96b27fa3705992ce2b : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ddb8460a34f4cb8a79afafa0be277a9 : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6ecb95dee0dd4b9a9db727a96888212a : SUCCESS in 26m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ebb8abdfadb4cbf9b05ac65620d634a : SUCCESS in 33m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aabf757034374367be270b6df79421f5 : SUCCESS in 1h 07m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d2ce6cc5d98448c88f6ca54f16d3a731 : SUCCESS in 1h 34m 37s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e8fac547c26f4c6eb05622f26088a0e7 : SUCCESS in 1h 45m 30s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b0bdf22e24ae40a7b4cc28fc8b975fe8 : FAILURE in 2h 22m 02s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/45b225a05f744894abbb1497094fbef6 : FAILURE in 1h 09m 52s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f3f88767c5c04c43915014a988e28355 : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c7471fe975224626b8b83c8e60dd7c10 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ebb16c292fb9426a85e85c43b9534a9b : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e910008843ed4259b38359d951db6bb7 : SUCCESS in 1h 18m 41s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdbd13cb2fa245cd91eef2aae6a0719c : SUCCESS in 22m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8a81c13b3b7046d585ccb3c8b24b9404 : SUCCESS in 29m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b90a615476d14e84bcfeb061d10d3a26 : SUCCESS in 1h 09m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9941fd2559b74bd09649ea48683a5853 : SUCCESS in 1h 28m 10s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f26e2a51ebe649c397d01fea16359819 : SUCCESS in 1h 55m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1bf02c258d6d46db9f148a226f8ea60a : SUCCESS in 2h 19m 48s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fe15ee3c22714b90914e1a36529bf400 : FAILURE in 1h 16m 58s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5c2583c2262d488fb1c523d9e2d1e0ff : SUCCESS in 1h 27m 08s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fc4b4f8fbc6841058299dcfca7c91ac4 : SUCCESS in 1h 01m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3d2658ed53104d639415781f2f1f7727 : FAILURE in 56m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d3f8fd9e1fa1473aa86981c3eec1e82a : FAILURE in 1h 06m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9349f9bf84c9c8071f39eafb99f8616c1090b91c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b419ad6f19604da0944507bab3f66e05\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/de22d091f3c04e388cd42a919b02a4a3 : TIMED_OUT in 3h 02m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af787229f4fc42a7b955455e35c2850a : SUCCESS in 2h 26m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ed57b4a11fa841cf8dfef83f992020f7 : SUCCESS in 2h 02m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbcb707ae7e94660bcbe461266e1205b : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/560afdead78b40e493f9ff5260bbceb0 : SUCCESS in 4m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5d0953a4f6f44ff1814b2087166a9dbb : SUCCESS in 2h 10m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8c787b266a3a48a5bae30f97435f4791 : TIMED_OUT in 3h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62d87b2593090a4aa8e5e9d5e9e7eecf121c76f7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-18 12:07:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a537b65ba032ba0dccbef9ccc8c123a5c00a9c15","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-19 05:59:19.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b2d31cc47d65d2e3e83628e590643d232c61d881","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-19 07:23:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9dc7ae3988fafb7b5da334617da0581293359abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 07:24:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89756d861aeee64125273b269d4319e544040e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 09:01:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/482ceed8a6694ac994d5dff3611b8439\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a68bd4103b34336bd3f45ed31133433 : SUCCESS in 21m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f02360e9a3de4d408e1c91eccc99a5d8 : SUCCESS in 25m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2a196112e42e478a900e34d0a2202ff9 : SUCCESS in 56m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/079c1f7e899949c6be89b798458ea35a : SUCCESS in 1h 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674d7a4b1cdc4bab9e62b64dc9a30bbd : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/903fc4813ee4427f873b12a2fdf27c5e : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c05ef99e1faa473297c0fd0daf131a13 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b55fcf84c24bd8a42c88c60edf8191 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/add21afef7fd405e9156094354e8c96f : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/21e8116cc3c24b93a1f67d1f285d45b6 : SUCCESS in 24m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d29028338a184f1894e7f24abdaa6654 : SUCCESS in 28m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/613da72fb6c9483a8fc524b693c7933e : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7caca88aa5814c57bbf2b2fb25979cc9 : SUCCESS in 1h 30m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/543bc918b739448dac73d7399c3dd0a3 : SUCCESS in 13m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2c709587104242df8225b15accbf435d : SUCCESS in 31m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1696a003747242e5914dd79337e1df76 : SUCCESS in 1h 09m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6edf588c0f404c7091da93a6eaf379e9 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b04515af6ed350455d0b888ea33d621e9a11dad7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 09:01:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"860daf6f26046e51872818de188f3089e0b7a59a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 09:02:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526d31349cf34e10946896fb86ae88d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072bdac644ea482490d58a97cedc66de : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97bb213216e4f8dbd601b2c3de5593c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1df09532d797a3b722185aef6444e53c2676bac9","revisions":{"1df09532d797a3b722185aef6444e53c2676bac9":{"kind":"REWORK","_number":1,"created":"2024-11-21 19:35:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/935974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/935974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/935974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/935974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/935974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/935974/1"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-21 19:34:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-21 19:35:57.000000000","tz":60},"subject":"Move uwsgi installation to openstack-base","message":"Move uwsgi installation to openstack-base\n\nChange-Id: I581ff79993d011b421d69295a6552cb14b8e2e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1df09532d797a3b722185aef6444e53c2676bac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1df09532d797a3b722185aef6444e53c2676bac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
