)]}'
{"id":"openstack%2Fkolla~936640","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I3fab494e939478458c2c299303f5c2c91943e0ec","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-29 19:36:42.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-11-29 08:37:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-29 19:36:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fab494e939478458c2c299303f5c2c91943e0ec","subject":"CI: Use debian/ubuntu images from quay mirror","status":"MERGED","created":"2024-11-28 16:07:36.000000000","updated":"2024-11-30 00:09:50.000000000","submitted":"2024-11-29 19:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936640","cherry_pick_of_change":936639,"cherry_pick_of_patch_set":1,"meta_rev_id":"de227ebf198aeaa32a0878ab6038a4144689744a","_number":936640,"virtual_id_number":936640,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 19:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-29 08:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-11-29 17:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-11-29 17:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-29 07:25:42.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-28 16:10:55.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-11-28 18:21:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-28 19:39:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-29 15:56:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-11-29 17:59:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8622ca7dc9d70b1a0d65e4d3a4d7ba0b2c645232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 16:07:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"3a32671797d0d0e3055cd3b5784e3c58fe0c0a9e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-28 16:10:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7b6c4a503b0e3758c2d692db65d990bf09eeb87","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-28 16:11:00.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a96f31cb936105a5324146abffd180906b95a96d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 16:25:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43ea66fb03cf436612e78b6c000ca5e7863bea05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 18:21:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a4c010baeb406da615576c446e00c1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ca791e89447249e39a90759a1833f053 : SUCCESS in 32m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dd9c133d6b524f0f9393146aaf81869c : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6930b1c2b99d4e69ba189d09098b8590 : SUCCESS in 19m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/002aa2d7a7e640cdbe75db229c70c6f5 : SUCCESS in 29m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef776856e72f4c0ba9fcd607011f5932 : SUCCESS in 1h 12m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/868b5a4519c6498a987f93fc686c05bf : SUCCESS in 1h 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e32e8c1c79a4dc19a011f980c6fc9e8 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff2043c265949d99b65bb7406de7069 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cec38aa03b6847f2ae7afa167e40f285 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aef2c3961c94a7d86584cb8f03332d8 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab6362bee7f46588ac7a098195d39b9 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bb17789cd7714987873dd3f5c6ea2d1a : SUCCESS in 26m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b4745692c2b4a8d85fda9fac52958d1 : SUCCESS in 34m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/47d629636b394958b2453b213e44e06b : SUCCESS in 1h 07m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4d637634ba684456937302fe9b13a77b : SUCCESS in 1h 24m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c47f1e7a93d7487c98aa4fc62ce00928 : SUCCESS in 16m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/34da3a757d024d87b7a355317b42228f : SUCCESS in 56m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f5d8e64335c5499393cc73117ffbc523 : SUCCESS in 59m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e6ad4d0699e24f4bb4d0c99dc2a49611 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"29e2cf811076cba9c01074a34271436c3a5099f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 18:21:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2b556446dbe19619c92b88ca3515e3399d58939","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-28 19:39:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b8c26ce49f4098ac11ac6ed12c4b271f934470b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 21:19:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90f3e7490dba448ba059d2ed7ec8c005\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/600c0b9c59d24f79954d54dc4ae2dce0 : SUCCESS in 19m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/60ccf5e127264970aa0d0661ff7888fb : SUCCESS in 29m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/49b1f26818cf491e8c69c0b36d8a5589 : FAILURE in 7m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1193a9a8f8594efa8ccea1c6128a7c05 : SUCCESS in 1h 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c2c48a55cd4dd0920fa7870cb2d4f9 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34a5db1bdbe04460b12383fdaa48161c : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/306836141a5f47848e9c93bc3a3e935a : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae648a5398a498d914740a389ac6810 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fb007ceddf3a4512bab80678bfbe65d2 : SUCCESS in 25m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0540c67a542c4d768b7efb3c5ad1b542 : SUCCESS in 26m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b90c49bc50649ad9cc294de95b917fa : SUCCESS in 1h 07m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/eba815e6051a4ad6a25496c18baad9d3 : SUCCESS in 1h 26m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/514dd231ffeb40df94ea0d4ff9b5669d : SUCCESS in 17m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b01d69c95248457fa48c157211f133f0 : SUCCESS in 1h 02m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e27a819e2fd4435b7c533373fcdde90 : SUCCESS in 54m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cec3d1953865439ab9a06fda72756531 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f4eb0945de1d5aa19bb77fa6f937822282e8ed48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 23:38:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b87740adf234caeaed95907897a00da\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57f5ec2159134509986d8b7f4465fd8d : TIMED_OUT in 3h 03m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b8f8e74710e8433e9ea783eaab9cafae : TIMED_OUT in 3h 01m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63da0d4574774d39972bce6ec6bb602e : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7688b35353a64fbe86125389f61dd4f8 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3ddf616410dde6f8c18e6a65cd16c8cb9a65056","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 07:22:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e3530a6cdb08b9b8876be01934a5ca7b4238bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 07:25:42.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6097e5bb2b8340fcbb8667f1435d5344754cf421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 07:59:00.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936666 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"a9f57e745b3b812aec6e61939835785fab73b8e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 07:59:30.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936666 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"588fc7f6c955fe2bce0b0b86b7cba0f909c896c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 07:59:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d9d73e1369035d7ec02dfd2b46ce3d6cd12ca2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:08:33.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936666 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"634650984845d909f9295ec1a6cf0df94f185e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:08:37.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936666 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"edf24eed9ed180f6ed0b6360dc829cbc6b39ed45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-29 08:37:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"250f221c18c9ba667bd4e1d31692122cc4af2533","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 15:56:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81bcf5bd01539e69ec15e0c39e07ce291b6c10d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:26:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88c0ee64702045b1868b5d3a90180da3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c1cc686bf63944b3aa67fc723827ed81 : SUCCESS in 26m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d8922184bb84677a4607d634a25a187 : SUCCESS in 1h 06m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/870e9b3102f04cab8c0e0a19fb4f5301 : SUCCESS in 18m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e0b786caa243482fb47eea794227d528 : SUCCESS in 27m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/958c83db37114d1c942bf76fe2d4f86a : SUCCESS in 1h 01m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b51c54586df24c3ea57fe0f72a461185 : SUCCESS in 1h 24m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae2112079dc4ba1baddc7cc35fe911f : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb78b10403d042709dcdb98a00a76119 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c2eb2f0096c43b3baa03ac16f38ac5d : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acc6ea11ef254de9b6f123ce4a598a78 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace542a348d44f7086dc532a015de9b0 : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/49e6c75275b34e628d0730f6de1bdca0 : SUCCESS in 29m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b1f6d415414b4684970cc88096bd1638 : SUCCESS in 35m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/90a46bd833c84daab8b3c363e5395307 : SUCCESS in 56m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9a70110ba3394beb8735f8576f50ceec : SUCCESS in 1h 18m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27060890d94443e89c39e63f9a671ff2 : SUCCESS in 22m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2e8822b451414dcdbb67b7a98c15a95c : SUCCESS in 1h 15m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9aecd5e69ad64ec49a8d678e26cdb6cd : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6a1353d9bb154f1ab56ee73f9c8dd185 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0c4f628bbb39b55a92be43dc55da419fdeea0baf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 17:59:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b0078e9372d01200bdb39c29c8fc50549db1074e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:59:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"167eced580c026c4ce3643cfaefbb4b7c4aa1d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 19:36:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272dba9ecec949e3bae4366c87782920\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ceafe2590f85477c99b94ca6e2e8516b : SUCCESS in 22m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2d332f1262948939034b8dbcd8f6b60 : SUCCESS in 30m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a2b1ce8586ea47e9961d277b96dfa21b : SUCCESS in 1h 04m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5c393658e1854128aae8b8afc7ca2bde : SUCCESS in 1h 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f92b9cf7dd430ea3f5f82d28356139 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dab7714298d46fc88981b5daffd0e34 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f11060ce7fc14712ac4f643298fa6369 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192dd0999e2c48d1822e2a69cd4b2af9 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/968d1b37bf894236b3548a3c96f6b09f : SUCCESS in 21m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/492f82614b964a358b7863aa1eb8c52c : SUCCESS in 29m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/97cd88c46a0b48f7bf5cade4672db97f : SUCCESS in 57m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/384dd4fb486b45d0b02249d8ada02d3a : SUCCESS in 1h 23m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a0f5a5376904fb580bd98ae9854ebaa : SUCCESS in 15m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/746f51b25d3247619de61ca333ebe7de : SUCCESS in 1h 04m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cafa5805b4e749c5815b67053fa6e996 : SUCCESS in 48m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cbb05e31d9e84394af9c2d4b54577cbc : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":2},{"id":"7edc1dfffa37672c978084ec9a017bfff6192978","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 19:36:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b95855bd662fe5b0d579c8cf017bb8959770d933","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 19:38:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f38e2afb1e724b0e9bb7a2e3f0632d4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b308ec02c0450686b9a5fc04d79010 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2},{"id":"0731edca068145a74ebcaf7fe8a293440edbbfab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 21:26:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de227ebf198aeaa32a0878ab6038a4144689744a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 00:09:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3efdb7e6985f4176aa345efb1b0b188d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1dbdc536e6bd44f18efe8633c94aa87d : TIMED_OUT in 3h 07m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa1d2355f230448a9427fb7fbb6a0c8a : TIMED_OUT in 3h 02m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07b34ccee7c84d2ca673373dfaa14057 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0beedef62004eac973b614e6106f64c : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a8d76e861076e5575906fb3c88906c5b9d93ed1","revisions":{"0ec817dbfcee11dd55b9336bfb1efc4bc5148818":{"kind":"REWORK","_number":1,"created":"2024-11-28 16:07:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/936640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/936640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/936640/1"}}},"commit":{"parents":[{"commit":"68e96e72f412afe793e103874fe67e32079ea670","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e96e72f412afe793e103874fe67e32079ea670"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 09:29:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 16:07:36.000000000","tz":0},"subject":"CI: Use debian/ubuntu images from quay mirror","message":"CI: Use debian/ubuntu images from quay mirror\n\nChange-Id: I3fab494e939478458c2c299303f5c2c91943e0ec\n(cherry picked from commit 7ba43d2820cd91b1108e5473eb0a9c26ae4a9cb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ec817dbfcee11dd55b9336bfb1efc4bc5148818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ec817dbfcee11dd55b9336bfb1efc4bc5148818"}]},"branch":"refs/heads/stable/2024.1"},"0a8d76e861076e5575906fb3c88906c5b9d93ed1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-29 07:25:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/936640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/936640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/936640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/936640/2"}}},"commit":{"parents":[{"commit":"68e96e72f412afe793e103874fe67e32079ea670","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e96e72f412afe793e103874fe67e32079ea670"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 09:29:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-29 07:25:42.000000000","tz":0},"subject":"CI: Use debian/ubuntu images from quay mirror","message":"CI: Use debian/ubuntu images from quay mirror\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936666\n\nChange-Id: I3fab494e939478458c2c299303f5c2c91943e0ec\n(cherry picked from commit 7ba43d2820cd91b1108e5473eb0a9c26ae4a9cb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a8d76e861076e5575906fb3c88906c5b9d93ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a8d76e861076e5575906fb3c88906c5b9d93ed1"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
