)]}'
{"id":"openstack%2Fkolla~936641","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I3fab494e939478458c2c299303f5c2c91943e0ec","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-30 12:12:46.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-11-29 08:37:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-29 10:37:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"I3fab494e939478458c2c299303f5c2c91943e0ec","subject":"CI: Use debian/ubuntu images from quay mirror","status":"MERGED","created":"2024-11-28 16:07:44.000000000","updated":"2024-11-30 12:13:55.000000000","submitted":"2024-11-30 12:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936641","cherry_pick_of_change":936640,"cherry_pick_of_patch_set":1,"meta_rev_id":"4134aea51504b4feab0671a6c7a1140cebf9cc5d","_number":936641,"virtual_id_number":936641,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-30 12:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-29 08:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-29 10:37:02.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-11-29 09:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-29 09:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-29 09:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-29 10:37:02.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-28 16:10:49.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-11-28 17:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-28 19:38:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-29 09:41:03.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-11-29 09:48:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-11-30 09:00:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"828c03be2cf4508a2ef55d9ee3a720e592e873da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 16:07:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"89935c8e0292d427b42b66e895f29bebcdde0c9f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-28 16:10:49.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a9d1631f7ced99b60cb8169726ebe6bb0aefd82a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-28 16:25:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84bbe96d7bd824b7720cd743bdcf13d74592991a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 17:41:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f708ddc0c62940edae92aebd149d32dc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2fb1d557cfc8447caa1f9e8acbf847fc : SUCCESS in 31m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/457e860fa1544460997506024bad99a2 : SUCCESS in 56m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eea36c59d917455ca7e314aa0a7b80c6 : SUCCESS in 20m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/343d0de23f354483b84323a3bd0bf273 : SUCCESS in 33m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2edd35f756594040ae8fec553580f734 : SUCCESS in 1h 03m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25167c7128d54a86b03b56b37eaa92c8 : FAILURE in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3cfb1d55bc435fb4f7e4b4248c8e94 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5f5e8b755ec47c5ba2d6cf0f00dc570 : SUCCESS in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/741d36721347418bbc92b5a8ce0b5412 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b6f9d0bad704a6fabe17c1f0fab34f4 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59245abb8bea4035a1973c4100c53af5 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfff08218e74639942877756d483254 : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c2bf8047fd854fdf84f24370be51d4ed : SUCCESS in 31m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/711a1d86715244d090c9ea3271dab594 : SUCCESS in 31m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d82756b8786443b3a666770b8fba758b : SUCCESS in 1h 03m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3193e405f33f4971a7c9197539f1379d : FAILURE in 53m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6a7518dc8c14dcd91d559fd2e0438e9 : SUCCESS in 21m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/77b2b01dd5754c4ab672c1165514423b : SUCCESS in 30m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3917b8a6987e4791ae2530df210baf17 : SUCCESS in 1h 07m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/92cb0c30b51a433e8e1c8944579b716e : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"49f43025c6f48a28784dbbf24cad0a0d694ff921","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-28 18:41:45.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a56ee278e803a44b876d6a5a0181b030a8b3ac07","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-28 19:38:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"281914db87c91d10b419737d6d44bdb47363da0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 20:01:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f06e585219594acfa4f31cfff0b00782\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a8ca6eaaab1b485991b2e0531128954a : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/29cd9acbb76140a9b5323a48cc3fd59a : SUCCESS in 1h 01m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f578e7644de64749909da002e7014d74 : SUCCESS in 17m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f63ac55248bb42e292e46391592de581 : SUCCESS in 29m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dfed5e15a5d749a1b4521fb310d309fc : FAILURE in 7m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7bf476d6a0b44221a5c351a43db63b77 : FAILURE in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c5d88b5bc246e4b6e96410cb95bd4a : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406aaff1dd3740aeafa0c8828a17d72c : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df0f68100305494f8857c91eae1b13fd : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38cad385fed847c09bd7e0aadc97c4d1 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1da12fed9e4c4190f1e96fba94ba60 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee41e7c11004380b5d0b4701004cf7f : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7b1034a676f74e15830f395305adf1eb : SUCCESS in 24m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f8673bd428b14907b72a23d13f39f7cb : SUCCESS in 32m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/142e62c532e6470fa3c38384b68c00cb : SUCCESS in 1h 04m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a082d3f19f5e45a28fb91ef6073cc721 : SUCCESS in 1h 18m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/60982e9a620045a7a7d4c84846159b31 : SUCCESS in 22m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b23133814a04d1d98eae817ce4559f9 : SUCCESS in 27m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0fb97f1ce46a4508afacddf581682f4c : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c34822048f9f414d8d2dac9efb5fb220 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f9e697b0843ba620ae12196d1942df2c6e84f04a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 00:06:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb02446cc214cc5bd983b9dd637bb88\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e5192cb9e5764aa89296b16a52cdc3d7 : SUCCESS in 2h 02m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9fc16864b6414b8194db0f1ff2f6d4c4 : SUCCESS in 2h 21m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/108801c02bbf497c8b81c5a5f03bb77c : TIMED_OUT in 3h 02m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38193ee28a384e99adec56534ae6e390 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f055208ef19b4b859b62925ecc10738c : SUCCESS in 6m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a7414b83b5fb4eeea5999c0891721a10 : SUCCESS in 2h 30m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9280766d0aca4bf39f276dcbb0fd969e : TIMED_OUT in 3h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab4aaeb1d9c8834147a7f1525578590ddf501dd2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 07:22:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1140e330ebedf066807140131c909712cfcf3f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 07:27:23.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"377a2a34cec74e6f5797eb098d98bc1fdd8afd02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 07:59:02.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936667 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"f231e18c1879f843d3e586f652d2f61b59494b38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 07:59:31.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936667 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e131d3633888ae09524d76fed858015b153d8626","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 08:01:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8057a6a9d77409c0e804a5310a6fd6aa2fc66dcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:08:25.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936667 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ac49a963ac77c407959fe8a01a04f834a67e5e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:08:28.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936667 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7f1239f3a8bd7fcde04bdc07697852700d46273d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 08:09:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"326e5ca28899f14a26ebab1fc81d4bebab4c2313","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-29 08:37:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"af8f78f27764faed758e6a10f9d6e84112337fd2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-29 09:41:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a5bbbd27cfddcebb5c88963503f0a32eaeda429","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-29 09:41:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efba4110c203b6a596a396ba4e9494d166f2bd15","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 09:48:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d2ad5162671f0551a576d4b2d5f7bcfc2b20c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 09:51:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d496c548a35e4816aaa1c47ad6098336\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ea65437abc234bb08f52934c2cf8b711 : SUCCESS in 22m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9b8f9248ee124b72acad3f7f6e4361c7 : SUCCESS in 56m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/312e1d62ac784102ba033212d03f8871 : SUCCESS in 23m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6aba9f5908fb4cf4805a86bab902acbc : SUCCESS in 29m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b1b77feaafd54f61a0dbcefd0aa90462 : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/99cd40618c814b9195487c8c747e9a36 : SUCCESS in 1h 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d679eea8f04412a2adcda687aa43e7 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de2d18e964764a5b81e5f4aa7ced1ca3 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b116afee274c49278761293324aec957 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2cc27cd8d98431b97f1376175383211 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cbcb5d9349e4cb8af6b90886c811b98 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f4d749e5624c59976faa9f81c38241 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/00bb61efb21a48abbfef608d074ae122 : SUCCESS in 27m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1931f90052814360a4abd789498f3a59 : SUCCESS in 39m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/718d848e0e2e4197b92b5e6f5ba10723 : SUCCESS in 1h 04m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a9d452b399be4682b5b4430f4c318ad0 : SUCCESS in 1h 41m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce14115442f84a4ea35f5fee9b1609a6 : SUCCESS in 24m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/98547c77217c41eaa0ad8b3ed10ea112 : SUCCESS in 33m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a48ab5e5667e45588c0f80359395b74c : SUCCESS in 1h 14m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b0602a68c63f43b18a8e338026027507 : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8203f436524dd91f9a5fde3ebb2f17fe973ae4d6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 09:51:51.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0b9162c1aafd58cdf2711f93bee6ac14e3f1120","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-29 10:37:02.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"bccb2bdaf5aed92f55c0ae0943cdc4a9ed227d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 11:20:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e0da7bf1858957cf2f277203f07284afd2e6ca98","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 12:48:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c590fa05fee0edcccea3a4a133d6bf8ee75ee0a4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 12:49:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ba4b31dcdedfad36a34914346105bc4d9e26dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 12:57:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00516f3530a64485a8b91df4da3f1be1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cbdda73d6bf14683980eb0414ff3119d : SUCCESS in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ec9d2dd7d5b345ed9d7ab909b0d7ea92 : SUCCESS in 26m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ab51b1e7aff47c7a82961ef4861bf31 : SUCCESS in 51m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/08540cca3d1940a6b7676284f8e73b27 : SUCCESS in 1h 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1c195f229f4bb7854ba923bb539520 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74afece4c5274e07b33b5d82cf9bccc6 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf55980b91a745ffb03c6825ab42be02 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb01ebdb1d249d4bc332027d3f117d0 : SUCCESS in 5m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1f59dd7e95034b3c9e4518063cd7c556 : SUCCESS in 28m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ea552d6a5bea4b06958c673659231a51 : SUCCESS in 40m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/04bdbed18cae46eeb05a36a09ec5de77 : SUCCESS in 59m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67e070e4acbd48c68d5ac2cafa25f8fd : SUCCESS in 1h 26m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/900eb8f1212944dea47820c840f0a8e3 : SUCCESS in 27m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/15b3d8e9bf8b41e1a9da8d52e078ae52 : SUCCESS in 29m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d53df2102424169b4048906f4fa1e34 : SUCCESS in 1h 06m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/96c510ef58694c6c8187deb99500f26b : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"169ac2e812869e2148e78aa56d104875f9b8098f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 12:59:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dd2370661d42a425a086dda541fdd84c281a459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 13:30:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dff1602a6c4d859ef0bbebc3409196\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4c0ef2f302ed4411bf85ac57c53450e7 : TIMED_OUT in 3h 01m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42bbd2b7becf460ca55efbf0be7c3b9e : SUCCESS in 1h 49m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d2e2a2948664a39b8cb62b885db4328 : SUCCESS in 2h 59m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfe0d9a7f42b48189e18ecd8e94ea081 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f01bc5d002a409a9250a07571f280f7 : SUCCESS in 4m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/087ce98f35414b5ebf334bbaeb6d2634 : SUCCESS in 2h 20m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d2ec81d40e1a43e2893fc6c86114da30 : TIMED_OUT in 3h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"175ddc5743ecdef6e194b4910140b955ad13fdc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:29:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/863745d28b1b41a7914d11dde7a9bc77\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/866c2f85babe44d1b772b30c8436b097 : SUCCESS in 26m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/74dc212cacf24e7889071076537aebc3 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2089d073fca7404da79f223e66375388 : SUCCESS in 20m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42bc7840d2584f97af90612abff40077 : SUCCESS in 28m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/630cef656b2c49f0ba9f0282dfa717ca : SUCCESS in 1h 00m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/164a0b9399dc4b9ea9d235234d527d72 : SUCCESS in 1h 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2cb50a20dba4cccab0906a24cab8c89 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75144b25f1fe41c9b81409e45d64f553 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1f8202d9c754ecda3dbd63b38d687a1 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8e0d16e2bec4873adc4a217ef331e0e : SUCCESS in 2m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/073a5276c16d4367aa12f7c9523a6f32 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469ff149e7a648a9b44f7db8a7435b50 : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b6631d7d1eb4da8a3e4291e7232ff9e : SUCCESS in 27m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/284ecfb5490d413c9ff23649d39c3615 : SUCCESS in 36m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/612349535a8145c187c2c9d3aaa76934 : SUCCESS in 1h 05m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3df7e5c0009c42b2bde0a9ce0b0645ae : FAILURE in 9m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c5ff303d0280494d9ac929ffb95d5590 : SUCCESS in 29m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4186f69a64e640e7864ab95ed11ac83a : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/465f7efcff784698950e9a7b9f62b669 : SUCCESS in 1h 04m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3569e3d776ea4fc3a097dcf22c66440d : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":2},{"id":"152729e02e64cf1bcb94ec2d4332c9523929378a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-29 15:55:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f066b1528bc57cae5cd7bc47a1d4a7eb51634d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:31:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5586b821e0ea42ada56ef26fc61384ab\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4f83ebf7928147208ddc1f4e30a60ca6 : SUCCESS in 25m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cdd97b889f384212ba0feb7daa47b09e : SUCCESS in 1h 08m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/489d3908a55448699787683ab7880c1f : SUCCESS in 22m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c3194577bd9d49f793dc1907688a3849 : SUCCESS in 28m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/db44a19a8a0c42278384d08cbe7c42b0 : SUCCESS in 59m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4355c9c44fc247d598d32c1f0227c811 : SUCCESS in 1h 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30121772ef040f4b4e35705b3d200ef : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/755d6b10a98c49b797fef573e6276d2a : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/220bfb63895d4d30beab38f49e00390b : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f6c1030bfd742a3ab1e1f43a9c807f0 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80e9b47e591d4574861a9f23019a2ad5 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812c30d1da6648bd810bb08b1d3989f3 : SUCCESS in 7m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b1a59250631c43cdb247d6d831399fe7 : SUCCESS in 31m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/618fc08a066045d8ac32168948c8439c : SUCCESS in 30m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c3229947a9454db9972453d29e8b6325 : SUCCESS in 1h 12m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7dba14d4be84454bb543aca1635eeda3 : SUCCESS in 1h 33m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9db76ae4d2284da59665a913bbaaa763 : SUCCESS in 23m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f952f22af2f944b78b26c9eabf3a3d7c : SUCCESS in 34m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c02e98b0e6084b8f9dbd6881b918d1f1 : SUCCESS in 1h 00m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d7b575477c3c4634916bf857b61d5a70 : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d7cd2b0c67c5a8c7d381f3d828ff1ae652517c86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:31:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3942cb419243d1effc9068acdc7479486c2f9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 18:58:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da676f941f8c4116b21f497e5396ee95\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1717ab310519446fbdd0892119dd718c : SUCCESS in 19m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/51606548601d4b13abca13dcefe2fe48 : SUCCESS in 25m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c9dac249784a4460b55a5dbffffce785 : SUCCESS in 1h 12m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1bfce802abe14ad9b0f4f19d6e01f191 : RETRY_LIMIT in 1m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bdddfa681149198ed2bd4dd49e0e77 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/651fadb6c11c4c45ad2a29120e416618 : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a2da884177448c68dc3bfda42161aaf : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6b2993f6054b76a9fa6040a3a454ce : SUCCESS in 4m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/94f5776ccb404745a7fab11efd3e7624 : SUCCESS in 23m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/45a896f30341471396b14217d300302c : SUCCESS in 32m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e0f369734f5e4af9a9ef43a2d5921320 : RETRY_LIMIT in 2m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/06b0c102b095421e99bd4c89c6c0f399 : SUCCESS in 1h 16m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c0b34597f0c414ea39c95a678730c71 : SUCCESS in 22m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/51dda9fefede4dd2be70e83cc8ab453f : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8340d4ee2db34f3d994df1436d95b35b : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c6ed4af10321456ab6f2876814b07283 : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":2},{"id":"249d54e920e58d8752f30481696e520c050fc412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 23:00:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec826b786e5642229704feb3089f6cb4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6776efab391946aa96f0baf1cfd7d157 : TIMED_OUT in 3h 01m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4aae9ca624cb4df0a1bced737a15de5e : SUCCESS in 2h 17m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a413a8eb51814d2fa983e51e639705c8 : TIMED_OUT in 3h 02m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfc57daede854f93a977c7e23c2cc179 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99b5c697974940fd9dfa22e5d4aef078 : SUCCESS in 5m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ba7d3bed2b34420889eae1845a48d6c2 : SUCCESS in 2h 20m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8ddcdbe8bad243f5861f64ffedc9534a : FAILURE in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4960cec7e58232a17dd9946ceadc7d94c3a413e1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-30 09:00:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4fa82decd334c739dda3783732ae480549613f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 10:32:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b630871a2f34eafbfffa104fa33f16d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/850049878a0c4c708dcde7c02f2d46cf : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/898c59e213e64815934e5628115a29c8 : SUCCESS in 59m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf4f981edde94884b6a2da5caab62278 : SUCCESS in 14m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8adbabd91fce46dba1ba86350a8e3a24 : SUCCESS in 34m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/24f6b78254744745942a67b04006d573 : SUCCESS in 48m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0303493156134aa89287ba9834f167bf : SUCCESS in 1h 26m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0c78ef6fe74dcc84d195495e72e6ed : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f201f8e62a87466db64391f9453cd240 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2eaba0c76bea48c59d4be3e5530c9e5a : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3e7f00648ab410a9727a484ceae1628 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95248075425f4884af24113c51706068 : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/506cf39c3331477a942a7b3fe81774e0 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/58a09792798947c7bcb6449038d2513d : SUCCESS in 23m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b10f294e5800490ca1d3f82192913a5b : SUCCESS in 34m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d4d16479cfce4995aa296d6ed4f992e9 : SUCCESS in 51m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8bb30162873448208adca74e2752a7a2 : SUCCESS in 1h 22m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cf751172b1e646c0917cc2edad39fe96 : SUCCESS in 28m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f1cfa1c1ce7542259c37757c2bde1747 : SUCCESS in 32m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9611ae776e34456280dd0c4d5191a2cf : SUCCESS in 1h 04m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/26a68cd31e6c4299a080b26e9f74557b : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":2},{"id":"dc6f730abb379d0658b9c7a5654566748b860353","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 10:33:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa174e9de481b0dbd9e66a6289bcc4b03bf78a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 12:05:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8856ce7826dd4fdb8fab5b9133a06d2f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/339fe886d4884550879d8b29bcb58c14 : SUCCESS in 2h 36m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c18442d6c4b548018b4bf02091771121 : SUCCESS in 2h 14m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3e628e16572942cd9c2e215437c4e811 : SUCCESS in 2h 28m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd591bd722314d608298d104b7917b31 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe356f0fca994cf29d142d113ccdb69f : SUCCESS in 6m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8d80f84fa3d646cd80fa335547d4d36a : SUCCESS in 1h 51m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d1c445da23114f75a124a375dc276cde : TIMED_OUT in 3h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd02d064aade2f01623bf7c2a0122927b494a322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 12:12:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/723ca2811c884e31a834d9131654a1d1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c7e4f06ba944dbcb87e1b4dfcca5a56 : SUCCESS in 13m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/488fedd9f8f248cf850f20c8ceaa57f1 : SUCCESS in 20m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c3f8a3b9c87f472eae6a14803ca3a240 : SUCCESS in 1h 06m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0667ead7210046b3a876143dd25c9d3e : SUCCESS in 1h 05m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0473b2dd41dc4c13bd1d8df87fda1d4a : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e63f41b32e434ef2972c3543588f7d04 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6feb150d612e4e60acacbf09af9fc830 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef370616eb4b4e2d91c43186aff65dc0 : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27aa58ea57614afa8a962d17a3e2e2d8 : SUCCESS in 23m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/266bcded5c5943e4878e3f4eeec8a5f0 : SUCCESS in 35m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fa1e47356fb2431b99a21371ffbd237f : SUCCESS in 1h 04m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/82b411175c1446c2bfac938d0cfaaa4b : SUCCESS in 1h 34m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b200f62a93c441acb0e7887ce8a26642 : SUCCESS in 27m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a99db861ec3849dfacf2f01e17636407 : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ce6a36f620fe4ce49109ae190e03fc29 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86f6a1c2851e42d892137252957e996d : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":2},{"id":"10f83adc4e7edc9ab3bfb760a574b02fe123c88f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 12:12:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4134aea51504b4feab0671a6c7a1140cebf9cc5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 12:13:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b90a120f28940fc98a178239fa3cc43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed44d795c494a5f97e44420ac1add21 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"890568ba95e04a2414ec62547270238a596874e9","revisions":{"6e11e69ebc06becf7e3a9a36144e49f189a93a89":{"kind":"REWORK","_number":1,"created":"2024-11-28 16:07:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/936641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/936641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/936641/1"}}},"commit":{"parents":[{"commit":"e49ba16fcc930ac54a610f7007ac40d5580ce9fc","subject":"Fix handling configs in base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e49ba16fcc930ac54a610f7007ac40d5580ce9fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 09:29:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 16:07:44.000000000","tz":0},"subject":"CI: Use debian/ubuntu images from quay mirror","message":"CI: Use debian/ubuntu images from quay mirror\n\nChange-Id: I3fab494e939478458c2c299303f5c2c91943e0ec\n(cherry picked from commit 7ba43d2820cd91b1108e5473eb0a9c26ae4a9cb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e11e69ebc06becf7e3a9a36144e49f189a93a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e11e69ebc06becf7e3a9a36144e49f189a93a89"}]},"branch":"refs/heads/stable/2024.2"},"890568ba95e04a2414ec62547270238a596874e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-29 07:27:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/936641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/936641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/936641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/936641/2"}}},"commit":{"parents":[{"commit":"e49ba16fcc930ac54a610f7007ac40d5580ce9fc","subject":"Fix handling configs in base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e49ba16fcc930ac54a610f7007ac40d5580ce9fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-28 09:29:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-29 07:27:23.000000000","tz":0},"subject":"CI: Use debian/ubuntu images from quay mirror","message":"CI: Use debian/ubuntu images from quay mirror\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936667\n\nChange-Id: I3fab494e939478458c2c299303f5c2c91943e0ec\n(cherry picked from commit 7ba43d2820cd91b1108e5473eb0a9c26ae4a9cb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/890568ba95e04a2414ec62547270238a596874e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/890568ba95e04a2414ec62547270238a596874e9"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
