)]}'
{"id":"openstack%2Fkolla~936755","triplet_id":"openstack%2Fkolla~master~I174c5afbd5517439c544f44cd56334fc85cf5731","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-10 20:23:06.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-10 20:23:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I174c5afbd5517439c544f44cd56334fc85cf5731","subject":"pycadf: bump to 4.0.0","status":"ABANDONED","created":"2024-11-29 14:52:59.000000000","updated":"2024-12-10 20:23:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98cee62d1d87ab058b7c7cf9bccf5482a61f3e2b","_number":936755,"virtual_id_number":936755,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-29 15:56:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 15:56:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-30 17:34:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"8ee5afac1432266cbe5d4a352fb787ec4e825353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-29 14:52:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc99d0f955ac39bfa0dd3211e552cbfbd7b9cba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:56:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d18cb527bfd34bb199381f7a4fbf1357\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7ec1e66d256a49518462f18724b76692 : FAILURE in 12m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f0350c5273c4422a966c52781726e624 : FAILURE in 18m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ddf7d8b0f43641868f5e525004fbfd13 : FAILURE in 10m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/16a552244e6f4367bc49bc702f2fbf7f : FAILURE in 18m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9902e18fe3344d2ba2614d853438a29 : FAILURE in 13m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/13f505885eb141ae83c65f5a528f8d5c : FAILURE in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6c16abd4914031a0493a34fbf146bc : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dbe35f799cb45ca8ba19db529dd255f : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91113d2e20d64dbbafb9b668135b6ec2 : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f35419bab00476d9fb33a92ff5c4cc4 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2874d773328d4ed6939a2d6595f95607 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/69aa3e45c64e4e1abb62eaa5e6405c06 : FAILURE in 19m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/eb19344c148340a7b1ab9adcf37dc1cb : FAILURE in 26m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/558f0a907585480eae1153d6442a5737 : FAILURE in 17m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f2bfb90648464b308c37db8e4c85d419 : FAILURE in 16m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/582b1350af4a49ccaa6468fd1490ade5 : FAILURE in 38m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f11dd3bf2fe6434497144203d5deb97d : FAILURE in 49m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/85b106c33ddb4dd28252a7a185872806 : FAILURE in 32m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/83ae06ade243447c9034bee9be70e891 : FAILURE in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"136d65317e0edbfe798d9baed28091ab94590627","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 18:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c433616e93494eb4aef0c73cfad0a44c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0156c1b5e80b4bf89bf5c393e6194f09 : FAILURE in 37m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f591ef2b328c4862937bc30e69824cc4 : FAILURE in 31m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3544db8435994eb1ace7ac3546a878cb : FAILURE in 1h 01m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf0d1f4d6f5a498aae349278d18fed86 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c9e5aae7a0f4ae8bd677e09bc8089a4 : SUCCESS in 5m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e2059bd5c4f34d08be8afd7bd8327be8 : FAILURE in 35m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4e526493c42347109b8860de38ad3652 : FAILURE in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcb52fc1c686fde4028fe20b748fe31ce27bab24","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 19:53:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c0d71672720564af46c760d3a94168f018a64f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 20:19:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56bc0e92b17f4223b834ab8f8236d00a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3767c5bae76b42be907861be97949449 : FAILURE in 15m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5dd8a689add9471384c6a1f5a52310f6 : FAILURE in 13m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fbcb9ca2eee94857b9ea106cd227fab1 : FAILURE in 12m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6ba547d561c34dfe8468c617e3a36eb4 : FAILURE in 18m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/17d288ba3eea46ab86dc68d73f2d0b2b : FAILURE in 8m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ae422a9ea2134351920ad8b088cb6965 : FAILURE in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef2c11998fe147ecaf61ba96ace9b027 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de401bd49045436a9cd9484e897949ca : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04d6076431b84978b54a8d0ca2d8b0e2 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40047c4322284820b5ee4274856c3268 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4290506e4304c429fd3ddd7e766da1d : SUCCESS in 3m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7c46656482c4bfa977f691284732c29 : FAILURE in 17m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a88cb8535ad1432686c9d45be121a673 : FAILURE in 18m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/11c0a9a72c54456abcfa47fe4fdc5f1f : FAILURE in 15m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2c43f1f6279247e3870f876d280ef56c : FAILURE in 12m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0893914b39374cb7815095cc8f5f3656 : FAILURE in 18m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8aba7415e58d417ca1ed776b03d63f10 : FAILURE in 19m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d962b86e3e484da08256b54c484d94eb : FAILURE in 14m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bb80bf47d67b4335844de2bd2b5150b0 : FAILURE in 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"98cee62d1d87ab058b7c7cf9bccf5482a61f3e2b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 20:23:06.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a20e4a3e9da65a92e170ee26179340e7d4d45de","revisions":{"8a20e4a3e9da65a92e170ee26179340e7d4d45de":{"kind":"REWORK","_number":1,"created":"2024-11-29 14:52:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/936755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/936755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/936755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/936755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/936755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/936755/1"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-29 14:52:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-29 14:52:58.000000000","tz":60},"subject":"pycadf: bump to 4.0.0","message":"pycadf: bump to 4.0.0\n\nChange-Id: I174c5afbd5517439c544f44cd56334fc85cf5731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a20e4a3e9da65a92e170ee26179340e7d4d45de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a20e4a3e9da65a92e170ee26179340e7d4d45de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
