)]}'
{"id":"openstack%2Fkolla~936834","triplet_id":"openstack%2Fkolla~master~I56cf5a869e854f760874d207802173c89b04c0df","project":"openstack/kolla","branch":"master","topic":"python312","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-10 19:33:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56cf5a869e854f760874d207802173c89b04c0df","subject":"Add Python 3.12 classifier","status":"MERGED","created":"2024-12-02 13:29:24.000000000","updated":"2024-12-10 19:35:02.000000000","submitted":"2024-12-10 19:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936834-python312","meta_rev_id":"adb706f7c43f072a900b7b3e2fb729b0e49dbe5c","_number":936834,"virtual_id_number":936834,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 19:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2024-12-06 12:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2024-12-10 17:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-03 05:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2024-12-10 17:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2024-12-10 17:40:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 15:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 05:26:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-06 12:06:17.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-12-10 17:39:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cf05f7232613c22286f4872e7b874a44e02c7924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-02 13:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c64db459476267b47528826e3a1f7a762fffd10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 15:10:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2562ad9cb544477caf6b7c19955517ce\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c80a7bd738f4a2d872d1212d12f0e6b : SUCCESS in 1m 53s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/07f313c3132c4988b4c77555a94768cc : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9256e13698b34de0b842676ffadaaec4 : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a88be79db6594787ae2db20093f7b3c4 : SUCCESS in 20m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c0d41494ef84c789e616e3907b769bc : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8630df3684249ae9ceed3e5d45f8d74 : SUCCESS in 1h 12m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d9b6e7b5010b45258f842bf9963b5f9d : SUCCESS in 1h 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646eb4c12e3c401888d151171844e426 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa9bb11b4aba4be895945c68be2665c7 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3f2c4c46a5742ae9363cb08c7b2080a : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a059e84e78647d29501f6211f9e90bc : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b0555b3b384744aca567597e3ffd6f : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ce508eed029a4857af57166e0a7f17d2 : SUCCESS in 27m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6e735caf7f7d43f3930647bf5c6c66e7 : SUCCESS in 30m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f6b36b8f3415400a9106083b801ce9f8 : SUCCESS in 59m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/38bdb0239cbb4edfbf7cac0e6f0fa6a5 : SUCCESS in 1h 19m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/512fa2555d8644399bc837eccc930873 : SUCCESS in 28m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b3e373c0fb394eaeb7dac97d21807f23 : SUCCESS in 38m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9dc01c7a18ac4ee695e0ad9d05043955 : SUCCESS in 1h 07m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/53aaa32b25c1459782b88178d6d7f2e3 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5780d34e365f5cf401c5a7cd5c94be7ee529cb79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 16:47:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2cc8355fb94055962b906e77806350\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3a2e7dc628f3413b90e542d5bd1086cd : SUCCESS in 2h 09m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f01d8f4ad9b4d75ba4beb835aa71b0a : SUCCESS in 2h 35m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac166fa2788c4a67868399a92f2c788d : SUCCESS in 1h 58m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47648b9ff0624232928313dbc0596408 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7af05722f114d4dafb449556cb78623 : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5595b22115fd4aabaf4a1d10aff7729c : TIMED_OUT in 3h 16m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4e0a490927264dc88f0181ee4035ffae : FAILURE in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16517f71babe40910adeb8eb0760d50c7a4ad409","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-03 05:26:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d8bc841e5265cfd9583e0eea1d8bcc099878607","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-12-06 12:06:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c81c9c229230d089660fc6d07e24bc27c09df0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 17:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4396745db78b562f99419fc76ec7a690608ae27e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 17:40:50.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"a91e952e0b79f126885ed75befe0e81e15fc4f8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:41:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7ac42550a85fca7d00e4d1769e41e2abcf5e4a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:33:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad64e5ab0c944a89327825d1f2b6353\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3553119e124d4ff4bda81b8fa05d155b : SUCCESS in 1m 44s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/20b05b2e94eb47a999d17b600d81386d : SUCCESS in 15m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9ecfa76303dd42339e06f32d5d0f8f79 : SUCCESS in 27m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/61e1c3db3ffe4236aadfe8ca7054f7e1 : SUCCESS in 1h 09m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/70a2866977734588885f9c7a6266b5c6 : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aaf59f97be4418ab5513ea455339af9 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2fca3f633184447a4a04064c1c95ead : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/607cdfae910e4939bcce75249e4f1eb4 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e67941e9154497fa822ed3545d1ceb1 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ac2e6869ae747ad8fec7df2f8202fc4 : SUCCESS in 31m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/649c119285e34ce49d77f9344c012cb3 : SUCCESS in 36m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/380149b6c7c245298202a3bd9a540f49 : SUCCESS in 43m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b02127410481492891e4b9b124d02d08 : SUCCESS in 1h 30m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9fbaa49d4eb64b6793c6a45f807fae0f : SUCCESS in 24m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/491f7570c0f54eb6bacccde755c9a1a5 : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b6583c34813845489447a6ac9b33e271 : SUCCESS in 57m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3246906d0b144b06bfe572af1004f1b8 : SUCCESS in 1h 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"386933857d70fcb346370ee2a945513192252544","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:33:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adb706f7c43f072a900b7b3e2fb729b0e49dbe5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/253fee1940c14aad947b452fc88b922d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f6e5b1c2884de49f46bde6e1b0bc8e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45ddb71947e52a6c18a3c0af73adcbf1dfe3da0e","revisions":{"45ddb71947e52a6c18a3c0af73adcbf1dfe3da0e":{"kind":"REWORK","_number":1,"created":"2024-12-02 13:29:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/936834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/936834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/936834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/936834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/936834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/936834/1"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-02 13:28:58.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-02 13:28:58.000000000","tz":60},"subject":"Add Python 3.12 classifier","message":"Add Python 3.12 classifier\n\nChange-Id: I56cf5a869e854f760874d207802173c89b04c0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ddb71947e52a6c18a3c0af73adcbf1dfe3da0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ddb71947e52a6c18a3c0af73adcbf1dfe3da0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
