)]}'
{"id":"openstack%2Fkolla~936919","triplet_id":"openstack%2Fkolla~master~I15f9b5cf3976f4298cdc7da4530f2d7d09614e88","project":"openstack/kolla","branch":"master","topic":"letsencrypt-rework","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-09 13:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-01-10 13:37:34.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I15f9b5cf3976f4298cdc7da4530f2d7d09614e88","subject":"[letsencrypt] Rename script for updating certificates","status":"MERGED","created":"2024-12-03 08:50:28.000000000","updated":"2025-01-10 15:28:02.000000000","submitted":"2025-01-10 15:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936919-letsencrypt-rework","meta_rev_id":"df0dceb0bf888c4e44afc3592ebe3f8ec40acb4d","_number":936919,"virtual_id_number":936919,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 15:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-18 13:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-18 13:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-01-10 13:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-10 13:37:34.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-12-18 13:14:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-03 09:08:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-03 10:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-18 13:14:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b4265d929f4384ba822164a3e844b041bf8600d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-03 08:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ca4498416502f72f9e84ed574c4bb9e645d5b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c1692221334a7086891062a392ee52\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1bb14159b64943d9ac196d5d16d9d2fd : SUCCESS in 23m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/07c7bdcb78bc4e0882e55add8a08419f : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3be11f0292ec4f4188afd5827b46534d : SUCCESS in 21m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f024f23cdcb049f2bc0e649d3f66b20b : SUCCESS in 29m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4a454b9e672b41bbb5e5777ab2f0e1c4 : SUCCESS in 1h 10m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/914f8a9dd1024b38b211d64a3b591165 : SUCCESS in 1h 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d541c8e4ae4554b8bf8a5798d0c753 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/344f6067e6694257a45142e01caa2594 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23cef2c45ba04089a25c951dea725068 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1205690161d4c5cad532f8b22d5c1bd : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa097e5af5b4fe39040c3f0c286de53 : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3659adc410a9477dad6a720e350e7f8f : SUCCESS in 33m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b40209fe1984b8bac573f3754d89f7a : SUCCESS in 33m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8999b726a8714901b47ba8b75d401bb7 : SUCCESS in 1h 05m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0bb891615ea644039ca6152fc1fee1e4 : SUCCESS in 1h 26m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fa30a929b2394deead6100c282de5f2f : SUCCESS in 30m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4c15bb0a0b744476a88b124c7ed6cd1d : SUCCESS in 27m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d74d86be73734945b9bc45d85c6ad6f5 : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/58be36a216e547e991749d64cbf61b0f : SUCCESS in 1h 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e69c7d59e00069c677a6655beaf6fb0f405865f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 12:26:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a174ace398445ba095dd0769cdcf6a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/25dc942be79240d290c71d2c514166ab : SUCCESS in 2h 52m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/19aa8c6b238a42f583a56ce838e16da8 : SUCCESS in 1h 47m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ef4e3f658bf44431bef0cbf30486c2df : SUCCESS in 1h 55m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/314585aa727a40ceb6d248bdc540e901 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0905546af077464d9db8bfb2eec021d4 : SUCCESS in 4m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/394bca79cd4d4b538215e74fb35ba541 : TIMED_OUT in 3h 02m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e1ca86b27dec4d2f94b871ddf03afe4d : TIMED_OUT in 3h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5eaf965428cc19a04d49aff0b2cf83962fbfd90d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-04 08:41:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"049d7f0bda4666dcae8b68e73303f8f0e59871d5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-04 08:42:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22d71357b138cd542ad20b841f985ea173c90ed0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-04 12:29:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c64071d1c359c7359805f8b7fedefffdaeb5e922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-06 15:18:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"55ef064e111d18f827ca5aa938f649977a745751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 17:12:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e808b0e25ed141beb81fd049d6990637\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ce7d1ae57dbf452bb68300143e435f01 : SUCCESS in 21m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3f0f2f8d8bfd42928f8cb71b3408dd39 : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5044c70749ae484f99f446a7d728d5c6 : SUCCESS in 18m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4236e920e00f4cfd9150df1a75ce8961 : SUCCESS in 31m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6b2b336a637a475fa8c9319bc96800ed : SUCCESS in 1h 31m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/51991a5af7654fb18c697a8d0978b57c : SUCCESS in 1h 48m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a71f5c1e0344fe84713bf125656336 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3ce894b830444db2f1451cb807e789 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51f3fb6ae18645bd82fc5031f2d49172 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46afe0a7f27d435cba7584c46266df89 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300643db7a06447eb91b572a5d315c79 : SUCCESS in 2m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e8f2c8ec168241c9904011aea8573e42 : SUCCESS in 45m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d39dc4699e1449d2865dacc85d9ef629 : SUCCESS in 40m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/89f9e268a21e46668c83d4303da69662 : SUCCESS in 1h 01m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b3dc2cd5675c4e2a8bc2b0c8b8c1baab : SUCCESS in 1h 34m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e6f74c6bc0014eac99bded90d48b619e : SUCCESS in 26m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e0ffe61768ed4c2f8ff119add3237527 : SUCCESS in 25m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20da3acf49ee4368b7ad87deb0ba1b82 : SUCCESS in 1h 08m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b257ac4013fa4d8a89246e8dfa31d7a4 : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6e23037d467f8d0721b23eaba8d161b0d2e9708a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 18:27:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/849f227cc1ba447798d0086f3bb6101f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad2928ee23814f268e49b78e08da0d2e : TIMED_OUT in 3h 01m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/78a021bcdb8844b8b73b1a6627b1893f : SUCCESS in 1h 54m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30d3e92d550349e3bd1ebbcb8805f99d : SUCCESS in 2h 09m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3696f72d94584447a9db9f96f23243cd : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b95f4f2c0874d14a3b408e58c1e7602 : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/857d65c191984ad0962037b165406485 : SUCCESS in 2h 01m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a367936f6d274f39926ad8b2ef5300ec : TIMED_OUT in 3h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9b2ebf00e35cb58b6eea88324c188207d736f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-06 22:48:24.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2de695361a41f02f9532909ea34da4eab7ef874a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:23:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60257ae6840d4b45872e9a78d61f351a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6146d02b22754d908755c0c6a91d0cab : SUCCESS in 25m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/af9b054fc46f4edc80fa7b37e8a3d5c5 : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/51acd2e5c43c47db941304cb6cbdbbd8 : SUCCESS in 20m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/932a9e61b23e47e38fb1fb93d39e72af : SUCCESS in 34m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8384c25c2ae24c9296a18764b55d840d : SUCCESS in 58m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bbd0feb28ecc463fba2b3b29df45a00f : SUCCESS in 1h 23m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4dd36e90f24f81a9fec38d457d5bb1 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38cb2f2c6f644efaa8ec18f7a614ccaf : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c33423af0cb046268a721f7d5be5a1be : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8132c6e6f55493a83f368a30b234345 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905c4a02664c4d6a94a319071cefb03f : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/01db15fecb3d4ea382358067f53bf598 : SUCCESS in 33m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2d229a4639744da8a0aec70749b1e2ec : SUCCESS in 35m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/885a6e0bb8bf4a6db60d43ee8b25a9e9 : SUCCESS in 57m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/82a6ab6fd84b4912b0aed5900f806e6f : SUCCESS in 1h 31m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c9dd0b8946e04e2cb3c369391a3bd482 : SUCCESS in 28m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/98a121b6caa549bf9d2bebbc32ec9155 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c68dccc56cd44a51b9b978e297d2b2bd : SUCCESS in 1h 10m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9dc428bf72864642b62bbac3900d188a : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":3},{"id":"7d50b8c1887be1947bb812d23b0e545a6b48b5e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 01:56:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b931481fdf34b4f91fe6422a3f9134b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/adf6f0efc0014393bdd0e5a9d8e4aac0 : TIMED_OUT in 3h 01m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4958f52488fa45f6902b7179b0f45800 : SUCCESS in 2h 20m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce724a4cea564ae19a7619d84a703849 : SUCCESS in 2h 02m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b86305fbbaa2469290cb184dd399d607 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/887d6242ce944f3db1547dda93af1cd4 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fc3c7bf1d6514c38bee44e30f4462d44 : SUCCESS in 1h 59m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f155b0652da6493788ea73898da283c4 : TIMED_OUT in 3h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ce912879103d43b1f052c744ab960126aae3f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-07 07:28:02.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5b6a2be0b8c174720231039840e87f365da27e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 08:10:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89b58dc9805345f7b1c05c0e5d2f3d96\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6b081e981f1d44788ce508da06e3daac : SUCCESS in 21m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a7b2d4e3715645c6bcf4974d08d08535 : FAILURE in 3m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/319a339ea1d3401bac3ce01ddf102579 : SUCCESS in 17m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/29106d5416db4fc7bebe6ac36654406f : SUCCESS in 25m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/84c9d8a4be534a2389235f93e84a8d18 : FAILURE in 3m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0f8383d337934eb2a32e50c4cc1edf1b : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f036be571994d439a5c285f96c4d3de : SUCCESS in 1m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d0ab86eceb34e3ea5eb4349704544b3 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a6874aecfc64fa79734e14d0917ce3a : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99b3b9027b7b4240b4044113cf968e52 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9c98d3b89848d4a6baa56a4c70c06a : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4d2fa33ced2d4ba0a5d152d7e1319161 : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7815fc8daa674894ae26b8a3a3cad269 : SUCCESS in 28m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/23d06d7d14cf4d5b8634222d8cee9829 : FAILURE in 4m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4b0a578434504a7a9ba6dfff987a7d10 : FAILURE in 4m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5788dc99eba3480d85f57217084d84c1 : SUCCESS in 23m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/80e219b1250540c4bf602429096eac42 : SUCCESS in 34m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d012c57cd13e44e2a497f44ca8c14bfe : FAILURE in 3m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c6ca428ee4bd4cabac7e8ca6c71df86b : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":4},{"id":"0b16f4e9f5d9c3a61cc193b13daa6b72b17ea8f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 10:38:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a98468301ca4506a7d9a4d6ef1ed7c1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0160879044b449bfa39fe52c6c532b85 : TIMED_OUT in 3h 01m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ef22d25a4804420d95dc48d4ea554adf : SUCCESS in 2h 28m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e966f6a569fc4192826252466947dffd : FAILURE in 4m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5865fa5a8e6943dc9bb3cd808cf9dc9e : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d95c2ff9d88141f9a29a1c13059ce02a : SUCCESS in 4m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ee52e6821f34460eacf39bc0b0afe317 : SUCCESS in 2h 38m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9c86426b4c5f48cfbc386a5620631a95 : TIMED_OUT in 3h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8aaa34426eac1e2508859af4e7d4e8ea2d2b8733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-07 19:59:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bc57eb7f0ad267bdbe426804200414adf673cd15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 20:52:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7af811fbdac24409b3c82cdccd670dc2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2367a54aab2f4fa68037c122ef0ec1d5 : SUCCESS in 19m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dbe02a5d998844f09af7384f2c2a5764 : FAILURE in 3m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c600e026db494937a02c8c700768ed6d : SUCCESS in 21m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2a2e04a7a33c46f0b2e990c395a83efe : SUCCESS in 26m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/214326b8ba1c4709a188c734fc5565ac : FAILURE in 2m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1ed750b81f85481f9bde8525e1210213 : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61cba2376e704304896bfe22e6d0bc02 : SUCCESS in 2m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c3c37dda76644b6a3f3dd8b32b19d79 : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/121bb9a5f2c342d4b7a559de23bb35bf : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a976f7dda034586b1c818a089889e2e : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a001190836db4553ab3441d8c0ebe8da : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/feda1bb02c684f05900a8c3007f8888d : SUCCESS in 25m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c68ee7a393514a42828dd77719f79360 : SUCCESS in 37m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5f35ab86a524cd3abfc4ba4a5a4f38e : FAILURE in 5m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a2b0ec25467647b3b61bcc8cebbd8fc6 : FAILURE in 4m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0f7cbda5b2c49eea195343e53d68ab8 : SUCCESS in 28m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f469dafcb8ad44e38a19285fb9fe33cf : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d0db8ec8942646738d8c87f1e08fbded : FAILURE in 3m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/54e09418d9244c238a88980bc4a024b1 : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ae0b927a5fa0aaceb37825d204bd46759555df4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 23:07:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5818d1330c46e6933f1b858ddf1e54\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9f27790d7e6442b6a3945567b29879b4 : SUCCESS in 1h 39m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d1306da06b64ee59cb6e244dff465c0 : SUCCESS in 1h 47m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f455d1ffe6846aaadb40ef10d81fc08 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ec96b0d64764073b8ccbe91b040bb1b : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/353b4cfde832477b91bce4a8929ab060 : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8fb2533b2ba440689e4661cef4425872 : TIMED_OUT in 3h 03m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d03f6e21671c4d36a4562d7e02334337 : TIMED_OUT in 3h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d87e08924172aa769a985b1d7c1e9cd0f029944e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-08 14:42:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8a5e953d1611be4a24139869434f058a79d99abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-08 16:21:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f16e9c056449ec9df0d8aa4261a53b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/48a6070019b842029fc0e45b5ae9989b : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/429340d17a984dc9829423c8b3e94ebd : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a0d8ee5b40b4bda8e2ffe9028c42ead : SUCCESS in 16m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d1843deafbab492db63076bc7b70aa7c : SUCCESS in 31m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2664f4736c314b01a88a1c57d97d41d2 : SUCCESS in 49m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/16ab34e5a249483496bc6cc5c4cde45e : SUCCESS in 1h 25m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fe6d6b79014baf8e558eb122cd02ca : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cb402a329cb4e9597bf1b69f14ecf56 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4417cff0fdda4808a46ca5e9965e4eff : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a8ecc76d50f4aaf928bc8eee07692f6 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e26642d64de44d8b82d65497f23e1f2 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/df47b62b481742c5bb4afbda76452792 : SUCCESS in 25m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57b75549e1084db7a03ed851ae1d02da : SUCCESS in 24m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d7e82b5434ae48c3ba22fb5ac7016423 : SUCCESS in 1h 22m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c04bd9706acb47de8a12e2b209173a25 : SUCCESS in 1h 28m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70436650243746d2a95ec6c92bf8840d : SUCCESS in 18m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/faf16b13e97f4494b4f43256c81d648e : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/574e6365d3cc4d0faaea5850ffef2100 : SUCCESS in 1h 08m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/376e77962c904548bc1a8c376ccf4893 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":5},{"id":"553d06423d512e14ad598aefe4e38da548bb4d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-08 17:28:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98984d16af504bedab8119d036c79763\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/210e68381b52481ba5349cd6c3d948cf : SUCCESS in 1h 45m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80bae89e21f54949935e280c87494410 : SUCCESS in 2h 26m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24521295668c4a67a71076f325ce7af1 : SUCCESS in 1h 48m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84c293ab2e3d43baa89e30a9bcf77510 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cd3b51ef23f4b8b8701ae65fa8a6680 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ca1cafb80c93474194f8b44e27b639ba : SUCCESS in 1h 36m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f53d10680f29460f8495fbfde6422702 : FAILURE in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fda4b95c11f4d94179ef6fbc20342cf3f4cd0d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-09 05:54:40.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"408e8342adf60e5f956ad044bea8fe45dfd05ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:53:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05b45fb7886b4c1ba085a3ee11c8d58b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/121181880f404d8d87f9ff676704cfe5 : SUCCESS in 32m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/87859b95a2b849a3a98c0f4c63978bbf : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e64a72b500b84b33bd92bac220d6f17a : SUCCESS in 20m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/470b718e07a341ccb8297e867fbf140d : SUCCESS in 32m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ac3771b551674e6bb1714bddfd516d96 : SUCCESS in 1h 00m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fae6654849a34d61a17517ba266886ef : SUCCESS in 1h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4763b0f6c0894701aa26299bb18025f5 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed4fbfc77fd4499b08e6fe70d8c5493 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19f83dc079004b32aa4ac7bd1bccbe6a : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6ad7985e2744179beef1bdd82ec2917 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3a03beb6f4f4072bce435c0244812c6 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8d75d975cec549109de41bf70c2bb008 : SUCCESS in 23m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/848129213c494c6a919ec9a0fc43d232 : SUCCESS in 25m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fad0b2af68e4479a937d19796cb85909 : SUCCESS in 1h 03m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/345be5d813cc44bcae02ae6a1a0d662d : SUCCESS in 1h 29m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8caad7bd2a6c4171a3e40380324dc951 : SUCCESS in 44m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9e8ff2111be746f9b9be52c74cf6f5e2 : SUCCESS in 23m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/74ae8dfc9152476c9c2fdc6fb02e5f7d : SUCCESS in 1h 07m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5089afafccb245588f70751f52b46887 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":6},{"id":"0dfc32dcdfb7e21f5ca9bb9bafd9a99bbb934bb4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-09 08:57:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"74744a1b8403af3357140a95cdc14debc0cb8ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:58:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a79e01f3e74e3baefa0cf36ebf6cbd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/af144813f83e4c9b85c9e5f8f2c39701 : SUCCESS in 1h 57m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d34cd0890d74fe1a35e0daf08b4e9e6 : SUCCESS in 2h 00m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/da0aa3a910cb4302896adc7e6b0b97dc : SUCCESS in 2h 37m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80401477b7ea4ebf9b777165e9264d11 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2075759af5484f3c92d4a90cb00fccce : SUCCESS in 6m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/471ff33a8fc44fe09dfc0c51c55b5c53 : TIMED_OUT in 3h 02m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/58b07ac4f7914617a72d20d29196d6cf : TIMED_OUT in 3h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cb3d61762908b3e74db03dbd9c4409bdad98b24b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-09 11:32:53.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b6f8195722d31d73ea24f97dc6e7efeb1f9cdcc2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-09 12:05:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"4950b906b6f20c8a7c2ea86bc5952158175c424b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-09 13:07:54.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"526d05afa984a663664d5d627cf50678a2beca17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:33:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993fb175d5ef4b3ba1e9b6bd5afd8e51\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d29f6b129bae4737a8ab6981a566eb6a : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d7864a0152834dcf8c2f6c04f02e753f : SUCCESS in 1h 33m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b6ba60c530a6461cb0c30ee68c09c2e3 : SUCCESS in 1h 02m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/87e69c13625642e5896bd133a2119c70 : SUCCESS in 28m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c88221918b884963a082565817a7fb6f : SUCCESS in 1h 14m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e1af8f18618c4923b8c63d8c8e267b41 : SUCCESS in 1h 55m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577ab92bb546414dafd5de4f1f6cfd1a : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e64631fe2e4a31b82b4dea85f801ea : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1325aab5e2bc4ba98f3ca3139997f4f9 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b26fcfbf9d147a597841f3ed712c9ef : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deed058c086e453e95d65e4e54779bc0 : SUCCESS in 5m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/29fa6e83d925419695e2d01c194b2ae5 : SUCCESS in 20m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3681158d78154f818f13152f565b95cb : SUCCESS in 32m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76491685631046c3bc58604c9186b829 : SUCCESS in 1h 02m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c34d6cccfe004e08b7dbf3d37777a731 : SUCCESS in 1h 16m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04fc14030aaa42858c6f1e85e9d3e208 : SUCCESS in 27m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0a4f03b91d1a4570858c4abfe385f9d9 : SUCCESS in 25m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d34e5e95a3604707b552bab73cd645b5 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4d21e5276dbf4c449985528dcdaea19e : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":7},{"id":"4288fef7d28548359223f3b345fda2adf988b133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 14:35:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e2c61fd140483e8eec18c2ffb9cab7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3328ba9212834551bce3a8b2aac03695 : SUCCESS in 2h 43m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a0ade5616c14ab99297c07a6d7475b2 : SUCCESS in 2h 13m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a7da5de634524f78a16e03e3e95216d7 : SUCCESS in 1h 50m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a6a49ddcf6c46969251b83e477c69bb : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82208771b6ea4c81a2a81e617a71668f : SUCCESS in 5m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f2ffc892f922461e8045445479eecc3f : SUCCESS in 2h 40m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d8fba282a233448b981df5a57ca8db57 : TIMED_OUT in 3h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ffbcbe7c1ffd8cbae5e793bd88805e3568988de5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-09 15:29:14.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"94c19b810a2a963ec514c1aaaa748592abd0215a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 17:55:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e4a33f51684c6d875819b011718bff\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a8c95cf751cb4f11b0fa4e4d74e95222 : SUCCESS in 23m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/21981f326d804bc6b2525812a7529857 : SUCCESS in 1h 10m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7df6fe687f8e4618b1d0fd266425ff4f : SUCCESS in 28m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f5ea3bbdf51141f8b55d0f4d352f2b4e : SUCCESS in 36m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/599841f774fa43158f0f4658b20a14bc : SUCCESS in 1h 15m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0881a270de474b448a9f00f721cf4ed8 : SUCCESS in 1h 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542fbc55b95348758c7a29cc9dfce052 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57b3bb7f9f674fd09a1f1fc59694085c : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13b6848a01024fe1999934484a62de9d : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba771edcaa054424bcbac80a24d53ce9 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01f2a7f2e034192b1ba7e47c49f212b : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bbc31212b8f9405292dd94c8a8f5ba9c : SUCCESS in 31m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bc3ac4fa9bea4584bc23c4acd87da1f2 : SUCCESS in 26m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a8651a4afa044d0b856e24b319af9995 : SUCCESS in 1h 06m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4672f85c372c4690bf124b256f5af3f6 : SUCCESS in 1h 16m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88120468ec29454ca2f2870ad3b03541 : SUCCESS in 36m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cead0d6e508944828e4f6f63b7894796 : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d72edc93869f4041bd881bea9da06c2d : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4d36dfbb211e4cd8b5b721e8183086ee : SUCCESS in 1h 23m 40s","accounts_in_message":[],"_revision_number":8},{"id":"80b6c7a27d13f4c4b1d67162c123ef7333a1668e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 20:28:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b287560a4a442b09ededb73b3a1f422\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d79ae86f56954d18a19e1e631d72f4b1 : TIMED_OUT in 3h 01m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/04ce1e513ddd42a99232fc73c0c69e76 : SUCCESS in 1h 57m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/252a41a68f3740ce8f3a83e8a275518e : SUCCESS in 2h 13m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f6078bd22984486a990f2ccdaabb428 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a28b79debfa44429a62c44ae75e579ba : SUCCESS in 4m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a13e13f68bad491b8e9c8f64d7d8a47f : TIMED_OUT in 3h 02m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9d3ef453ffe94dc1b92c49f02510d623 : FAILURE in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ea2b57352d711e09461a443041287047bff67615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-11 12:51:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3a55e9e375cfd67344ae0da277602d91d30d04bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-11 12:53:17.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f5dd8bdb6299bb138e6b426f569132b3b90a5a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-11 13:47:28.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"d4dd28f8459f803a2051ef49c0f41aaed07991c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:18:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ca814d4260452188bc1b908eaa63d4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3e0c39264ebd46288d4a4d9963743b91 : SUCCESS in 25m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/68a7ba72f1f546e4bfa1d1e8da9db218 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c21e5c60b6b4878983582cac7bdc107 : SUCCESS in 17m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d03e5f504b8649f29dc80b9fa3c66856 : SUCCESS in 25m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/851419ed8da44cda9ed200e726fdddff : SUCCESS in 1h 01m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f197af071d6a4e4abb7c9e0763706749 : SUCCESS in 1h 30m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baee8a4d69c347d98a07133bc075b5c2 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a79ffb8e4a394efb99831dd5800cf906 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dadcc39849774bb5a6abba55224861c0 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2447a6da8f546cc912affd17ebebc2f : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f8688e9aa8468396b901f7b5f31abc : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/efcbf190fde1483ebaf6fc929d059542 : SUCCESS in 28m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb541aab405e4850a2531ceb18c3bc58 : SUCCESS in 36m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cea13b7948f24195bed01e6661647cbb : SUCCESS in 1h 26m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d189575223e74f958456ae8af4983aa8 : SUCCESS in 1h 19m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/caa34c0c45a947bd975de106da118bba : SUCCESS in 39m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9796c87f31ea48809bdbb8a3437ee7d3 : SUCCESS in 24m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/92a8ed460f1d44aaa0a01e5656411baf : SUCCESS in 1h 02m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c319fe86c5e488abc8a8e18b80c0c4b : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":11},{"id":"ea8e4fbeed304c829d1a3c1463257bec2589dbd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 16:55:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f52619adec614048b6d41a257f562469\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fea5d41704344b6db3090db983364de8 : TIMED_OUT in 3h 01m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14b4e442b234491ea399d74e825ccb8c : TIMED_OUT in 3h 02m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2316704ad334760ade23c4b41521f24 : SUCCESS in 2h 09m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fabdd9a6b60e4a079d665a8573e8d587 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2d4d4eb97614a9391b6a96451f41e56 : SUCCESS in 6m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/84e01954fa8d45f3a3dc7df161fd0a75 : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eb98627074064c258a415934cf2daeb3 : TIMED_OUT in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8f72e7d7b148a4afc6208947ce0c1e6af063ae82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-18 13:13:55.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_27339\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"_revision_number":12},{"id":"2f28e6fd184e48ace49c70df3901bf7fffb52c19","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-18 13:14:20.000000000","message":"Patch Set 12: Backport-Candidate-2","accounts_in_message":[],"_revision_number":12},{"id":"e3b79301524a5f8efce88452d938ebb144db41a2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-18 13:14:36.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"9374ace94ee1417b1244784d6a793e13fa20cc21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:15:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/062bac335c6943938d91db38122da08b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7b012ade91fc4de291ccd08b3dabef62 : SUCCESS in 25m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2d83ec7a50eb4e4da2d449f22f7e7d5e : SUCCESS in 1h 07m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c6ce133f4fcd43ce9438e89584fe436e : SUCCESS in 48m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5d761025de6546969e05a6c2cf796ebd : SUCCESS in 33m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/868eac562b5341378af8ed02bae17cef : SUCCESS in 1h 27m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f7f96248e0814c1396979bb2e946ab35 : SUCCESS in 1h 26m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31eded175e6c4f0292dd2b16a8ba58ff : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b1457b7c4ba46acaa434c57c71313df : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a52596c25c348139ed5aa56973eb0e2 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c6bd6cfc4dc444ba4500590f23d847d : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6da544095f4d3f9edd53527750f863 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/88a8ab871429424b96a601ac343789fa : SUCCESS in 29m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0ca9d09a4ec3425a9354ec91a5a068b4 : SUCCESS in 48m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3224da086bf24c928bfd0407bad7d0f1 : SUCCESS in 1h 13m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e351eb156934336bd2c67089a8420f8 : SUCCESS in 1h 37m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a1977695cf5b41acac65fa8ef9e8330f : SUCCESS in 21m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/68f1f1b51b594d1281fa9b62058a3611 : SUCCESS in 27m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/edd76e5c9bed41f78bc0d5c5807f8976 : SUCCESS in 1h 08m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ce8fdf57ce054f5482dd049ad594ada2 : SUCCESS in 1h 55m 42s","accounts_in_message":[],"_revision_number":12},{"id":"ff7791d62dd32dd0ecb4f3a6fd21e0c7b8cd2bad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 18:37:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09543bef9fb34b0d9b589a7b44be596e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/628545508ac74d1fa3bba206abef1dde : TIMED_OUT in 3h 01m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/382a5dcf4fce4a0994b715d713e0cd12 : FAILURE in 2h 32m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ebf7b8d7297c4342bbb8fdd6f7fa5cc2 : TIMED_OUT in 3h 03m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31a2c4a17ed94bf0813a2f3202f3eefc : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71d96750bb4047ef812b10874a114410 : SUCCESS in 5m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5c02775f17df40048d00b15c961faed7 : FAILURE in 2h 45m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1f24a0fb2b14466e9a3a3db20e4afe0f : TIMED_OUT in 3h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"26283c7405073a63a667893b169acbe552a7b23b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-10 13:37:34.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4e0f4f866d975d83fd517f62049ccf968fa6ba1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 13:37:56.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"6468fede51a3472dfdc82c9e69e9f2d3492b9bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:27:03.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1392c195334cfeb1bf839610732ef6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/248c328fcafa4f47af04b075ee482554 : SUCCESS in 13m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ce214327997d4b6b8594f6392dc82c1c : SUCCESS in 39m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/75e03036d5a54e7d82988a2f4305dcd9 : SUCCESS in 1h 07m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/97c2c44c2fc44b4fb572e6b5a48eb3b3 : SUCCESS in 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a863088414411580178eb1f34046d7 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b9e804e5194454381981398bd5df21e : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/859d34c0ab894287b947ff60dc358ecc : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f654db09014adca8becfe73073ccf3 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9c2f677f664f4cc29644231114c0183d : SUCCESS in 30m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f065929af2649b3acb30976c85f1fed : SUCCESS in 23m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2753650064e4effb2336068400eeefc : SUCCESS in 1h 05m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/be4b6168e35c41188847cdc618b36fc1 : SUCCESS in 1h 38m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9217262b4778498195afb7a4591e5938 : SUCCESS in 25m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/be8f1de57e7b403cb73620e84763e77e : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/67e39baa6fa642609b61f037dafd777f : SUCCESS in 1h 14m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/27fa90779f774f9e8b7ab96c23e7ec9f : SUCCESS in 1h 33m 03s","accounts_in_message":[],"_revision_number":12},{"id":"f44e13f65e797be4072aee184afb9615798ac742","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:27:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"df0dceb0bf888c4e44afc3592ebe3f8ec40acb4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:28:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfbc487760b34b6895fab5031aa9b8ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880848e3f7bc4f3fb6a22d33e788e5ca : SUCCESS in 39s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"36ef420f827dd1d8150720f2ab84ddd34c54ca36","revisions":{"67c533a8c8e9a003e86f2c96cc70b5753cb39e67":{"kind":"REWORK","_number":1,"created":"2024-12-03 08:50:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/1"}}},"commit":{"parents":[{"commit":"d22245c71125c8c1c053fa772a362f0446467d0b","subject":"Change copy-cacerts behaviour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d22245c71125c8c1c053fa772a362f0446467d0b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67c533a8c8e9a003e86f2c96cc70b5753cb39e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67c533a8c8e9a003e86f2c96cc70b5753cb39e67"}]},"branch":"refs/heads/master"},"a83c0912b54992e9683a0a174a2725ad209d1182":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-06 15:18:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/2"}}},"commit":{"parents":[{"commit":"681cb1b26d2bcf05ab70edd362987d201b9f9b78","subject":"Merge \"Add Let\u0027s Encrypt EAB support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681cb1b26d2bcf05ab70edd362987d201b9f9b78"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-06 14:49:23.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83c0912b54992e9683a0a174a2725ad209d1182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83c0912b54992e9683a0a174a2725ad209d1182"}]},"branch":"refs/heads/master"},"f8eb652dca0318192e866f57d17d9bf5463cb780":{"kind":"NO_CHANGE","_number":3,"created":"2024-12-06 22:48:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/3"}}},"commit":{"parents":[{"commit":"681cb1b26d2bcf05ab70edd362987d201b9f9b78","subject":"Merge \"Add Let\u0027s Encrypt EAB support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681cb1b26d2bcf05ab70edd362987d201b9f9b78"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-06 21:10:22.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8eb652dca0318192e866f57d17d9bf5463cb780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8eb652dca0318192e866f57d17d9bf5463cb780"}]},"branch":"refs/heads/master"},"b4242c3cbe996866cddb17acbf4f26490b6d8dd7":{"kind":"NO_CHANGE","_number":4,"created":"2024-12-07 07:28:02.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/4"}}},"commit":{"parents":[{"commit":"681cb1b26d2bcf05ab70edd362987d201b9f9b78","subject":"Merge \"Add Let\u0027s Encrypt EAB support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681cb1b26d2bcf05ab70edd362987d201b9f9b78"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-07 07:25:54.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4242c3cbe996866cddb17acbf4f26490b6d8dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4242c3cbe996866cddb17acbf4f26490b6d8dd7"}]},"branch":"refs/heads/master"},"f5f8a27ddb567d0ef190acba6f78ff8e2b2bb77c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-12-07 19:59:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/5"}}},"commit":{"parents":[{"commit":"1537574c29b4bde0f384a1c7155bf9ff11776ec6","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1537574c29b4bde0f384a1c7155bf9ff11776ec6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-07 19:54:53.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f8a27ddb567d0ef190acba6f78ff8e2b2bb77c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f8a27ddb567d0ef190acba6f78ff8e2b2bb77c"}]},"branch":"refs/heads/master"},"5cb6fbbe1dcb42b68ac3b11008551ebb021b55c0":{"kind":"NO_CHANGE","_number":6,"created":"2024-12-09 05:54:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/6"}}},"commit":{"parents":[{"commit":"708b460fac6978c734435f5615644c63f57ba9f9","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/708b460fac6978c734435f5615644c63f57ba9f9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-09 05:50:39.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb6fbbe1dcb42b68ac3b11008551ebb021b55c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb6fbbe1dcb42b68ac3b11008551ebb021b55c0"}]},"branch":"refs/heads/master"},"aca1e8079924edd1e4197c487d830e37320915bc":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-12-09 11:32:53.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/7"}}},"commit":{"parents":[{"commit":"c153764de3ed712be61215b958bbd8d984e8e347","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c153764de3ed712be61215b958bbd8d984e8e347"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-09 11:31:55.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aca1e8079924edd1e4197c487d830e37320915bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aca1e8079924edd1e4197c487d830e37320915bc"}]},"branch":"refs/heads/master"},"18f89659337da480589161f1d77a66dc3ca268cf":{"kind":"NO_CHANGE","_number":8,"created":"2024-12-09 15:29:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/8"}}},"commit":{"parents":[{"commit":"3e69a4d7944f6a1adf7152138be0536feda077a9","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e69a4d7944f6a1adf7152138be0536feda077a9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-09 15:28:05.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18f89659337da480589161f1d77a66dc3ca268cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18f89659337da480589161f1d77a66dc3ca268cf"}]},"branch":"refs/heads/master"},"cec1c81523057ee7a9a60ee69340a7bb3fdac492":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-12-11 12:51:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/9"}}},"commit":{"parents":[{"commit":"02cd156c0d4e782c2c2ba63fe6b3f7b77863a80b","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02cd156c0d4e782c2c2ba63fe6b3f7b77863a80b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-11 12:50:56.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cec1c81523057ee7a9a60ee69340a7bb3fdac492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cec1c81523057ee7a9a60ee69340a7bb3fdac492"}]},"branch":"refs/heads/master"},"a0fec014cf56590d38cf023fbe5c0c409179de3b":{"kind":"NO_CHANGE","_number":10,"created":"2024-12-11 12:53:17.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/10"}}},"commit":{"parents":[{"commit":"002360d2e61b2a41e3b1ec81268e38f73e2196ac","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/002360d2e61b2a41e3b1ec81268e38f73e2196ac"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-11 12:53:09.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0fec014cf56590d38cf023fbe5c0c409179de3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0fec014cf56590d38cf023fbe5c0c409179de3b"}]},"branch":"refs/heads/master"},"5b62430cc9aa13aa00b7b695a9f9397abba1c86b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-12-11 13:47:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/19/936919/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/11"}}},"commit":{"parents":[{"commit":"7180c5124d35a25b256538f30cccb304a0a2d7f4","subject":"Fix EAB support in letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7180c5124d35a25b256538f30cccb304a0a2d7f4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-11 13:47:13.000000000","tz":60},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b62430cc9aa13aa00b7b695a9f9397abba1c86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b62430cc9aa13aa00b7b695a9f9397abba1c86b"}]},"branch":"refs/heads/master"},"36ef420f827dd1d8150720f2ab84ddd34c54ca36":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-12-18 13:13:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/936919/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/936919/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/936919/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/936919/12"}}},"commit":{"parents":[{"commit":"b1be32617803e2a6c24f449a37622717e2a6d102","subject":"Merge \"magnum: Add helm binary for magnum-cluster-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1be32617803e2a6c24f449a37622717e2a6d102"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-03 08:37:29.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-18 13:13:55.000000000","tz":0},"subject":"[letsencrypt] Rename script for updating certificates","message":"[letsencrypt] Rename script for updating certificates\n\nThis patch renames the update-haproxy-cert script to\nupdate-loadbalancer-cert to make the name more generic.\nThe change is necessary because support for ProxySQL\nwill be introduced in the next patch, requiring a name\nthat reflects broader applicability.\n\nTo keep the diff in the following patch clean and easier\nto review, this patch focuses exclusively on the rename.\n\nChange-Id: I15f9b5cf3976f4298cdc7da4530f2d7d09614e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36ef420f827dd1d8150720f2ab84ddd34c54ca36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36ef420f827dd1d8150720f2ab84ddd34c54ca36"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
