)]}'
{"id":"openstack%2Fkolla~937345","triplet_id":"openstack%2Fkolla~master~I9c805b866ee359d2da150438ca78be25b1462988","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-09 22:39:03.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-10 00:17:46.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-09 14:51:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I9c805b866ee359d2da150438ca78be25b1462988","subject":"CI: Stop using LABEL for ephemeral0","status":"MERGED","created":"2024-12-09 11:00:27.000000000","updated":"2024-12-10 00:18:53.000000000","submitted":"2024-12-10 00:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937345","meta_rev_id":"00cdbadc96b4d148bab860b7e2e48d47988cf649","_number":937345,"virtual_id_number":937345,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 00:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-09 22:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-09 14:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-09 22:39:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-09 22:39:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-09 22:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 12:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 13:10:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-09 14:02:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-12-09 22:39:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"020ffe8fd9db4defc920ee8eb061aba82cf2620c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-09 11:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25bad279fc9f862f5bb9f6f0ec75b208ce07f690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 12:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3148bbeec0e4d5283ef3616ec8e7573\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7609028d7c204db29612bc624d71d286 : SUCCESS in 25m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1d87f7cd770b48caa98220e908db6e0a : SUCCESS in 58m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c60506fdfcf641528784f031dc4f3768 : SUCCESS in 22m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fce6b266c42e4dbd9ab7bda4c0f19f9d : SUCCESS in 28m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d97686e3d7a948f9b3dc199cb75636f9 : SUCCESS in 58m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/93bd921595944c14b79c13e11fbe06ea : SUCCESS in 1h 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4556a0f4782d437fb4ef88173f6f299a : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3bb0df27776431698e1baacf136d23f : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81fe339ecc3440d59bbeceb571976dfa : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5b45e42abfe4b12873ac95d8f8d8e40 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819f3e93c6b44fe89583a72918372360 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47b86d14d00f48828a4525bfe7e951ba : SUCCESS in 23m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/da46c67d54ee4c6485d1c53c250373b8 : SUCCESS in 38m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/73177dd90ba34c779590a9f298c768e8 : SUCCESS in 1h 00m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba0df982c61b4ead8eb577f0cb497ad7 : SUCCESS in 1h 26m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d51e53a31b1495c968176aae4c79d2a : SUCCESS in 19m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0b80cb828a9d4cc5baa7ff2efb04c39c : SUCCESS in 25m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/398246499dec485dafa12a4cbb1cccf4 : SUCCESS in 1h 09m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d11f85674ebd4f468e0060fd09373802 : SUCCESS in 1h 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9b5df1ce9e94e660de48608f414ce27bd9f0138b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-09 13:10:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbe76826b858a2327882a0657ea5678b08ce6882","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-09 14:02:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"59064264aae2546257dde6f630df98daf95453ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 14:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d2b9bbdde54700826e18ef482023a7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/77ecc5dda72643f98c95060425826470 : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f26e40d31d984e268be909d5cd328c3a : SUCCESS in 2h 42m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3321b51ec4a04391b1940fc7d082ea4b : SUCCESS in 2h 03m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76e92f7c153e4163a234e644a00cd104 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b90618f3f0e049ea9ace09e4ed45c992 : SUCCESS in 5m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/97638565f9c340faa421a41185204514 : SUCCESS in 2h 41m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a2a6d98a32314f7ba132e4fc1f72d271 : TIMED_OUT in 3h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f07c9f38e8ee59d2b68b1284a652535237bb037c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-09 14:23:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a0531a6b4dae9ee03f08288bc6d6feb6a636ad3a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-09 14:23:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37112bc6bd5140f0bb710e37d01e9aec0a5a2df3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-09 14:51:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"58f169a89d1f79b3e6815b5e51c2ce5d79763403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 16:49:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3c50e69aac74999a204733525cf59e1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/581c0f9222fc44ef9e18ed4e4397c8b4 : SUCCESS in 21m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/31ae7a838ff9476b8f2f2ed1799494e0 : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad1e8b1d402c4224acd2f9bda46a1c94 : SUCCESS in 19m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7247c7189b5a426e9a84361b9d793a7f : SUCCESS in 33m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1e051a32a08948b08dd2ea8e8f0755db : SUCCESS in 1h 11m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/70a8e79f05b74011820e3dfad6af4681 : SUCCESS in 1h 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e509dd333234abf88fdd571a49b8da2 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/116d8e4bb09c4c8ba2bcf6f06ff342d1 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/167294e8b9114b1ea0ddef772a03962d : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b2a0240eef441f0add6ea8a66b0cc26 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7416d8215bbb4c6095f1cdcc71a5b81f : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1f4634de2b2e495083fea42bf3b660e0 : SUCCESS in 17m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/40bebb3039484f638f393c3d4b3ee4df : SUCCESS in 33m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f87118d907184e39aa9d71769d727af2 : SUCCESS in 1h 01m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/087313db5c994450b8d8391d7aab4f56 : SUCCESS in 1h 37m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d680e04ec7f644788dbc9582e96196bd : SUCCESS in 28m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/23051fb9f9894223a43a905f4e10ab01 : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e9bb0a03962648e0b5f1666cc93cb9bf : SUCCESS in 1h 11m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77646b365d4d45d2a7df8f9178b401ab : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b561c9484397d5fcfd77a94cb85a8a1a8ff23f73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 19:50:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b838bd549c46f698d8c62564c3144b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2b3dfa10926147699f0aa25df4b7d861 : SUCCESS in 2h 45m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/603b26e6b5514f94a10487d9260b7242 : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/582aa1ba62604d2a9e26dfc0e4492e51 : SUCCESS in 2h 16m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a57b26e9cb54d5ebb4f594dd09c84a4 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ccf83a71900a442d8072bbbc220cd4f8 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/729dacfaf5904c71af040a221a106c80 : TIMED_OUT in 3h 03m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e82ce7b5a20a4f22a367e6ba71c6ca0a : TIMED_OUT in 3h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1d7fd4adeebb53482e5029d636e3c038a8fba50","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-09 22:39:03.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbcfc53a22aa89b311a7f78abe1f555a8a9406d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 22:39:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f32efd29be0ee6129503e791d10016e49debc7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:17:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e781daf0fd14977ba776c10fae8c2b6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2d3e7c8ef1704b46b46231c4fe4eebc1 : SUCCESS in 19m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a837a1cbc83c47218ca1ea12ec7d6531 : SUCCESS in 26m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b5afccf299194248979c78d2aafdbb05 : SUCCESS in 1h 01m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5aebd2e8a59245d9a0de0a66b3c92e33 : SUCCESS in 1h 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b37316a3514f17b880cfb1a91a7d5e : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bba3fbb400741d6a0edb3a9525f6674 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a53ee5ce5554fa3934a7f7a93420119 : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3222b15f42d742ea86c6ec67aae2eeb9 : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1058c6f798942989d20af707659bf13 : SUCCESS in 26m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cf313430c40047e69b7f289a650ca1ce : SUCCESS in 21m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e79babd34faf4cb68fe8fbd25f5e4790 : SUCCESS in 51m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aa0ba00d50404a999207dbb7c0d5f0b8 : SUCCESS in 1h 28m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/61670f12783d48de814d6c0326a1a717 : SUCCESS in 24m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2a54d7395d82484685a588e18a2b426b : SUCCESS in 17m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a9bd34b349ab4e0b91312470d336a155 : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9b1e169b58c8460eac501eaa231a0689 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4a3248b9e15b845d1395762b645dcfdfacf67d3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"00cdbadc96b4d148bab860b7e2e48d47988cf649","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 00:18:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4e4fecbd0841f38f87b06921015a9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0cde364e1e4bc0b065b29f197876e4 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6a15966aab9d24864e313cf135eb4bba2ab6fb2","revisions":{"9c37611dc2a505020798a5af54a834cdb734186f":{"kind":"REWORK","_number":1,"created":"2024-12-09 11:00:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/937345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/937345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/937345/1"}}},"commit":{"parents":[{"commit":"681cb1b26d2bcf05ab70edd362987d201b9f9b78","subject":"Merge \"Add Let\u0027s Encrypt EAB support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681cb1b26d2bcf05ab70edd362987d201b9f9b78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-09 10:59:42.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-09 11:00:25.000000000","tz":60},"subject":"CI: Stop using LABEL for ephemeral0","message":"CI: Stop using LABEL for ephemeral0\n\nFor some reason it doesn\u0027t work in raxflex, i.e. the created filesystems\nlabel is not seen.\n\nChange-Id: I9c805b866ee359d2da150438ca78be25b1462988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c37611dc2a505020798a5af54a834cdb734186f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c37611dc2a505020798a5af54a834cdb734186f"}]},"branch":"refs/heads/master"},"a6a15966aab9d24864e313cf135eb4bba2ab6fb2":{"kind":"REWORK","_number":2,"created":"2024-12-09 14:23:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/937345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/937345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/937345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/937345/2"}}},"commit":{"parents":[{"commit":"681cb1b26d2bcf05ab70edd362987d201b9f9b78","subject":"Merge \"Add Let\u0027s Encrypt EAB support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681cb1b26d2bcf05ab70edd362987d201b9f9b78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-09 10:59:42.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-09 14:23:01.000000000","tz":0},"subject":"CI: Stop using LABEL for ephemeral0","message":"CI: Stop using LABEL for ephemeral0\n\nFor some reason it doesn\u0027t work in raxflex, i.e. the created filesystems\nlabel is not seen.\n\nChange-Id: I9c805b866ee359d2da150438ca78be25b1462988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6a15966aab9d24864e313cf135eb4bba2ab6fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6a15966aab9d24864e313cf135eb4bba2ab6fb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
