)]}'
{"id":"openstack%2Fkolla~937434","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I9c805b866ee359d2da150438ca78be25b1462988","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-10 11:02:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c805b866ee359d2da150438ca78be25b1462988","subject":"CI: Stop using LABEL for ephemeral0","status":"MERGED","created":"2024-12-10 07:14:00.000000000","updated":"2024-12-10 11:03:57.000000000","submitted":"2024-12-10 11:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937434","cherry_pick_of_change":937433,"cherry_pick_of_patch_set":1,"meta_rev_id":"8fe08bbed652f424eb1b93d7b78a695486d5a231","_number":937434,"virtual_id_number":937434,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 11:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-10 09:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-10 07:21:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-12-10 09:17:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 07:14:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-12-10 07:21:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-10 09:14:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-10 09:17:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"5e7410711c4f206d6db471a00e8d15099263406f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-10 07:14:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"add190693b0b9ba30058cca512cb84315578af07","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-10 07:21:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"201a4ebae1461a684fd3a26da996fc8985a7fad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:14:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47adc3de5b864cc8b47b00fa7354836b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/621dd606975e41239fa379261d9bea5c : SUCCESS in 27m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5ae9c39f0b9847ab92751846aeb6ff4c : SUCCESS in 1h 08m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8e2f5dc9d49e485e9ba8f80f8ddd525e : SUCCESS in 15m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dfc9f8110714418e9414fe1ca9f93d55 : SUCCESS in 31m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9bbfe1eb48674007baebacf1289c56fb : SUCCESS in 38m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8d949c97d91c4bcab41779e3f7830f60 : SUCCESS in 1h 37m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a488b71320448882b4cc01a53fe053 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0a73302f14646f7a152e6aa1ef6e27f : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beeb119f8d2c4ab3a216dfd6864bcf29 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24f44b523dc44d7ca1f583c6589a7458 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9b1846ab505432383cbe8d9a34ac111 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/713514545267423eb7af855fa5da2f22 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccfbad491ef4470390edc675b99dbc88 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d2c60eb4aff64c41a9c752a414650d52 : SUCCESS in 32m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/664f248b046d4b63a3e112e11fd30d3a : SUCCESS in 29m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a42f532df70e41e2a1d4ed5858140fa5 : SUCCESS in 54m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/adeaef06cfd049cebfd663168890e0e5 : SUCCESS in 1h 36m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec1e59c5748241a39d1c0170467f8fa0 : SUCCESS in 18m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/40e145915cf7493795b077e481e8b3ea : SUCCESS in 1h 11m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c14105a31504b92bb0351606d9ac963 : SUCCESS in 58m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1e5d16cb4ed24dd6a9abc32862111520 : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e3da87150d8d48b273a2fd7cc143ebc7b7a0f325","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 09:17:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0769202fff5518b3ef8657e4544b0e0ab3daaf5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:17:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75fe76e32365198f520cf319f4ab6e5245af5783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:15:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8dbf45995e34bc7a5c95b6c56c985b9\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a8a5444667e41ec916e78a6f59df5eb : SUCCESS in 2h 46m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3c239917e0074f3cb8de0d0e3203d694 : SUCCESS in 1h 56m 50s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e16de492fb64596a961a1c69fc389d6 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be165e77190542458343160bd3a3d51d : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/128ec134995445579750a457ec7122a3 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/144049137c814747b14bb24a3758096a : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27781ca55d19c600f7244521b30f992ee0f0af6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:02:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d1d53d7698d4a3085b0ac1b758fd3d3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e2dc9341ec9740a8a044689dcd792c67 : SUCCESS in 22m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2e1c2d8f0f18493ead6c6ad789659534 : SUCCESS in 35m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9c054f88546b489bb56484ba29e85e53 : SUCCESS in 1h 03m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/969842c5e307427b80cff169fa04f8fa : SUCCESS in 1h 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085ca0e5f9244f43a920a64d39e84430 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1191d15160ed489b9205b47fb1da58c4 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04b5bc24caef4ff19ad035f1640d20d9 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e042ce9a36164cceaa8abfd621b45f4e : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e36dfff83034fb2b50d6b2e710798d0 : SUCCESS in 3m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2686adba15e449658dcbbdab0cc61f86 : SUCCESS in 27m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff14e31f7abf47cf8dd9861f822a0d49 : SUCCESS in 32m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/70b3413fff1c42b4813ccd9a24639912 : SUCCESS in 34m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/63b2fb38eb954d31a393a66b2f019434 : SUCCESS in 1h 34m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0051782f93d64f438f9e2c2ea99d6d25 : SUCCESS in 21m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/96ab0fe02b8b47169ce99ffb6b373559 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9354053d99ab45388da6e0e860a4cb9b : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/48a098240e4841d8996d45cd5545f501 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0d19efaffbb7774a3cc066ac980e8224c4a9d956","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:02:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8fe08bbed652f424eb1b93d7b78a695486d5a231","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:03:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae8f4913556a48a8be79ef1349c9ef51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2947f6e5e5924fdab889e30072a1bf16 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01a9e7b992672ca7432bb3fb6d076cb243c9ae05","revisions":{"01a9e7b992672ca7432bb3fb6d076cb243c9ae05":{"kind":"REWORK","_number":1,"created":"2024-12-10 07:14:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/34/937434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/937434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/937434/1"}}},"commit":{"parents":[{"commit":"96bf08833bedbef4b48c7d1a16d937404b1db28b","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96bf08833bedbef4b48c7d1a16d937404b1db28b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-09 10:59:42.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-12-10 07:14:00.000000000","tz":0},"subject":"CI: Stop using LABEL for ephemeral0","message":"CI: Stop using LABEL for ephemeral0\n\nFor some reason it doesn\u0027t work in raxflex, i.e. the created filesystems\nlabel is not seen.\n\nChange-Id: I9c805b866ee359d2da150438ca78be25b1462988\n(cherry picked from commit a6a15966aab9d24864e313cf135eb4bba2ab6fb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01a9e7b992672ca7432bb3fb6d076cb243c9ae05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01a9e7b992672ca7432bb3fb6d076cb243c9ae05"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
