)]}'
{"id":"openstack%2Fkolla~937517","triplet_id":"openstack%2Fkolla~master~Ife393e7f1b00fa37921e7822668d6a40dd467d5b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-12 06:59:55.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-12 10:01:05.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-11 22:50:30.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-12-12 10:01:05.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-12-12 08:20:11.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ife393e7f1b00fa37921e7822668d6a40dd467d5b","subject":"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"","status":"MERGED","created":"2024-12-11 07:41:28.000000000","updated":"2024-12-12 10:02:06.000000000","submitted":"2024-12-12 10:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":921032,"submission_id":"937517","meta_rev_id":"dfbdf9a985a314d2d5e6e4e9024f0655da00924d","_number":937517,"virtual_id_number":937517,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-12 10:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-12 06:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-11 22:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-12-12 08:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-12-12 08:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-12 06:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-12 06:59:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 07:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-12-11 07:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-12-11 07:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-11 07:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 07:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"e676f035f0c6fa02185e3238f162f38601ba4d50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-11 07:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a0b05fe7f659bd1f7a488a024875b81a14f8ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be09608c883444009f52860104754e45\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4b20906f990244c7b69fde320503d7c7 : SUCCESS in 28m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93337f2c7d25485483c580028a5a6ddd : SUCCESS in 1h 04m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e77a1e44d834bfcaffe061261fe94ff : SUCCESS in 16m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9cc82871900f4b7baabd4034c1d09c85 : SUCCESS in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fff72b93d79e42a7be893ebba8219633 : SUCCESS in 1h 08m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e26ce954efac44abb2ea0bb732126e40 : SUCCESS in 1h 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb53afc540242a489b41eb260215d27 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1185c20dac8f4d9facaa1a576a581679 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5ce16291aa54d95813f7bf28feb2aa3 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da051545529b47aeb41e8e720b34fc4d : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6c4bb1eb134b00bea1db2112acb074 : SUCCESS in 5m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9ff9df38a43641d892a1e29435dce763 : SUCCESS in 29m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/506ae4652d9d4d5faefac1e2c9ba95d7 : SUCCESS in 32m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/feff05a832e34a7ebe9357b93b06570d : SUCCESS in 1h 09m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dfe3c5524b11455c9658d5c12f78685a : SUCCESS in 1h 35m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/489ae7711d824da9a03d6301b5955a5b : SUCCESS in 27m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f5b52582e2e44eea5be87d6a9fd1c83 : SUCCESS in 35m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b20e4b95c33340f4b16902153ccb1963 : SUCCESS in 56m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/92d330cb2d3b453c82cf02e9b30a848b : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7b325192a95a1f5040193362ebdfb407b1a4361a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 10:48:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85ab9950f03a41a4a3a80f714a810fc4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d7b61d1d09fa40668cecca93b1a91b05 : SUCCESS in 2h 39m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8f319e622e744e1821dd11b6a94ce4b : SUCCESS in 2h 15m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66f49c2c8e784c74880652fe3444d6ad : SUCCESS in 1h 40m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7316590f238d4252bcba4d2c13beaed6 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1416b8d1b17e44729891cca8af22c0d0 : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f1309cd7b15f439ea52a6819eedf7876 : SUCCESS in 2h 36m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3ca1b1912ca849e19d147c27b532f385 : TIMED_OUT in 3h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0a21577f3f7642a4debcb5ca97e581c8256b5aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-11 22:50:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"080d42599eb2562b5ab6c213801c0b6fd22619af","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-12 06:59:55.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53e6256bbaf6a1823f013d40cfd4914191b5a9f9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-12 08:20:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27407cd7d1174dca1f8da116061015a28cd8b50b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 08:20:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dd547a468835f1945c000cfab309af9fc8d83d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 10:01:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f216e391f94f7282213cb205ce169d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1f072ea653e4c8482cd5b029180c490 : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/27cf56bd4f8a4a0c985511d9d01920ff : SUCCESS in 30m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0d4e75aab74144bea2b6af0288038d28 : SUCCESS in 57m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/088132c6dec44ca2af49ebdec8f99e33 : SUCCESS in 1h 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfc915e3d0d403d96bf3b56f14d48a7 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ab918a59c745568ce735d7a1d4696a : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca458ce401844823a9f20d7aeca4c7ea : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a521092205d34fe4b2aac0822e20201b : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2ec5758127a41458e2d8bfbd22b8daa : SUCCESS in 26m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7bfdefe5f205456e91d2d50d790e8a77 : SUCCESS in 31m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed9e1f115fcb49119f157d7782a70813 : SUCCESS in 59m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9feec3a5df064ae28596c96022684828 : SUCCESS in 1h 35m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c86ae78782a24a86b0b93bbbe6f39b93 : SUCCESS in 20m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4c847bffbfae48a78c7119e4e3b5dd59 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81b59537366f42f4a9a7e4c23cc52516 : SUCCESS in 1h 09m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e996c75e1ccd499aa74d1f775c5a715f : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d0a4d743bf8fe8be954d7abb4b0b8e08f8e4df1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 10:01:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfbdf9a985a314d2d5e6e4e9024f0655da00924d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 10:02:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac86a423144c4e36b456e0442aa5a143\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ff89c351594559b4e28f4ee452986c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f33c1ef176c16d0be9eeb393d86dd8271ea8498e","revisions":{"f33c1ef176c16d0be9eeb393d86dd8271ea8498e":{"kind":"REWORK","_number":1,"created":"2024-12-11 07:41:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/937517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/937517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/937517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/937517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/937517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/937517/1"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-11 07:41:28.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-12-11 07:41:28.000000000","tz":0},"subject":"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"","message":"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\n\nThis reverts commit 428f37124986019bedc897469f9b1f3470b62082.\n\nReason for revert: Should be fixed now\n\nChange-Id: Ife393e7f1b00fa37921e7822668d6a40dd467d5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f33c1ef176c16d0be9eeb393d86dd8271ea8498e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f33c1ef176c16d0be9eeb393d86dd8271ea8498e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
