)]}'
{"id":"openstack%2Fkolla~937619","triplet_id":"openstack%2Fkolla~master~I52105f6cbf7529ce107cabf8a38a10525be7435c","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-16 12:10:04.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-16 14:19:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-14 21:12:33.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I52105f6cbf7529ce107cabf8a38a10525be7435c","subject":"magnum: Add helm binary for magnum-cluster-api","status":"MERGED","created":"2024-12-12 12:36:47.000000000","updated":"2024-12-16 16:08:33.000000000","submitted":"2024-12-16 16:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937619","meta_rev_id":"0260d728cc715b85d6c38758a43b9e556fa452de","_number":937619,"virtual_id_number":937619,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-16 16:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-16 12:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-14 21:12:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-13 13:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-16 14:19:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-16 14:19:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-16 10:12:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 13:09:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-12 14:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 11:11:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-12-13 13:28:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-12-16 10:12:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-12-16 12:10:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a64b1acd547f63083b2758496e26af30920b5b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-12 12:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"725e1c2227d8a97996dc943b736635d2140475d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 14:14:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b4907abf1394a9fa9810f44fe396625\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/74d7aac5a15148a79d0a0d718169783f : FAILURE in 27m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/902e8d6c76e346b1ac042c2ff1528f80 : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1daf122c6a9f42f4b8efe6d575e56e56 : FAILURE in 15m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bbbdccd1ce0244a09639a2709bb06cef : FAILURE in 23m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ecc3c8ce77344a1abcb67133bb222646 : SUCCESS in 1h 03m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf8cb280f882453a871ffbb8d3e23b58 : SUCCESS in 1h 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1738c11f3ff946cabe1b92bae5c3ccf3 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f91e32183ed0438db8e6690be23e8201 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ee61ac3d78447968e6bb845890fd986 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddff1897422246169fe056ebc29ed890 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f801f93e88cf4aa5a51c748258af19f3 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ed889a32387645f78906f7ed89e5e1b4 : FAILURE in 19m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2ab06163ef342318834bc788424001b : FAILURE in 35m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bbd9e586aa73428eba4ef1ff79a13ecc : SUCCESS in 36m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e91bacb7197446fb89fd319e032075c9 : SUCCESS in 1h 24m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/48940e9e6abb4d63b80b39bbda5db0c5 : FAILURE in 14m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1965415164ac4216a60a5a1d50a74ef0 : FAILURE in 26m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c9d20e02e2845968651ed992e337c95 : SUCCESS in 1h 10m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a18d08516b1c4ec8b1624d086967d5ae : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"94d0bef1dba1fa3e9c15490c942a0728a7ef3a8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-12 14:59:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"559025f818effee2e9be57a2a4dff94d6eecb631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 16:35:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365c6407f37e433a80ec8665479e34ae\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/441cc687bacb4604a510c217e5c41520 : SUCCESS in 33m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d916f0762cfd4762a9b24a2aae5f0618 : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/49a7d2bd7924496eb4f7e9b36c0d0f53 : SUCCESS in 32m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/79c15e3c94784f8fbdf0ee04d249bea4 : SUCCESS in 30m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8fd7595440b4f58abc0114ec37cc6dc : SUCCESS in 1h 07m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bc8c075059a044bcb05d224fafd08352 : SUCCESS in 1h 30m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1855f25a28bd4d618b7b7dd1b81a5817 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/420757d8198b4120bf5960b970c77969 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ecda2953c394625ad21db177f550c40 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5cda1a2a72d460ea584e7a652dde07f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5a1a5fe7334d44a49825c111875ffb : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9577f6e4f365421d9ce84d3459269c01 : SUCCESS in 27m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cad8d7cd69934d04baba19377f492713 : SUCCESS in 37m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/911e9e14b06c4888b2d179524af02f15 : SUCCESS in 1h 01m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ec715556e06e483ea7fc49d7ee7ae317 : SUCCESS in 1h 19m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5dec203daeb24689b438d9c6ff7f6c73 : SUCCESS in 31m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/87ebc4503690424582c0176302470d4c : SUCCESS in 25m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ce95b3bf874b44e198476779fee40353 : SUCCESS in 1h 04m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/20b2ffb841504b68bd52a91aca247324 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6f5c4bd2f2e8d285ff0579977d4a94515b125d38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 18:18:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5711d30f554fbea2776f3a2996723d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/294300b3f7454453bd3c16e2981b869f : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1d7695d468d474caaa1786d41a0cc4b : FAILURE in 2h 19m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/405c467f438944ed94f0ceb1f2b32c34 : SUCCESS in 2h 48m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d7a5da727544bf198d9e0b3ac1bdd99 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66e7269d2e3d4d9b9506906ae7c61e69 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3eee4aac35fd4d9e935d55c208c065c0 : FAILURE in 1h 19m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/356040728f9f4af5bebf099e61703b33 : TIMED_OUT in 3h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8858cd92945eead867a2365c12bc645d76f5a476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-13 11:11:00.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"36b26d841ea12c81cc91411f261ae167876961ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 12:49:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21bfa2161fa8448389c78ad9016d89f3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/225e696ede184d4dbabeb86ea7e444c6 : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2cb3d7b4320a40f09b1a1d0790e27c8e : SUCCESS in 52m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ee204e01d57b4b808ad90252e4e33695 : SUCCESS in 19m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fdff8334aa374aa69f24140579895546 : SUCCESS in 34m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d0bbc1f14ce1463fa776bcbeb57b707a : SUCCESS in 1h 11m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/92ab0c522bba47d48045acc8749ec4ae : SUCCESS in 1h 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00faa62eb46b4e3ab0f6061fe4427f96 : SUCCESS in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445d971602b24e7b983c26a6eaf7fd0a : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd4fc55fdafd4dac96a9bdf463dad67d : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aed8ba6abf81420fa71fece03ec6d5dc : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9694c6e6955b406f940450049162b1b9 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/96ae0d4112464dd8afce02ceca10df1c : SUCCESS in 20m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7dbf24c7e8b4df7b5e187cae3d75856 : SUCCESS in 35m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9c68fc608774a62853a67630af078bc : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/de157451a01948918cef1ea62ed896de : SUCCESS in 1h 28m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/da0abee731914197b7bdc1130c3202cc : SUCCESS in 23m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5c8aca19e35b429091b612e9d2ee9f0a : SUCCESS in 32m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d24b751e6273422ba13e147d732e9c2d : SUCCESS in 1h 04m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/29ef15ae438b46a09e46c27174b0b1f1 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f2a74948568c2896609428b57a219a025474ec00","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-13 13:28:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"756adf3e51c6a7c5bd9ffb84be10da2bc0a63deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 14:14:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63493d2203de4290916a152d5265f51d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/261f0069f5f04ecba0ade7d8949c8512 : TIMED_OUT in 3h 01m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3e50b3dfbb34d928647dce0073e4403 : FAILURE in 1h 07m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d0e9d1bce08440a3a0fc8bf9b7b383c3 : SUCCESS in 2h 27m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57baf9086f6c417e959b82f33ce8c601 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5613a0ce954143169f0fe5a01df8b480 : SUCCESS in 5m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/81af09613d1f40a3869321b37c91a3fe : TIMED_OUT in 3h 02m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/989bd1bca9fc4e178dac3915a556e9d8 : TIMED_OUT in 3h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"32887927d1574934d2e99fd16810334ef6db15c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-14 21:12:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"78a72dc0530c85eb1371679129844847925d0f14","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 10:12:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2b63e1b2c586024f241d0f6869391e3660f5659","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-16 10:12:33.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"57beb1c824ae4048e2b6231aa5a0a62174fd614c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-16 10:13:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efa23db6e12d7948e04edb35ab44f0e735f4e370","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 12:10:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30223d43cca90421cb3073ba789a92235d11cf18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-16 14:19:15.000000000","message":"Patch Set 3: Backport-Candidate+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b02dc35d7bbd27bd5c2408550f262707d6c0f489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 14:19:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fc2790c622cbbf1c0c1ce820b802eeefe1311aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:07:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d72bd03205d4485a97a86da5406e633\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f8df339d77d49e19ef0bfd0b254f129 : SUCCESS in 22m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8b4ce411e669484e89dbabe08a127810 : SUCCESS in 33m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a0c020b335f54e718baaded260889b51 : SUCCESS in 41m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/58454e852694458883ef27de51a8c3b2 : SUCCESS in 1h 41m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549b1f5856c94631a3cd98c50ac122cf : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f57b8397e6bc4639b8f3a6634c051378 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b47ac2fd786470cb464d8c3079f781b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b82d1ede3bd45b483ccde551e0fd56f : SUCCESS in 3m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/96d3a327b8614c35902d55b7dd6dca97 : SUCCESS in 30m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e2940e181de34e5caf34387c244d44f4 : SUCCESS in 29m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7331b998e48f487cbc2278a8ecf70089 : SUCCESS in 1h 12m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5dfbec04a9e34ff89cefe007a64c14f0 : SUCCESS in 1h 23m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a420279158f14b5fa352296177e4c6d5 : SUCCESS in 17m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c307959dbb2b42c5b6a294af35cfaa79 : SUCCESS in 31m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/379114f0625b4d2ca91d14cedfb2260e : SUCCESS in 59m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ff9ef775d8dc4a43b0fe299cabaebdee : SUCCESS in 1h 30m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f7a08457c0b4e2776c10e8b7edda7bef6b79601d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:07:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0260d728cc715b85d6c38758a43b9e556fa452de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:08:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0de775ab05a549208ffb7a57e89688cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8dbebe27e24f5895eba64c1ea4f692 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"592f341dd2e90095cfbd704ff7d66535da851d6f","revisions":{"9f020c73631fb90542ee6256f033137c74ee4b34":{"kind":"REWORK","_number":1,"created":"2024-12-12 12:36:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/937619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/937619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/937619/1"}}},"commit":{"parents":[{"commit":"27f31009cb1725f373376817d4f4f6a3092ecf78","subject":"Merge \"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f31009cb1725f373376817d4f4f6a3092ecf78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 12:36:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 12:36:45.000000000","tz":60},"subject":"magnum: Add helm binary for magnum-cluster-api","message":"magnum: Add helm binary for magnum-cluster-api\n\nChange-Id: I52105f6cbf7529ce107cabf8a38a10525be7435c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f020c73631fb90542ee6256f033137c74ee4b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f020c73631fb90542ee6256f033137c74ee4b34"}]},"branch":"refs/heads/master"},"35a17cf0a7edafb1d38edfa253d71f09475a7165":{"kind":"REWORK","_number":2,"created":"2024-12-12 14:59:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/937619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/937619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/937619/2"}}},"commit":{"parents":[{"commit":"27f31009cb1725f373376817d4f4f6a3092ecf78","subject":"Merge \"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f31009cb1725f373376817d4f4f6a3092ecf78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 14:57:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 14:59:01.000000000","tz":0},"subject":"magnum: Add helm binary for magnum-cluster-api","message":"magnum: Add helm binary for magnum-cluster-api\n\nChange-Id: I52105f6cbf7529ce107cabf8a38a10525be7435c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35a17cf0a7edafb1d38edfa253d71f09475a7165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35a17cf0a7edafb1d38edfa253d71f09475a7165"}]},"branch":"refs/heads/master"},"592f341dd2e90095cfbd704ff7d66535da851d6f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-13 11:11:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/937619/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/937619/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/937619/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/937619/3"}}},"commit":{"parents":[{"commit":"27f31009cb1725f373376817d4f4f6a3092ecf78","subject":"Merge \"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f31009cb1725f373376817d4f4f6a3092ecf78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 14:57:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-13 11:11:00.000000000","tz":0},"subject":"magnum: Add helm binary for magnum-cluster-api","message":"magnum: Add helm binary for magnum-cluster-api\n\nCloses-Bug: #2091608\n\nChange-Id: I52105f6cbf7529ce107cabf8a38a10525be7435c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/592f341dd2e90095cfbd704ff7d66535da851d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/592f341dd2e90095cfbd704ff7d66535da851d6f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
