)]}'
{"id":"openstack%2Fkolla~937734","triplet_id":"openstack%2Fkolla~master~Id81ce23fed7c19211aa8b15abce185d990be8389","project":"openstack/kolla","branch":"master","topic":"grafana-remove-angular-plugins","attention_set":{},"removed_from_attention_set":{"35511":{"account":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"last_update":"2025-04-22 19:18:03.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-16 10:12:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Id81ce23fed7c19211aa8b15abce185d990be8389","subject":"Remove deprecated AngularJS plugins from Grafana","status":"MERGED","created":"2024-12-13 19:16:28.000000000","updated":"2025-04-22 19:18:50.000000000","submitted":"2025-04-22 19:18:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937734-grafana-remove-angular-plugins","meta_rev_id":"34ad435d869efa1fca5820198c63ff417d791d0b","_number":937734,"virtual_id_number":937734,"owner":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 19:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2024-12-17 12:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2025-04-22 17:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-16 10:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-16 09:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-04-22 17:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":-1,"date":"2025-04-22 17:12:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 20:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-16 09:44:22.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2024-12-16 10:11:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-17 12:00:43.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-04-22 17:12:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"874821706aba094829ed7b96e154b327f71d6d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"date":"2024-12-13 19:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f506b62ccfa14d0017a64de78d845491d1cff64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 20:26:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b794acfa8b03422fa90b467c05c0b9f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b39de877425d402999e3fed447b76616 : SUCCESS in 26m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/79ce9af9c2294f0988b37a7c465d10d0 : SUCCESS in 55m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d355ce6e00fc4b3192ee959fd9599ba5 : SUCCESS in 18m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d4cb1ee90ab4a768482e88ee81173fb : SUCCESS in 23m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/512e8774aa7142d3baccb812e218097f : SUCCESS in 50m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4c3b0db7e5c4484da33206c872f844fa : SUCCESS in 48m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a945a85205f410f8283d759809cffaa : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f139861c9cd425389a13b0e273b3c67 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/672d038ab17e4bf3a0453ef407018273 : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8476a1614214c5f9fc8a31057bb76b9 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45240de0a584216b1bf06b4bd648382 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f60b203e944e298c5655418a6b83a0 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/369fec17260546cc8a89547bb4a8f082 : SUCCESS in 21m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5109b21dd0a64e158ff9836637d93a37 : SUCCESS in 29m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dbf943e0b15248f1bdc300bd021c2616 : SUCCESS in 53m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c141c32b05f478ead0e914868b92064 : SUCCESS in 1h 09m 22s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6eb39c07f99d439d8f2d95c5a592d68a : SUCCESS in 34m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fd498517809b4cdda0d22cde91197dac : SUCCESS in 13m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9b5d0dfd28e8446cb385dc212d93fd68 : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1159c42f75b44ac3bb39ce08ef62268d : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73d4edf7062343fba6131da60202373e : SUCCESS in 44m 40s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/796774eb3a26475aac52c567b0d0c1e2 : SUCCESS in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d3d94492779ac0c0a17fdbd507fc7224730b355","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 22:23:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99498102ac494569a536a616e248b998\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3f120f57aba64f06861bce016056e9d4 : SUCCESS in 1h 46m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/21c1bc1c9b2841bd8d02dd18b495b7c7 : SUCCESS in 1h 39m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9a1eeff2bba542bcbc351be0f9772aea : SUCCESS in 1h 42m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e64b7070332e48af81e55e3fc39f0385 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2379ecba5624b6ebd15a09156f17439 : SUCCESS in 6m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0098e01a76084f358c8dd21f2e9b46f8 : SUCCESS in 2h 55m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/77ad1e58b4e44c3c9d7ac5ff6b456922 : TIMED_OUT in 3h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31086a7a8d8ff06bbe4b4b13f629ff134a66ea5b","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-12-16 09:44:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b15ed858aeab517884900985de88ecc4b2aa4cde","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-16 10:12:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"39a92bff25a4c5730def13dd5f8af18fd358847c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-12-17 12:00:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"757e27a3c9e023a8ebbab99140acef6616c71b93","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 17:12:36.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7403d5f7d39fc8e6d792d3822bf643f5798066f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:12:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb79d2620c6741cc1c039b141f6df9aa4ac3fb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:18:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/964a1eae8918470ba75878c716b0ba0e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0bfa2e426e7c4755aef16094ad7dca4d : SUCCESS in 20m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/52c5510c25da41a8b1ea4d57d8ac7742 : SUCCESS in 25m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39db1e80683d4f348aca6a908accaf18 : SUCCESS in 1h 02m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7898a5b1b2ee416986f5df513cdfc0bc : SUCCESS in 1h 35m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0de94c7bf104f089fe1f3798bde15b9 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef04d2c873ab4329b210e87d8d1adde4 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85985a9c20f149cea7427692439b0e14 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06596f2a94b2460faa5a75d3e46bfd0a : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd60ace5be2b4144825b7a05b7c47c7a : SUCCESS in 3m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec6ff7fde5c84b4da019df6e4d2ec6c6 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5598a0cae27d48afad8d93f37eaeb0f8 : SUCCESS in 13m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ef4e22745c104b789060c650de6312dd : SUCCESS in 1h 19m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0353ebb8d936435fbf59f33c9f076cc6 : SUCCESS in 1h 05m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d25c8016b84d4943a3b685a8df31514e : SUCCESS in 1h 29m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/236681dc176c4b2b8ac1d7bfd8719807 : SUCCESS in 21m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/80a41df94dc64f609fd8101685c815bb : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3bb49efc2955400db448da310c307eb7 : SUCCESS in 1h 04m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ffd5a3429b7e46d496b7cb2ceeae312f : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":1},{"id":"09ac6089a11dbb8578cb61d5f35800112b405c40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:18:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34ad435d869efa1fca5820198c63ff417d791d0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991c55f86ac2476898603b60c97df780\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5418b379d7eb4b3d873f28436c664653 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634ff352eb7448c69c16a0ee9c9fd258 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3575d7d8bd14feef50e3e1222557a6a07158fdd4","revisions":{"3575d7d8bd14feef50e3e1222557a6a07158fdd4":{"kind":"REWORK","_number":1,"created":"2024-12-13 19:16:28.000000000","uploader":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"ref":"refs/changes/34/937734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/937734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/937734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/937734/1"}}},"commit":{"parents":[{"commit":"27f31009cb1725f373376817d4f4f6a3092ecf78","subject":"Merge \"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f31009cb1725f373376817d4f4f6a3092ecf78"}]}],"author":{"name":"Dawud","email":"dawud@stackhpc.com","date":"2024-12-13 19:13:08.000000000","tz":0},"committer":{"name":"Dawud","email":"dawud@stackhpc.com","date":"2024-12-13 19:13:08.000000000","tz":0},"subject":"Remove deprecated AngularJS plugins from Grafana","message":"Remove deprecated AngularJS plugins from Grafana\n\nChange-Id: Id81ce23fed7c19211aa8b15abce185d990be8389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3575d7d8bd14feef50e3e1222557a6a07158fdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3575d7d8bd14feef50e3e1222557a6a07158fdd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
