)]}'
{"id":"openstack%2Fkolla~937837","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I52105f6cbf7529ce107cabf8a38a10525be7435c","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-30 11:40:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I52105f6cbf7529ce107cabf8a38a10525be7435c","subject":"magnum: Add helm binary for magnum-cluster-api","status":"MERGED","created":"2024-12-17 06:09:19.000000000","updated":"2024-12-30 11:42:33.000000000","submitted":"2024-12-30 11:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937837","cherry_pick_of_change":937619,"cherry_pick_of_patch_set":3,"meta_rev_id":"645a74772e914f703bbb417a24eda550c40191d2","_number":937837,"virtual_id_number":937837,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-30 11:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-17 13:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-18 12:19:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-12-30 09:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-12-30 09:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 07:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-18 12:19:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-30 09:37:54.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"430882f8ada614c2e18ccaf85425fd139ba66fc8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 06:09:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/magnum/magnum-conductor/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"ae67d6fa6afbee6e1014ee91a83a637ee5d3c508","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 06:09:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b965a59371b4650c1347baecf9171a5796726a1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 06:10:03.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"016f3a6a4abf78234c78ef50cb1689607a76eb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 07:42:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41ff70412eb40c7b02370e9ed7330df\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e5ff51779349420387dfa631716b06a5 : SUCCESS in 26m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eba0f0dfc8944694b5ccb269074fb937 : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41d45408bc4b4ea3b79e32a2757ff634 : SUCCESS in 21m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84cf758c7bc544558d711948cd5e1cf7 : SUCCESS in 17m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/18e695e171ef45c081a8bb37b5b450d8 : SUCCESS in 1h 02m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a86f1fa961d4878ae7664a588c9c93d : SUCCESS in 1h 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d86a95bef64d6ebf6d0b7313ee6e5e : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a9fbfb807524eea85b06ba53fd4fc8b : SUCCESS in 2m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51b2c82b9bdd4ec6b2e438f2c13d1d0e : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f268db926c0749d89f6427e88e0299fd : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b9e0121a77f4b11bbef534c33f537b5 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3127bdde99a842f9bc050d2c2fdfd25d : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/66470983c02d4ff28e5dff069abc86ca : SUCCESS in 24m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6642f50a98f346438df121ab10f086e6 : SUCCESS in 30m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6bed637c1e8c40a68fcaf734a39604d0 : SUCCESS in 1h 07m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6acfb1fe64b14ca9b9d5df43262b6d6b : SUCCESS in 48m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a22c8008a5a040eea59b1c22ede68ed5 : SUCCESS in 26m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c94665554d744700a3913b2cd669ced5 : SUCCESS in 30m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89894da52edf46c3b202da70b56873bf : SUCCESS in 1h 02m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/02379383d94e466fb27b3ff133c0e9ac : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":2},{"id":"99b8f7ee6b35c1002e48231aa17183cf54ff546e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:34:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2df31c57a4da4aa2b348efdf54566916\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c94bf7d9e88f458e8ea3ca97a3cdc5e2 : FAILURE in 2h 02m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e57e252755dd4909bb4acdfabc998e3d : TIMED_OUT in 3h 01m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/813771766b344c269178b70ae726189e : SUCCESS in 2h 51m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de50e4e0fbdf46e29642bdce6f1d0000 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15ca7cf3616e4662b6fdfec025e2d4a3 : SUCCESS in 5m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/df350035baf8490196de2803a039fc50 : FAILURE in 2h 33m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/245fd8a28de044ba970cc1778dd21ed8 : TIMED_OUT in 3h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8415162f343a040b58d60f4c362fc762b0e0b882","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 13:43:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9c5ca83331f7f2862915943e5847117bcb329b6d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-18 12:19:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c54e977982a8fe0196b40b16110bafacda82cc6d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-30 09:37:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44dfe264063b2480f0e9cd70cb7f11bf526f4a7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 09:38:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f73ef1890646efc14442f3e8dd4f96a538bfabbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 11:40:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc392b2c91d49bb86821ae2536d751f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aba67ec8b5f14580a86a40d507dc2605 : SUCCESS in 29m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dc573004451b4188880530043b205b6d : SUCCESS in 26m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/46ac41f3a87d4a63a11341ca436ba16d : SUCCESS in 59m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6fcd7388ad44f7dae8758d592419018 : SUCCESS in 1h 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fb95a9a52c41bf98520f9a4a292351 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d5dd8e218b64d22a6dea88f9c67e51b : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f845483381da48de936c2e641709f962 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a627774c1fe442baa7c351843c3e766 : SUCCESS in 5m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2f9646814b42406dbfd1667b6ddd3d0f : SUCCESS in 23m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3691699b243f443fa088c04e8985225b : SUCCESS in 30m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a534d2dfd3f94e1291dad3eeb6f44d2a : SUCCESS in 1h 09m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e7a1111df8de4289b4b1460a56f7c96b : SUCCESS in 1h 28m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de87be6acb2242da880e66bdf6f606f0 : SUCCESS in 24m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5f4843530502466697d2988958f83be4 : SUCCESS in 26m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/52e2c65ff7e94627a24074e1ea28ea0c : SUCCESS in 1h 02m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/59130d76762a4341af270df51946d478 : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":2},{"id":"93b068a865083b529f42541ce8b35fc69f8eefda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 11:40:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"645a74772e914f703bbb417a24eda550c40191d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 11:42:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66cee1c6db0a43f0b09d428f2b0e57db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6343e2d070480693bf9f3321315fe6 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fe38110b4907ab6775a2a6345c20151e5a5dfff","revisions":{"032073366db81c9bd9a179d977f4023e3458bd64":{"kind":"REWORK","_number":1,"created":"2024-12-17 06:09:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/937837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/937837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/937837/1"}}},"commit":{"parents":[{"commit":"fe63120941a32654fa093517f81735a56b2bb607","subject":"CI: Stop using LABEL for ephemeral0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe63120941a32654fa093517f81735a56b2bb607"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 14:57:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-17 06:09:19.000000000","tz":0},"subject":"magnum: Add helm binary for magnum-cluster-api","message":"magnum: Add helm binary for magnum-cluster-api\n\nCloses-Bug: #2091608\n\nChange-Id: I52105f6cbf7529ce107cabf8a38a10525be7435c\n(cherry picked from commit 592f341dd2e90095cfbd704ff7d66535da851d6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/032073366db81c9bd9a179d977f4023e3458bd64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/032073366db81c9bd9a179d977f4023e3458bd64"}]},"branch":"refs/heads/stable/2024.2"},"8fe38110b4907ab6775a2a6345c20151e5a5dfff":{"kind":"REWORK","_number":2,"created":"2024-12-17 06:09:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/937837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/937837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/937837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/937837/2"}}},"commit":{"parents":[{"commit":"fe63120941a32654fa093517f81735a56b2bb607","subject":"CI: Stop using LABEL for ephemeral0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe63120941a32654fa093517f81735a56b2bb607"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 14:57:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-17 06:09:44.000000000","tz":0},"subject":"magnum: Add helm binary for magnum-cluster-api","message":"magnum: Add helm binary for magnum-cluster-api\n\nCloses-Bug: #2091608\n\nChange-Id: I52105f6cbf7529ce107cabf8a38a10525be7435c\n(cherry picked from commit 592f341dd2e90095cfbd704ff7d66535da851d6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe38110b4907ab6775a2a6345c20151e5a5dfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe38110b4907ab6775a2a6345c20151e5a5dfff"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
