)]}'
{"id":"openstack%2Fkolla~938185","triplet_id":"openstack%2Fkolla~master~I46da8fad93febbc462cbe42a0d96ef7d423b8965","project":"openstack/kolla","branch":"master","topic":"kolla_uwsgi","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-29 14:23:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46da8fad93febbc462cbe42a0d96ef7d423b8965","subject":"cinder: Fix log dir permission","status":"MERGED","created":"2024-12-23 11:36:42.000000000","updated":"2025-01-29 14:24:51.000000000","submitted":"2025-01-29 14:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938185-kolla_uwsgi","meta_rev_id":"80deb37df0057157bf18b4e0a583c4e3482df810","_number":938185,"virtual_id_number":938185,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-29 14:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-01-29 12:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-23 15:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-01-29 12:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-29 12:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-01-29 12:31:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-29 12:31:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-23 14:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-23 15:02:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-29 12:31:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-29 12:38:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"4e42657eceea3e213ef7ec0ef05ca7de1360393c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-23 11:36:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3affa03ea49e191953d2e08670efaa164fd38d95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 14:28:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19ba0000dbb4746b6048ea8b81d11f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/82fc52264718484caf711d4b49dbfa96 : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1a1c575ccd104c19bcc6209d9e6c512b : SUCCESS in 39m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0aa3fa4c6a104dc49d4a88e85a05aad4 : SUCCESS in 13m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/50662b44472c40239c8a92d2b8791fd0 : SUCCESS in 28m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8edf5a2461054d0bb68b6590d06f8f05 : SUCCESS in 1h 07m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/944d0c292876497e8706cf307cb06f5a : SUCCESS in 1h 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9559642bb14a1cb2347be6e16ebac0 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/923cf8a7cbe64117a1daa3e005bc56da : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94629258a2054d2e950e3a3f857cae4f : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930176b79aec45b4a8117cb380ac6afe : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387a0af341804168ab744aafec320073 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cf1850545f554542933e23633b1fd02b : SUCCESS in 28m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/77888c9566b149ce80fbcb3d6287380f : SUCCESS in 35m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e8002d4cbbc0400d83c30f832eca6049 : SUCCESS in 1h 00m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/339688f3db8c4f318452e00cc58fdbd0 : SUCCESS in 1h 34m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ce4adf66e3f949a3847096f258bdb260 : SUCCESS in 1h 54m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/83f433464ae646aca91ed70ed9eff7cd : SUCCESS in 2h 43m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e2b73d8c4d6b4763b063fe9519ed8d37 : SUCCESS in 16m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/61832c27d8274950838b9567245741ab : SUCCESS in 22m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/478a5919a6124999a8247f53d15d92d0 : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0ce9e8a4cd244f6fab947e293033d6e6 : SUCCESS in 57m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9013c316525348b585bfe6f3f7df0b1e : SUCCESS in 1h 33m 25s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/21c2a12e50a349f286fa41a6ed9fd857 : SUCCESS in 2h 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c93363013ccb35c5271b086d9fe2c693f15407e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 14:44:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d45c44b32141168c869e72d29854a2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fdb440cea55047b888372df20e49ae92 : TIMED_OUT in 3h 02m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd0fda918fc645b38a242fca643b08ce : FAILURE in 1h 42m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/174e4f88376f4d23a42a31ea2d17e4d0 : SUCCESS in 1h 44m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/360f139c92d7468db78491b2ee6d0b2b : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/caee6e0a3251483f9a70ae5be583746e : SUCCESS in 4m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a0999291fdcf4961975c53799148f5d8 : FAILURE in 2h 34m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/32eff199ddd344669cfb8513988daffc : FAILURE in 3h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdd44cbcdb97d0200d726f0fb93b59f4c7bdb09f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-23 15:02:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b12ab7d9165dda737ed535cb4b0c187242c66159","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-28 07:07:16.000000000","message":"Topic set to kolla_uwsgi","accounts_in_message":[],"_revision_number":1},{"id":"894d9106f728c1b2cc40fffe8dff5f66930c2e36","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-29 12:31:37.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ada2b2103cf5873c869c72019c8e84014a8a998f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-29 12:38:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74da4d19c8449f5fd436b37ed5844e747489c52d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 12:39:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eae617553da759dcf1d4a4a5b38d9e44557985ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 14:23:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee374573374444fa0ad269c92c1bf5e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/24759d95ae514a9eab67dc7e9a72b8a5 : SUCCESS in 13m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/073455e414a844ffbf9093e5c0deea8b : SUCCESS in 18m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3491c42440f04b7bbc9e1d62fb43228c : SUCCESS in 1h 37m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9440bcd8e11f447abc4c274dfbe4d17b : SUCCESS in 1h 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd090a6478743d1baaa09352f11ed3d : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0410ae5fc24446978f2cf301f92b645f : SUCCESS in 6m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/391291b1c99b46b399b25eee7df35df2 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9795a0e62eb2449d9c44d38391b52bad : SUCCESS in 4m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e702fd17ceaa4fd29911d2b002817a6c : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fc8b2376fcb74dbf943431c5045a7219 : SUCCESS in 37m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/34dab439805c44e99ba9c6324fa493ce : SUCCESS in 44m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff7db51821d94341a473fd084fd1b857 : SUCCESS in 1h 04m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/42873cb851684afa92931e4f024eca98 : SUCCESS in 51m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4550516234dd4fa587e080e891d81c65 : SUCCESS in 27m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fa04a81539714f9082465ae83501b834 : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/21fcc80410ee4961828c7ae08961bc53 : SUCCESS in 1h 41m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5895ae000873478297c13c26faabe8b2 : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9e9dd0904fd6a32c994bfc462da0c70f2b61e326","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 14:23:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80deb37df0057157bf18b4e0a583c4e3482df810","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 14:24:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162b92deb5614340bd1c3557d06bd854\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb49d4e0481643f18be1c53766a5c7bc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"969cbc736706b42ef1a6bc1922c2b7b2ec95506d","revisions":{"969cbc736706b42ef1a6bc1922c2b7b2ec95506d":{"kind":"REWORK","_number":1,"created":"2024-12-23 11:36:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/938185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/938185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/938185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/938185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/938185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/938185/1"}}},"commit":{"parents":[{"commit":"3cb7bea93be35bef0f4a93cc874cbc422d4fe084","subject":"Merge \"Move uwsgi installation to openstack-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cb7bea93be35bef0f4a93cc874cbc422d4fe084"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-23 11:35:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-23 11:36:40.000000000","tz":60},"subject":"cinder: Fix log dir permission","message":"cinder: Fix log dir permission\n\nIn order to add support to uwsgi (which is opening the log file\nafter dropping privileges) we need to set proper permissions\nfor /var/log/kolla/cinder directory\n\nChange-Id: I46da8fad93febbc462cbe42a0d96ef7d423b8965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/969cbc736706b42ef1a6bc1922c2b7b2ec95506d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/969cbc736706b42ef1a6bc1922c2b7b2ec95506d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
