)]}'
{"id":"openstack%2Fkolla~938402","triplet_id":"openstack%2Fkolla~master~I38688e442b995d38e9826fbb545b6ab09e06bdf5","project":"openstack/kolla","branch":"master","topic":"kolla_uwsgi","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-25 06:42:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-10 11:56:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I38688e442b995d38e9826fbb545b6ab09e06bdf5","subject":"keystone: Remove ARGS to support uwsgi","status":"MERGED","created":"2025-01-03 14:33:40.000000000","updated":"2025-06-25 08:13:08.000000000","submitted":"2025-06-25 08:11:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938402-kolla_uwsgi","meta_rev_id":"c60ba0c5160f7a23f11591b8e0721a44748a9d6b","_number":938402,"virtual_id_number":938402,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 08:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-25 06:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-10 14:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-09 12:20:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-25 06:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-06-25 06:42:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-09 12:06:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-03 16:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 09:21:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-09 12:20:26.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"85e1da273118c378cf2d8b4e4cdfb018e6e79d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-03 14:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"139ba39ca829698bb7b29cf8e2d8e90583ae4592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 16:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11a508afff7f40898908e6d3d7175c9b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9a7c2b1043144f1cb985b073c6da30c4 : SUCCESS in 16m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e8d2fe93d74745fc8924fcbea27fa144 : FAILURE in 27m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/24c6eb5c690e422c8817bbc1f3930d4f : SUCCESS in 23m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/44e024443c1f4958a5ef10f542f56b97 : SUCCESS in 29m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/61b44e0cc7c74a1ea72bb0a23780823b : FAILURE in 28m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aed4ec3982df42d9bdc6ce6f51ed1f16 : FAILURE in 1h 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d5cd205a1e246c29990a4bd09cc9115 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a49a71e5315e4a7f89c4fd5fcbf9c72f : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/071f7402e53848aab50764588bebdc84 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63d414b9eadb4a4b93ce27dc8532e84b : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6285b86e079f4efbbdc07280fc718b53 : SUCCESS in 3m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b7789e47d0714cabb1cf104ffa2abf8e : SUCCESS in 23m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/71b5c74616df4a669cc8c3daad74805e : SUCCESS in 34m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cad08801fd4c4f26bdea264d445dcc60 : FAILURE in 32m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/108949d1c547467c80abf6aa5811e2af : FAILURE in 1h 13m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9f347e59718c45229370fc95867befd3 : SUCCESS in 29m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/719b288546a448b2b744cc2fd96861a4 : SUCCESS in 34m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e0a4651a5b3f4c37be7ec4ccd53bd4d3 : FAILURE in 29m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3cd09c5c754e4114a296b35654aa09a4 : FAILURE in 1h 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"09cda1e44cd6debaa632666ac946972009dcb7fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 17:39:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdb50de4f9a48dbbd59f7897a0fd974\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c25eabd120e147528e0e853aaf68bc47 : TIMED_OUT in 3h 01m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bbb46a91f58a4bf4b7e011f03a7d3d12 : FAILURE in 1h 58m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f96ae68d566642dca6d60bffb79940f8 : FAILURE in 1h 14m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fca057e6df414d5bb0e7080b5fecf2c8 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5849a1d58f34465990ba04c1edd702b7 : SUCCESS in 4m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/921e096aba10458b97ffdd36f306b672 : FAILURE in 2h 55m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7bdc5efcd86c4d93a0fa4363dd3ced62 : TIMED_OUT in 3h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56c6e5bd80b89b975dff5a50bbc49169aba85c66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-10 11:56:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a9f745d825b49d1502393a782e49becb82c38aba","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 08:56:11.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"628986c2915dfc67fd975564a3a9e6e90d0b7069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 08:56:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49f44ffab0ac27ecda424974afa502096fcb20ad","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 08:56:38.000000000","message":"Topic set to kolla_uwsgi","accounts_in_message":[],"_revision_number":2},{"id":"5f38e5f254046f7a1dc350df62d1538cf4bc3150","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 09:38:40.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/952093 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"9671da5615a917adf52dce060f0b42e93d8b9b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 09:38:41.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/952093 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"fd3a481cc218f7e1dada4fbf219df1a376b04296","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 09:40:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"528e8d2b6b77c4f7c1154322cb59974865859109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:14:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca45b488462a4bf2a1924b8142acd756\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ee7149bf3274c0bb61216e727aceb74 : SUCCESS in 19m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d076e48ee9b746b2a91f9bbf5c87bcf4 : SUCCESS in 19m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fd76532530d548fa942747c4caac72db : SUCCESS in 39m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b586d9b98a6c453aa4e3bc21b39c6ee0 : FAILURE in 1h 33m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ddfc5e6929444e826a06300aa1561c : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fd3ec4a532b4ef084f6bd4222230303 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7542d1deab443b78eb97765ec159189 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37191be1a2ef4f92be61e6954ad488cd : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1170bb25bd284ad3b4cbd62cb67256ba : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a796421ac394830baa3a2b6ddaf9ef0 : SUCCESS in 6m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/166dad00557345db98c312044255eb51 : SUCCESS in 2m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7186af3954954894b828ae7ff52a2c91 : SUCCESS in 19m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/79e81b9fa4894b68943baffed414865e : SUCCESS in 12m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b7652f54f6704c0bace76ad0d6d20f73 : SUCCESS in 31m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0a7e702ea0d4499da5e63611b0d8cbd2 : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3e81f74678a60cf7cb642e41e60df43c34201554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:18:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e5f254a6584dfaa63c9963451e48ad\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0ca58913465243358a17d74f1df038c8 : SUCCESS in 27m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a7847be46a5b4918a283f44859fba00c : FAILURE in 1h 36m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92204c125ee0413ab2bb7810c0980bca : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3e35e3aba06473c967c94f5a7ed8b63 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c0c90657a0e4289ae458c845dbae1dc : SUCCESS in 5m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/53b69ce5a32745c1b1ffa214a9010a56 : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58bbfab5a22f029ab7db32cbb464604fb1e2131b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 12:06:01.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"88a8ee00906d00d3ecb1bec4e43783bb02382345","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-09 12:20:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c57a84078d807708d3a145d751f402ab790cb0d8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-10 14:18:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4af4f210337e7622c6e29ebdd147cd5a77eee05a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-25 06:42:33.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a51990c0fd7d8f2c45c3f94628b081dc6404726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 06:43:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5008ab967c9e2890f22b0f006d91a3c37295aa7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:11:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abf2261ac4c543269f30447bedd90522\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/31dcfd25c5544bd9a512572b44763527 : SUCCESS in 22m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4925f013f32e4ccca214852dc8a33601 : SUCCESS in 26m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ea2f67024a22460bb9f9017d80b55d94 : SUCCESS in 1h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fca40039a041a595a0e71b4f7388ea : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d758a12bb6f49cabfe21a0103e6274a : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/228d57ecd7094ef48fb7cdcb3c98508d : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f92793f0a74783a002be86811769cb : SUCCESS in 6m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e4ba51ed6a8b41689696eca24dfc0123 : SUCCESS in 3m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e8ebe92ccd3944d2a27e10327b50a947 : SUCCESS in 19m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/32dbab8ff8ba45398dc6468f6d08facc : SUCCESS in 22m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/081a3c4ec5834842beda2f0e50fad4f5 : SUCCESS in 59m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b41d3735898a4571a3b87676210b82cd : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2ef34164105598f77e50a922174084567b7c78ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:11:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c60ba0c5160f7a23f11591b8e0721a44748a9d6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:13:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed21ee34c00f4227a84d3d2c154572f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfda19519018409d8eff14d3c43c4867 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1b54b521487d185813bb4d7d4b4617dd59f0172","revisions":{"3c7232c04774fc9500bd3e8a734d7d8d6d93345d":{"kind":"REWORK","_number":1,"created":"2025-01-03 14:33:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/938402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/938402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/938402/1"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-03 14:33:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-03 14:33:38.000000000","tz":60},"subject":"keystone: Remove ARGS to support uwsgi","message":"keystone: Remove ARGS to support uwsgi\n\nChange-Id: I38688e442b995d38e9826fbb545b6ab09e06bdf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c7232c04774fc9500bd3e8a734d7d8d6d93345d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c7232c04774fc9500bd3e8a734d7d8d6d93345d"}]},"branch":"refs/heads/master"},"d1b54b521487d185813bb4d7d4b4617dd59f0172":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-09 08:56:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/938402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/938402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/938402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/938402/2"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-03 14:33:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-09 08:56:33.000000000","tz":0},"subject":"keystone: Remove ARGS to support uwsgi","message":"keystone: Remove ARGS to support uwsgi\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/952093\n\nChange-Id: I38688e442b995d38e9826fbb545b6ab09e06bdf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1b54b521487d185813bb4d7d4b4617dd59f0172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1b54b521487d185813bb4d7d4b4617dd59f0172"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
