)]}'
{"id":"openstack%2Fkolla~938426","triplet_id":"openstack%2Fkolla~master~Ie3fbfe2ef59a29ed7808826f0b1fa5b8bf5ec487","project":"openstack/kolla","branch":"master","topic":"uv","attention_set":{},"removed_from_attention_set":{"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2025-03-20 09:32:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie3fbfe2ef59a29ed7808826f0b1fa5b8bf5ec487","subject":"Use uv instead of pip in install_pip macro to speedup builds","status":"ABANDONED","created":"2025-01-05 21:27:48.000000000","updated":"2025-03-20 09:32:43.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"20be348a4a252edb3f3a05d237a2dfbf58d584f7","_number":938426,"virtual_id_number":938426,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-05 22:10:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2025-01-06 05:59:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-03-20 09:32:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2025-01-10 09:07:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-05 22:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-06 05:59:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-06 08:18:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-01-10 09:07:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-20 09:32:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"05581ddacb510dda226103d2d691e8cac03fe9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-01-05 21:27:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"290160d6288f108c8446d5a3dad4b8c3d8ca2018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:10:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb026011be2d4dd1b38675943442e29a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/99f1139b51e94c7ab3da9d900f1c4f76 : FAILURE in 16m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d7b5faab7f3a4c2a9d9e70313ec76a35 : FAILURE in 12m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed3ee6cdd12f4d3092c766efcf2e72e6 : FAILURE in 12m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/86c70c043ecc409eae8b52c5c069e013 : FAILURE in 30m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95ffc879d0194090a1b54b803ddf2e18 : FAILURE in 14m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6765174997354c6ca147c4ea3bab315a : FAILURE in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbdb386227a746209d22c6e05a1a20bd : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e801716422445a2af6e5ff2622cdf96 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02053426ac33472884bb6d9cf6a178bb : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1523005a76494096bdb08016434e66b0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb0a56c8d2f45328fb939885aaffd62 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8200d6cee0c0403b8c15963e46c65156 : FAILURE in 25m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a0c1a8ece6a40f18fe02ae2e2a49b71 : FAILURE in 35m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c6a0aac2415a4827a993a2497cac0ec6 : FAILURE in 18m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7d4a399f040f43ceb10ff8f6b7e96c10 : FAILURE in 17m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/20b99e38293e437eb83b1c64915e3de2 : FAILURE in 17m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d515d9ad41e940ef97818d844b594270 : FAILURE in 19m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6e9c4e84947642fe8a00ed76617ce337 : FAILURE in 21m 16s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/afb98227d8bf4e8eb25bf60415a1dcc7 : FAILURE in 21m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/27c21469ef784f0a9b9a3d783446ca36 : FAILURE in 20m 37s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bf73507a5e4d46a89cb2a5be3b07f52a : FAILURE in 21m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e04c76ec34454d178ba813cc016e88d1 : FAILURE in 18m 41s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a60fafa241784c668a36e5bb8b57818a : FAILURE in 20m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/74b1da79c5f244efa32ef4c82a37c27e : FAILURE in 18m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cabc2f744fd4fe6b4e2690292247e70 : FAILURE in 17m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86c0c99c9fb045f2b76cbd6d0c44fcfe : FAILURE in 16m 24s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d7bd5be19725487fa4859e9b473b7833 : FAILURE in 17m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c6831fe18bcf4062b2e41e0e557edbeb : FAILURE in 12m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0d9843c99fcb4f27b13b4b77bfbc5c67 : FAILURE in 17m 39s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/01a0e171e7534cf39feb6c8cbd26cb91 : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2f236794b0f44f3b9b7f386a5fc49df6 : FAILURE in 13m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b23159f93ba9410dacc731d5b3466bf6 : FAILURE in 18m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f698f3877f454e128a76329d989a5bd4 : FAILURE in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ad1c6172f54ca52cf2dec9ad0f7e1f3a163cc94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 00:35:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d51f701e50441dbb8104a7cf2f2ba9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f18aa52e1e6243f2b8ee04312ac38408 : FAILURE in 1h 23m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dcd7f0bdbed34bc1ae9994049b517d20 : FAILURE in 1h 09m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25713670e48b4e80bedd0848ea630464 : FAILURE in 29m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fe92b1c55d949fe84a93ef8fea06117 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f44ef386a244cc7b58a112e3b01be4d : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fac0cbd1a8534619a7ce300d44dc6c41 : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/420fc8b4b2f6438c9a779302b5122bda : FAILURE in 1h 52m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8eabef0d78c4e71b1a39f46e9abcb4befbe5c549","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-06 05:59:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"319b4b382d7bc9188188b2678e439768353015e4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-06 08:18:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98e9a0e07f18d555ad0f91e247a6f2129aa5cbc4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 09:07:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"293b5e41e8c4f5302393aeefa84f976371f0b96c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 09:32:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20be348a4a252edb3f3a05d237a2dfbf58d584f7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 09:32:43.000000000","message":"Abandoned\n\nconsensus to not implement this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96d0bc46e621219d5005b96e76d4508202ecf01c","revisions":{"96d0bc46e621219d5005b96e76d4508202ecf01c":{"kind":"REWORK","_number":1,"created":"2025-01-05 21:27:48.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/26/938426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/938426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/938426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/938426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/938426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/938426/1"}}},"commit":{"parents":[{"commit":"27f31009cb1725f373376817d4f4f6a3092ecf78","subject":"Merge \"Revert \"CI: Pin requests to \u003c2.32 for publish jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f31009cb1725f373376817d4f4f6a3092ecf78"}]}],"author":{"name":"Christian Berendt","email":"berendt@23technologies.cloud","date":"2025-01-05 21:25:54.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@23technologies.cloud","date":"2025-01-05 21:27:45.000000000","tz":60},"subject":"Use uv instead of pip in install_pip macro to speedup builds","message":"Use uv instead of pip in install_pip macro to speedup builds\n\nuv is an extremely fast Python package and project manager, written in Rust.\nIt provides a drop-in replacement for common pip commands (uv pip).\n\nhttps: //github.com/astral-sh/uv\nChange-Id: Ie3fbfe2ef59a29ed7808826f0b1fa5b8bf5ec487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d0bc46e621219d5005b96e76d4508202ecf01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d0bc46e621219d5005b96e76d4508202ecf01c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
