)]}'
{"id":"openstack%2Fkolla~938546","triplet_id":"openstack%2Fkolla~master~Idfc26fe380a546d0fd712a531e5913090b01d808","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-24 15:32:07.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-01-24 15:32:07.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-24 11:51:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-24 06:58:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Idfc26fe380a546d0fd712a531e5913090b01d808","subject":"Fix genconfig","status":"MERGED","created":"2025-01-07 10:59:13.000000000","updated":"2025-01-24 16:18:15.000000000","submitted":"2025-01-24 15:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938546","meta_rev_id":"1adbf168541784b2d364e7e28b8db324c4657fa6","_number":938546,"virtual_id_number":938546,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-24 15:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2025-01-24 06:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-24 06:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-01-24 06:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-24 06:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-01-24 06:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-24 06:48:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 12:42:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 08:33:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-14 09:23:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-23 10:49:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"cc2dc0319f7c31da9d1915ccb796b9fbf20391d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 10:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"470ee3fa1d0c8aec175bd78c65c6443f39d62a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 11:00:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffb18251949c7f5ac2b6af083e72f7ab182b8b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:42:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7463bdfe4f6543af8b9fe7ce486acd8d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a55260bf433c4773b3248fbae7af6a21 : SUCCESS in 26m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0fa9ee8e8f074a0b8058cf2df9638cde : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b722626554fb4bef949ffdeb183320e1 : SUCCESS in 13m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/39b582edb0ce445f82189b7f93e0e8ec : SUCCESS in 25m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19b0f5443d654cf9b2ef7554689fe49f : SUCCESS in 1h 09m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/84847f8f24d042c487dc5752ce8432c2 : SUCCESS in 1h 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f31c70a7a04e0eac76845264ee5264 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b921b43bf2745e2992f0310031ecf53 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff4cfdece78a4a56b73155e3a3795eb5 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c37f46f57ddb4ca8be8d46c3e2fdd535 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04900180b5cd4650bef9af8430eed842 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87940223778042d8a4b1cdd4822ebd95 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/de7c071cb01f416ba9b3051f1fed4e46 : SUCCESS in 29m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b0687c3e5244e94af966844a8e89ed3 : SUCCESS in 38m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8dd0786364fa498a818f98f430e7a1fc : SUCCESS in 1h 06m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1893ff59dc94415093e4db3cfaa7e8c4 : SUCCESS in 55m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ecb2244bcacc496cad0c7aab2887ee73 : SUCCESS in 25m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e9fdfdb8c3314ec487b85bf43ff4abb5 : SUCCESS in 32m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5e544d39a3f64c37a2482ac7f1d9ef02 : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1b53b258947547aba8711eec05230fc4 : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1eeb86b95d7358c0ac7ffcb6ae67b728ff1730ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 12:43:44.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"09d93f66b34cf6ef97a61cce4c2ef5124283fcd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 12:45:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d005e27c5ec99c2f400380482c42a25cd88eca06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 12:46:02.000000000","message":"Topic kolla-ovn-sb-relay removed","accounts_in_message":[],"_revision_number":4},{"id":"8cd8e50b869f37fa5d1639c7b83a2250a5b290ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:32:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9e9263e4e1495fa49fd17693cff75c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/84e472b2b664461ab1ce24fd3c93d99c : SUCCESS in 24m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/040b00d881f34fc6a9963853649183ca : SUCCESS in 1h 06m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/db037479bb1345dabefef43d348930d7 : SUCCESS in 12m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2fe82d87f241452ab165067035f26ae1 : SUCCESS in 23m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc929aab5a184885916f83ef8289d669 : SUCCESS in 1h 04m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d2c2f562672145eabde5e34f9f22d0da : SUCCESS in 1h 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ddbf202891e41c8ac13bff7116b8437 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a011b4d85ba84323bdd286ea634b87d2 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6327710290604ff9ab36ae54c1e5348e : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6f72af98024628bdbc5b6d571257a6 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906af03b2ec644188e4f3e70be6e2b18 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb7d6d49a9ae4ddd826d53b4914d0c81 : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/01c1d730b61c484f82ede65cb18b39a3 : SUCCESS in 28m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ffadc375aef24248b720a109b371f167 : SUCCESS in 27m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc2a10a8f4ea479ebe3485a7ca490f37 : SUCCESS in 1h 06m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dcd34fbcb6ce4efda2aa36c2e19607b5 : SUCCESS in 1h 34m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6c7e258bd0f94ed3b222400ef44073e2 : SUCCESS in 25m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/53e912e6e3d54002a5730f638f11a6db : SUCCESS in 24m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38db63e900d54ec7bc15747984fecbb3 : SUCCESS in 59m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f3b3ac6ef9824b57ababa3a273a40ae6 : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0bd2e58cb4e80b3700d39ae46616beedeef3764d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 15:52:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ac08f0df00246c1bc60f9d0031d01fc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/22adaa03016f4c73a8f0dfa0b9a23ed1 : FAILURE in 2h 04m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e747179e80f345a3b291551570b33387 : FAILURE in 1h 09m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28110ab7c9294684b951f9ecbb0a1635 : SUCCESS in 1h 42m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db48dc4789fd4f258e09b6eecd8aeb44 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/309782430e2e42b39e4bb47548b91593 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/962e5ff31b64472aa37dc602a148e211 : TIMED_OUT in 3h 02m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3e2dd6e27a584c7db0bba80e892cdfd3 : TIMED_OUT in 3h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"daa12c93987df788b14dd939bf5eb325ec37a449","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 08:33:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"72e04e09414bb9b494cdf6930d6ed60a25f30901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 09:36:22.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"15645338f4c490da21afb60463ac54892f9512ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 11:16:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3553f9cd844c92aaaed38b82b38963\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9936f4de55444e5dbb0821cbd050ff5e : SUCCESS in 25m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/21f97ee0de1d4ea0a3951f42626c3720 : SUCCESS in 1h 05m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b7e6f9c1bef43b3a8dc09cbff90ddb6 : SUCCESS in 22m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/58d1e9cc030f4de5b54aad78b701b09a : SUCCESS in 33m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce6ad4f20ad444b0b7e2b96f47138e6c : SUCCESS in 1h 04m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4e6dcebb1e4e4edf9a49859751a2d6fb : SUCCESS in 1h 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385aab666ea648aab762bd758d6482b3 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66e90766d9d14ccaa9e276218e812df9 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/417dd573141a475a8d74a4329cd6b17c : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b741802168724801a10e7e6050589014 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99738aaa2b0345afb7b8c59a67b8b019 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a42cd15ec55402d82e6944c48747181 : SUCCESS in 7m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b1701f9eab77466c9bccdf1b01c586f2 : SUCCESS in 26m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/147bf46172b549579bde1b93a860ee41 : SUCCESS in 31m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6099c765e84741b6860cdb495fa3c80a : SUCCESS in 1h 01m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2059d9123e684177a21a9e2b610a341b : SUCCESS in 1h 27m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b437cd7845a048e59dc3e412d1d2fd9e : SUCCESS in 22m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3fbe45124cc4096a338a882aa5e24de : SUCCESS in 47m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e087e7cd5a24114a5a60e4bdfa6a2c6 : SUCCESS in 57m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f16118d6d0344cf68b8eaf65668b4eec : SUCCESS in 1h 33m 17s","accounts_in_message":[],"_revision_number":5},{"id":"4504805b2a36f618bf03429713d3b8ee14d4971f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:40:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c63fc650a9d4bc0bbff4c2b33810561\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3851b1819c4c4bdaa6091654b31202ad : FAILURE in 2h 41m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/117c959a9e3b47ef87fe8e174a3fab4a : FAILURE in 2h 46m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e2f39bfec17f4a6789ad9a03aa2d87b0 : SUCCESS in 2h 03m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb4477e867e24590b94379e8146127db : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45b4f3fafd774248af231339b2a312f9 : SUCCESS in 6m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f5bc2a0602cf4362846138ee0942d055 : FAILURE in 2h 44m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/731f1725c637433bbe14319008f83216 : TIMED_OUT in 3h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee0d8ccd7e9e9bc6b76161ccacfcf74093cf6a06","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-14 09:23:46.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f37740c51c0f688c1d3878515db573ef751028a2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-14 17:07:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ffa0633483acc12e296d714f1cc9621a768fe3cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-22 07:56:44.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a2720e027709694ea464546fec7f892a6d78bb1f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-22 16:52:19.000000000","message":"Patch Set 6: Backport-Candidate+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f2ad99b0d570c73e1e25d90c49f4fa771a981650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-22 16:54:32.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6ee5ec85bcef65c24d831da023c5c34366432995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-22 16:57:15.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"7cca77b136bc7b919b3de9a4267881f776b459b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 19:32:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b578e59c06b14a038c2025d427cbbc30\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d82592a9e46d4748b1c3d94de92e7de6 : SUCCESS in 23m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/943c61b5b5004c58ac2b20ec285fcfa0 : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/debb16165d6043f2bf2e9ee99f65d2da : SUCCESS in 23m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f81ec9b2854a4f62a52997cebe207321 : SUCCESS in 32m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/979c380ec54d420ba0a2ef621fe8bf94 : SUCCESS in 1h 02m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c1796bdb07ad442bbdf1a49e9c72827a : SUCCESS in 1h 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0983f9024a463fa7ea23e17b92a0ac : FAILURE in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d8c6f2900af464b823406a3b31f8dee : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd39a9672669414ab6acd2a6d7452b3b : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01217ea2675b4632bb761ded954197dd : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a67e395a84b4b1696520d3e5180519c : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41435a8967e54762864f04a7878b7eb9 : SUCCESS in 3m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/62eb92f72a52489d8252e3ea4bee891e : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ba7db3ee89774becb503bd2c1715401f : SUCCESS in 27m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1e93a7e3895448f89d835145b8552a0a : SUCCESS in 29m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cd6ec45f70b248129c51ecb9e2bb457b : SUCCESS in 1h 08m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/55dff2587f1d40b585bfa40123b65f17 : SUCCESS in 1h 38m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/96d01638c64e4c80bbb718cb7733be7e : SUCCESS in 26m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/68bbd0d666a6407fb4bc93431341bbd2 : SUCCESS in 35m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0424e0bc2c0e435faefc5ba5c955e355 : SUCCESS in 59m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cfd09e3c652047949dd3e827770f4eac : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":8},{"id":"2bff5b3dc88d5ad2d2e98571f31c671cdb714742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 23:55:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8821b48676462ab1d8da89a5adca8c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/784ea2bf34804500928108f2b3853f3c : FAILURE in 2h 10m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/627680cb40154ca1b4366be692e0569e : TIMED_OUT in 3h 01m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15f4a05e464c40f6b546df25a08eba3f : SUCCESS in 2h 31m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09d8c36ac65b4ce79b79f3ac2139d6c0 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d158bbbfabb9458a889eec3271563ac3 : SUCCESS in 5m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d697a4b056534109ba6d1b58b165d22e : FAILURE in 1h 34m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2de48021e5674ba3997d59b4ecd4ef62 : TIMED_OUT in 3h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3734d0efba3f5b6c3b906242593449ca3b856f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-23 05:56:14.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"cfe2fc0027afa8a38a115280fcc1dbea9986185b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-23 05:56:31.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ad487f57ddeae9b70720eb64018a1005061b8cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 07:36:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794b55d8a73246b1bfefc4bc8f37211d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/adeafb8df9204701bb21f5401b16bcb7 : SUCCESS in 29m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4acd8e5171704837a0c5e13c43d9a15d : SUCCESS in 39m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/351f50a0306a46dfb884e7bd7069ceab : SUCCESS in 24m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7538c4358588456c9a53636811d2edc0 : SUCCESS in 38m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6022b940b0b64f618d04be50736159b9 : SUCCESS in 37m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e146dd8fb42c442994b4fb8a535e75f2 : SUCCESS in 1h 34m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad853876bc204c0dadf299376e9d7226 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0cfd61498a42d1998704450af1d948 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/215c80922eeb459ba3df65fe34000852 : SUCCESS in 4m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fea70fc16a1443a87faade9e81c6deb : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b9de281882484bbc725b164c0f3354 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b89dd95b7ce4843a7cea5322153d21b : SUCCESS in 3m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1196949e0cf444c4863ac7e0aa46c18d : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/56f26b5fd87345b79ae1e99b08f12c70 : SUCCESS in 28m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3dbecb59995c4381a28aedaf0a698331 : SUCCESS in 31m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b87b57dc76804fe9bcacbf8aa4988bf6 : SUCCESS in 1h 03m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aec8ceaffb4a4c44a6dcb7fb702aee5e : SUCCESS in 1h 23m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27e0d190dc354fa39ab1ab774cfe4480 : SUCCESS in 19m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/84b60110668c483f9c144d8ebccc7e9e : SUCCESS in 23m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e070bc6468cd48f88c72434479df837d : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/611ca00bf17443588f9cd458b322e64d : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":9},{"id":"0e90cb95b3143f4624770ace0b3644869df2d571","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 09:03:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/542204d0606941c58992a5e4499b17f2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b48046d344644c65b7f5464618b345ff : TIMED_OUT in 3h 01m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ab6a8cefa404e4c9a099e700df53e6c : FAILURE in 1h 51m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0266337e14644bd287b5a865393398c0 : SUCCESS in 2h 21m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51d55ab5b33d46aabe6db3be1c693997 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6dedac3e685645e9b34880f7680f9812 : SUCCESS in 6m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/045e60a74aa04c8a80dbb25c5c2bfcf2 : FAILURE in 1h 27m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/37a37397578b48a0a98e09f8a6969865 : TIMED_OUT in 3h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f280a4af374a82edb12bd9fd89a6feb9f6de1c9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-23 09:51:33.000000000","message":"Patch Set 9: Backport-Candidate+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0cc38af724d4cae68a64158a1f7805b96c741b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-23 10:22:58.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ffe73f8c47debbd1d4aed4feae8ed4c9364bbd9a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-23 10:49:22.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fb7f9fb9cf806f82370e84c67d3b803c00e89ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 12:01:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1cc163b1674658adf6093107fe4699\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c48ac49462244a08b2ad6470bedc757c : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/51cd82a9ec964ea29e8da762387aa011 : SUCCESS in 1h 14m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45fe0d41ba23436083a9244503ee151f : SUCCESS in 21m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a8c5c80200ad45369770df61c89d870c : SUCCESS in 19m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95615aae7e78444ea6dc9f83ebdcf74f : SUCCESS in 1h 13m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/09d8a358ae6442b7adeb1a5d1e4431de : SUCCESS in 1h 23m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d5b182b4f74279a5d5de8e0afc2daf : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c6540580b7b4bb6a46e0e73bc54d76c : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ce5b39f50224103ab37a46382607a3d : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dfbbd9b3b914b12b537c30fd5223ecb : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d044aea808934b438c51fe613126821f : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/358618b35d00403c803cdeb106b9c7e3 : SUCCESS in 3m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a2ddfb2b0354f68accaf34f6b582fc8 : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5e1f5ffa9b5148aea4615f84fac120db : SUCCESS in 25m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dc9dfbd4bf7940cc8396c6ffa7fd18bc : SUCCESS in 35m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92594de7e8f84259bfa9caddfb37e155 : SUCCESS in 1h 09m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/28f6f2e9a13048aea7549863ea97dad8 : SUCCESS in 1h 32m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/621b95c4b7fe4aa48d5ae78a692d5755 : SUCCESS in 20m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/50c03bdb0e3f4ce0942b2ed7f69297a2 : SUCCESS in 35m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fca8c39e427d48c099c35c7cb755951b : SUCCESS in 1h 06m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7b73f71da8d543b493a71bd9029db021 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":10},{"id":"a4d5b487dec2030570db48b32a16457994e041e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 14:42:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe6971bbe3ff4d52bf1a31b9f0e007f5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e5b82b650e2a496795b55d47bc69dcad : FAILURE in 2h 54m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a1bf4cfa6d34cd9b2375b668f6962ae : FAILURE in 8m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be98b6a6a6e94c2da84f0e075484c68e : SUCCESS in 2h 50m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26c3800e5199487ea1b4957e377f8d65 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e75b60fec44b49cc9b49a38077b1ceff : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e271887ace8641419de8e0721ea2ff5e : FAILURE in 2h 36m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/59f419eb65fd4eeda84b91ff710c0fc9 : FAILURE in 23m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b1472e6ef44334ff2e61af597531c02767e08ed6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-24 06:44:57.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"7b65946c198edfbe3e651a56220c03bec7356efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-24 06:48:38.000000000","message":"Patch Set 11: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"e7585a27918973553235c7875a17c4f1eb5395a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-24 06:53:12.000000000","message":"Patch Set 11: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3b3f24d2c557b57de82d97b13269d13cf4b7d706","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-24 06:58:58.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"ee12808989f1dd1eb4279dba5d6fc1ba8cc8888d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 08:30:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b934968fbea4ce784d35bbfcb96caeb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f082a0d8008849f6bceb9fe0dcc8fd62 : SUCCESS in 17m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/57fb30a62b6e4f4d994be7f37ce07072 : SUCCESS in 1h 08m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aaac484e1713447aaa673c009032eaae : SUCCESS in 13m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/833deef03a1d40eaad32d51b641f8079 : SUCCESS in 28m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6a742c788f845869fc34d8628df303e : SUCCESS in 37m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cac91e29847b4c12abfe2447188d3883 : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da87a6c585c24171b80470d9bd3fe2ae : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3644b9a1376247b989ad10f966f949ee : SUCCESS in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adbb42685bf14a02a3806d2e6bb11d21 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68ecb904167c4b2b84adc7cfdcfb439a : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a162429169430f8035ab24349fbcf5 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4707268a2c4d398ff9931458ba8d78 : SUCCESS in 2m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/01c799dd6484456f983243278e5a6b03 : SUCCESS in 2m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/94cca6b802f547fd994da3b73b29e442 : SUCCESS in 27m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/68e4bd358dbf4c338084de8093eea595 : SUCCESS in 32m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/60758ee8dbcc4ef08e0cf2fd3cc5e9e3 : SUCCESS in 1h 09m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f16399329af94b6f94064fc4a54a532c : SUCCESS in 1h 24m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d5f551a74edf4f099510d4c79328f6b3 : SUCCESS in 22m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1a0d15ae96b444eea0a50713c466d117 : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3a3d6394d15e486e978aed3d5d48b810 : SUCCESS in 37m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7b0b362b0b664c9db5c4f1e929b5afc7 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":11},{"id":"017b4bd4ddb56705a77ba48e37dddae80957600b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 08:31:29.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"89b374928547f6de0e3c451b1412668572d78924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 09:59:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2e5fa987a9436b82752af8c4ac8ab8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e4c75911fa3c40e9b8557ae787de9388 : FAILURE in 2h 40m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd694042f37144eb8e7fe71d9daa0122 : FAILURE in 2h 56m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74667e6af50e4ee195c8b48dfa2f3343 : SUCCESS in 2h 45m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e10e6330edb43be9f2742b9d8c14ee8 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d13ae37477294a58a88d4693dc0ac7ea : SUCCESS in 5m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/df96700c9b27433eb010a42d2d2bc2e7 : TIMED_OUT in 3h 02m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9215c8338251493f8b1fe719f4ebb0a5 : FAILURE in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b9439039b1a68a7dca43e06446e2cb6b17024ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 11:47:41.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c42a712c47a48d880c36d466865b5ec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5e35b837dd72404f9ef228bab48a49f5 : SUCCESS in 22m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92258dcf28274223b7660a3262bfced1 : SUCCESS in 20m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/53774ea69272483793a61758bf7b7fb4 : SUCCESS in 1h 04m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/31b8b329ea8d43eca8088a9a40cbd161 : SUCCESS in 1h 31m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d3aabb42864cd68374f8d2fa857906 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95aade671f4443fa893104c44c182fb4 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9b7900560394c1292107004184227cf : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f234717afc4ef6ba38b4e2aec6df4d : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c708b1910e274137bd61e7d55c9ae417 : SUCCESS in 3m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7663ae8acf9f4a668d26f33fea82c3a4 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f8bb26da985e4ac78209b00d0e4138fa : SUCCESS in 40m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/230b3b6208cd4318a52a64b19f3cbe25 : SUCCESS in 35m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce6118f19c1a44098b88a919b92ea93e : SUCCESS in 1h 19m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b9611af8daf74f068f94350b09bc3b56 : SUCCESS in 1h 32m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4224abcc8c72422fa21f855d09defe90 : SUCCESS in 1h 18m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f13966a6f4c24af3a336bdbf658050d5 : SUCCESS in 1h 18m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ba77cf223ea4e2d8194653f30499629 : SUCCESS in 39m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/255a08a24f14416cb82e7ead013161e7 : TIMED_OUT in 3h 02m 33s","accounts_in_message":[],"_revision_number":11},{"id":"4e125bdd6d4b9ffb3428ce6e00525394036be80b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-24 11:51:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2958cad95894553fbbbafc6b166649d11441fd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 13:43:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/652f130907f245e885a9d629e009ed48\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4e24b538e9cb4cada9dab48a264681c7 : SUCCESS in 29m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/90a498ab193148d6b6a3cef5c44df91e : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a139a081d39349f2a5c37a4d90715ddc : SUCCESS in 25m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a05334eaabe4613b9a209c76f6196b9 : SUCCESS in 28m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8b6348ab5db04a28859de98dcdbafc80 : SUCCESS in 1h 08m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/de4430afcf6744ca996ba6c85754e1f8 : SUCCESS in 1h 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f80f9f67d94ad7bc1e76a82329edcd : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42343f6a0cc640db9fb680e091594272 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7626287578c547f0b05486f23b68cd59 : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/331e5631393f40fcb3df39938616ed56 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1189310b4b64c18922f7e4255f94c3e : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68d27440ced4e8bb2ede33b9b402d7e : SUCCESS in 3m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c778ef83f4114326a56877c16173e3b1 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aeadcdd6bf72498eb1494af1a953cf96 : SUCCESS in 24m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d4ce74592191452ba18a820f77fd3dbb : SUCCESS in 31m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/40b955053944429bba82bf36eee80783 : SUCCESS in 59m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bdde77dc721b4b11a7210cb44c7a171c : SUCCESS in 1h 19m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02660b68706d473fbc20575eae7be72d : SUCCESS in 25m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6cc7686ff8949838c97fa38966a0816 : SUCCESS in 34m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/674d24f2a2f04c6cac3a20fd2dd2d31d : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4c92b4e4e30b4901a3b093c8221f1cef : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":11},{"id":"cf759ca33a036b2e04ea28123fc5532547cfa511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 13:44:13.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5609546a8905f4023d3ab40f3237eb7846f325e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:32:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6960fbd3504de0865d15fd23098f33\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5efa4a72125f4d0f8739acb79168ce42 : SUCCESS in 20m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/934e8f2f71ec4d28b0b893f232c1fb48 : SUCCESS in 29m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5e39cafef23c419dac9b38d79d784b9b : SUCCESS in 1h 04m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9152022ffc694ad68765465c86b19093 : SUCCESS in 1h 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90cb33ff057943b396db67e6d26065e4 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b174ddd56e14c088fdf1e8488d205cf : SUCCESS in 7m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d679b4961f43409d938dc878fb793353 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b00236926a4e60a74f567895627008 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87b2a26d81af4a5f927e98df9f675868 : SUCCESS in 4m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b73173b252b6433b82524b11f4bb9309 : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43159aa135d6463ba2697dc728a02c45 : SUCCESS in 22m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8066612ab06249a190eae7bd0276b063 : SUCCESS in 31m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d5c08ddc075c4cebbeda3b584e2f02a1 : SUCCESS in 1h 35m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dad03c37a1034bb68dba2f299aa85029 : SUCCESS in 1h 32m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/24d9f865a6904547ae0a25c17881dcdd : SUCCESS in 13m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a9ab2e5d01364074a96ef1f30412f24b : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/de319ac71cee41af975eb51a1ccc8f4c : SUCCESS in 1h 10m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d51ff210ec3843c091359998c1f214a6 : SUCCESS in 1h 30m 34s","accounts_in_message":[],"_revision_number":11},{"id":"0fbfee90b573a3ecf2d44cfece77415aacc9964b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:32:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"72f81fee0c6039258cfc906ab23e847e7e2ffd92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:34:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01a7ddbbb814bbc838c2c6698879420\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d50fb645c5d4a4aa95726fafec11686 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b1b99bf0cc1433abdec984366842345 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":11},{"id":"1adbf168541784b2d364e7e28b8db324c4657fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 16:18:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1063fa6b6b1f4e21b44daaa535927815\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c80cb0e92e944ac6a43be4befb694e9a : FAILURE in 2h 58m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59dd4e1741d94e2d83e570c24d14d9ef : FAILURE in 2h 26m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b0f0696b7ce425492e34d7da42054cd : SUCCESS in 2h 32m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2af01f79c734125bb511f214ca03589 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd7cd0199902493fb8563efdab681804 : SUCCESS in 6m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e14e4205d2a246998772685730b28912 : TIMED_OUT in 3h 01m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6934b5515b494bfab0ddf7d8158b33d9 : FAILURE in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d3bb06d5ce3fb9df53fd6e043a3fea9bb1faeae0","revisions":{"6a1ffaebf5bad304492b2a6998a35c232209e3a1":{"kind":"REWORK","_number":1,"created":"2025-01-07 10:59:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/1"}}},"commit":{"parents":[{"commit":"83d68434b91d98c23e6453a0c4596401368e0250","subject":"Add ovn-sb-db-relay image with newer ovn-ctl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d68434b91d98c23e6453a0c4596401368e0250"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:59:06.000000000","tz":0},"subject":"WIP: Fix genconfig","message":"WIP: Fix genconfig\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a1ffaebf5bad304492b2a6998a35c232209e3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a1ffaebf5bad304492b2a6998a35c232209e3a1"}]},"branch":"refs/heads/master"},"d6c976f63c8b111d4586b439c99fc186ca925e40":{"kind":"REWORK","_number":2,"created":"2025-01-07 11:00:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/2"}}},"commit":{"parents":[{"commit":"83d68434b91d98c23e6453a0c4596401368e0250","subject":"Add ovn-sb-db-relay image with newer ovn-ctl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d68434b91d98c23e6453a0c4596401368e0250"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 11:00:22.000000000","tz":0},"subject":"WIP: Fix genconfig","message":"WIP: Fix genconfig\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6c976f63c8b111d4586b439c99fc186ca925e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6c976f63c8b111d4586b439c99fc186ca925e40"}]},"branch":"refs/heads/master"},"420fac1640c56b323c98c25196cd06a7df135e8e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-07 12:43:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/3"}}},"commit":{"parents":[{"commit":"83d68434b91d98c23e6453a0c4596401368e0250","subject":"Add ovn-sb-db-relay image with newer ovn-ctl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d68434b91d98c23e6453a0c4596401368e0250"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 12:43:44.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420fac1640c56b323c98c25196cd06a7df135e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420fac1640c56b323c98c25196cd06a7df135e8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c003e028063a680131675fb675e6b5eccca6dde6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-07 12:45:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/4"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 12:45:07.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c003e028063a680131675fb675e6b5eccca6dde6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c003e028063a680131675fb675e6b5eccca6dde6"}]},"branch":"refs/heads/master","description":"Rebase"},"e3e3f0d08cef6e5efc3450ceb60a34f11c914889":{"kind":"REWORK","_number":5,"created":"2025-01-13 09:36:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/5"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 09:36:22.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e3f0d08cef6e5efc3450ceb60a34f11c914889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3e3f0d08cef6e5efc3450ceb60a34f11c914889"}]},"branch":"refs/heads/master"},"d5be93c0917583092e8320e7c21f59ace220ec6a":{"kind":"REWORK","_number":6,"created":"2025-01-22 07:56:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/6"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-22 07:56:44.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5be93c0917583092e8320e7c21f59ace220ec6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5be93c0917583092e8320e7c21f59ace220ec6a"}]},"branch":"refs/heads/master"},"6439cbebd65eb0fb2ee9bc5fc16d45e99c97ca57":{"kind":"REWORK","_number":7,"created":"2025-01-22 16:54:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/7"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-22 16:54:32.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6439cbebd65eb0fb2ee9bc5fc16d45e99c97ca57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6439cbebd65eb0fb2ee9bc5fc16d45e99c97ca57"}]},"branch":"refs/heads/master"},"a280d447b67b727413867465cb68aba425d45296":{"kind":"REWORK","_number":8,"created":"2025-01-22 16:57:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/8"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-22 16:57:10.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a280d447b67b727413867465cb68aba425d45296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a280d447b67b727413867465cb68aba425d45296"}]},"branch":"refs/heads/master"},"822ad5214f7a319a490c2b8a2fae24a63d50b8d0":{"kind":"REWORK","_number":9,"created":"2025-01-23 05:56:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/9"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-23 05:56:14.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822ad5214f7a319a490c2b8a2fae24a63d50b8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822ad5214f7a319a490c2b8a2fae24a63d50b8d0"}]},"branch":"refs/heads/master"},"a9d37e58322e72507cb278da5002799f61c95619":{"kind":"REWORK","_number":10,"created":"2025-01-23 10:22:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/10"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-23 10:22:58.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9d37e58322e72507cb278da5002799f61c95619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9d37e58322e72507cb278da5002799f61c95619"}]},"branch":"refs/heads/master"},"d3bb06d5ce3fb9df53fd6e043a3fea9bb1faeae0":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-01-24 06:48:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/938546/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/938546/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/938546/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/938546/11"}}},"commit":{"parents":[{"commit":"aa3e06cd33ef9a469d51d5da41e29d304e42fcfb","subject":"Merge \"Time mismatch in /etc/localtime and /etc/timezone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3e06cd33ef9a469d51d5da41e29d304e42fcfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-24 06:48:38.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nCloses-Bug: #2051575\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3bb06d5ce3fb9df53fd6e043a3fea9bb1faeae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3bb06d5ce3fb9df53fd6e043a3fea9bb1faeae0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
