)]}'
{"id":"openstack%2Fkolla~938562","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I57e36ef07e8840eead2efde77e0ab9dfd4842931","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-07 18:48:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57e36ef07e8840eead2efde77e0ab9dfd4842931","subject":"neutron-dhcp-agent: change dnsmasq.log rights","status":"MERGED","created":"2025-01-07 15:13:37.000000000","updated":"2025-01-07 18:49:16.000000000","submitted":"2025-01-07 18:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938562","cherry_pick_of_change":938351,"cherry_pick_of_patch_set":10,"meta_rev_id":"9605340dae7ea1609f3ef7437391f9bc8542bbe3","_number":938562,"virtual_id_number":938562,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 18:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-07 15:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-01-07 17:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-07 17:09:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 17:04:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 17:40:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"a325e61263df83f9fb1149742d440bc0569629f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:13:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/neutron/neutron-dhcp-agent/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"29594b81ab0701334ae7b5c9e93d40556db629b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"90c58658fe50df818dba2170d69e13a944bc3463","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3bc9750ef455985c9a32b4ce6efc480be2980a58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67aeb27d8b01ca5d93e5c05568c48a71af614a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:04:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9172a562eb974dc7a5151d36aefe4704\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/05ca5d4334774ba89fec6c36f78c133c : SUCCESS in 20m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/40cc015be4d14a868a2837846ec55f84 : SUCCESS in 1h 02m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf10f83baf1d44fabdf5f2bad6d04a54 : SUCCESS in 28m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3b0911db5dd74ee29c24e421b79c6101 : SUCCESS in 36m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c990f7a72e8846898910e32b1ba0f5ef : SUCCESS in 59m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cac387ced1344fd785c6206f4fb85980 : SUCCESS in 1h 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fc68bbad6a4b3182f09ebbbfd0a48c : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78e8affa8ff74767863d1de87de179a6 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1741310883a40328d6eba884ea41d73 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eaa009ea608c4de0a93cab4c557291c3 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f31b2ddd107433996fdfd7cef049a49 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d894cc100b234214bd92aee81bc85f88 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/af9106564594467d9f0bc835d9133dd6 : SUCCESS in 29m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/af8657f5a8c0462dac60fc5a82762c59 : SUCCESS in 29m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3dcd29f2ee014eb3b0c863bff2bc6cf9 : SUCCESS in 1h 15m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b6908292881b44f1a9717ce1191de1d0 : SUCCESS in 1h 19m 54s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/733fe1c178e54f66ab490cf79659ab6b : SUCCESS in 1h 00m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/69e8a2e9e12e47659c1dddb274e256c7 : SUCCESS in 1h 43m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/568ffd51991a45208aeb9c44fd29b961 : SUCCESS in 20m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cd01d2dbc7b04f68b4cac0d292442d23 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8abf6cfd81c24d249fee754a078f82c2 : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2acc479e22ed42d79720e836be60050a : SUCCESS in 1h 20m 52s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/22135fbf068e455990d0998b2ddcf517 : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c7fdc873984c4b5b994197a537248617 : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2294341e2c4da85bcd5896af4e9ef66978de708d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 17:09:33.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00baa5ce383b76c138efb868814d86f21f45bf20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:09:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0441a9f0e4a3c5a9361b0bf5fc8cc1ab733ec78","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-07 17:40:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b33bcc3581245a1ac417499c0beb3f709333b3c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:16:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f50abacdb5c44dea62dadc3117dccda\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e0db3b6ef5784d27a4b23bc15d9f3f3a : TIMED_OUT in 3h 01m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d04e47eb8b14fcc9f308afd2ca6def2 : FAILURE in 2h 33m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f59be76d74f14075b764d6bfd0180e6d : SUCCESS in 1h 51m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83d4485b0b434e61b9c4f3e9d704a0d2 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/903117dab7b547c6963869a06abf0b29 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/227f2507ff714457a246d60866e1dab0 : FAILURE in 2h 39m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ad1dae686731421da5cc296820cdd5b5 : TIMED_OUT in 3h 01m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a549beac93fa30d6bcac5d61e890ca69db9efce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:48:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4dd93cf078f48009b434537b0dd4fc5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b25f8141dad4069aa9e6a71550e2e63 : SUCCESS in 19m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4523db9aeec74adc949fa12f2d04e290 : SUCCESS in 32m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ea1e95427a44b47a3a48d1e80f0e1f2 : SUCCESS in 35m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/422152f0fd674537a833a4a8e4df5e0f : SUCCESS in 1h 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e67a750655470f84b170e641170d8b : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8723f0d4c484bc19a72779d9f35bc72 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30ab5da36f67430488c64eee80b17246 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac3d4f5a9c4449fbfba31f32e1d324e : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/304531e1f4304fd99103924ff2c67f06 : SUCCESS in 21m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8a579f3705b46c7a9ee303190dd0e0f : SUCCESS in 33m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b21a3fd35c8d4dcfad7b672830c7e7ad : SUCCESS in 1h 08m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8c12287b8e6f49a7aa51002931437149 : SUCCESS in 1h 29m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4d4d892e0547411785265d652267b6b6 : SUCCESS in 22m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f48f0eae008f499ea869977a2f68b2b7 : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/565b8d5392544d2fa99c440886be720b : SUCCESS in 1h 11m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8bc49b06af8b4947840c0e7cf20f8877 : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a220cf8b449b3829d88e16827febd26ac7c87335","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9605340dae7ea1609f3ef7437391f9bc8542bbe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:49:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ce3095da875487f84fb26c1d3a05e1a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0662257685b469b83e2bfdaff193795 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0972d82a26463c69d0334350b23b558d3b5f3edf","revisions":{"2d6ac90fc07b83f430a3d348f97c74ace621a9d2":{"kind":"REWORK","_number":1,"created":"2025-01-07 15:13:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/938562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/938562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/938562/1"}}},"commit":{"parents":[{"commit":"8fe38110b4907ab6775a2a6345c20151e5a5dfff","subject":"magnum: Add helm binary for magnum-cluster-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe38110b4907ab6775a2a6345c20151e5a5dfff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-02 09:47:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 15:13:37.000000000","tz":0},"subject":"neutron-dhcp-agent: change dnsmasq.log rights","message":"neutron-dhcp-agent: change dnsmasq.log rights\n\nFixes /var/log/kolla/neutron/dnsmasq.log rights so\nfluentd can read logs from that file.\n\nCloses-Bug: #1693712\nChange-Id: I57e36ef07e8840eead2efde77e0ab9dfd4842931\n(cherry picked from commit 0f5e39954292a51e7a967d4015a6f024d6ba42f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d6ac90fc07b83f430a3d348f97c74ace621a9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d6ac90fc07b83f430a3d348f97c74ace621a9d2"}]},"branch":"refs/heads/stable/2024.2"},"0972d82a26463c69d0334350b23b558d3b5f3edf":{"kind":"REWORK","_number":2,"created":"2025-01-07 15:14:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/938562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/938562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/938562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/938562/2"}}},"commit":{"parents":[{"commit":"8fe38110b4907ab6775a2a6345c20151e5a5dfff","subject":"magnum: Add helm binary for magnum-cluster-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe38110b4907ab6775a2a6345c20151e5a5dfff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-02 09:47:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 15:14:02.000000000","tz":0},"subject":"neutron-dhcp-agent: change dnsmasq.log rights","message":"neutron-dhcp-agent: change dnsmasq.log rights\n\nFixes /var/log/kolla/neutron/dnsmasq.log rights so\nfluentd can read logs from that file.\n\nCloses-Bug: #1693712\nChange-Id: I57e36ef07e8840eead2efde77e0ab9dfd4842931\n(cherry picked from commit 0f5e39954292a51e7a967d4015a6f024d6ba42f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0972d82a26463c69d0334350b23b558d3b5f3edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0972d82a26463c69d0334350b23b558d3b5f3edf"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
