)]}'
{"id":"openstack%2Fkolla~938563","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I57e36ef07e8840eead2efde77e0ab9dfd4842931","project":"openstack/kolla","branch":"stable/2024.1","hashtags":[],"change_id":"I57e36ef07e8840eead2efde77e0ab9dfd4842931","subject":"neutron-dhcp-agent: change dnsmasq.log rights","status":"MERGED","created":"2025-01-07 15:14:19.000000000","updated":"2025-01-07 18:51:22.000000000","submitted":"2025-01-07 18:50:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938563","cherry_pick_of_change":938562,"cherry_pick_of_patch_set":2,"meta_rev_id":"f404f88eca2a985f38a61359863cea967d5073b1","_number":938563,"virtual_id_number":938563,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 18:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-07 15:15:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-07 17:09:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 17:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"332631b949b4905b47d0414d736e305c3a05838d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"443d160dd1140bc4e1a53492a0b03cec2b2a9e83","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"92b893e5e85c06e7ebf2d4f155f44f2d1b67aebb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:15:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96bbf5024aeecbdbc6d56e2d68b26f2639fb6586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:08:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74815c13eba34225bc7c82251f88f5da\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d442987f4da6443ba0743f53c08aeba6 : SUCCESS in 22m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d9604c797c84eefae32f1a732a7b31f : SUCCESS in 56m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2e33ef79cf0f4f3a9ee204cbfc4e7b55 : SUCCESS in 20m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/55e3ca190e0d4749ae8fda3512b33c44 : SUCCESS in 29m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/07c827deb5484ab2a72dd561371d00f2 : SUCCESS in 56m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/78e3f15998624e29aeda4e8336d2d7d3 : SUCCESS in 1h 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9025ea61d9946eeb50dc9b60ebcd12a : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2536d6504e14905835899bf029bd9cb : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d4bdc655a5840a3a2c8e0191252add1 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91cfbbf4a2784cbbab620fea30291d43 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fda484345de4e3ea6a3e55ef7e78642 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a9e9ab53ccbe4596a2f1f02cc8788073 : SUCCESS in 34m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1433e40721194ebb9ae81cd4a1772eaa : SUCCESS in 32m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/25d181801795477293766440da27e41f : SUCCESS in 1h 16m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a8e9bc7d06ac4cbaa19e30a606f77317 : SUCCESS in 1h 42m 00s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a552835103f448ca898549d15181a13b : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6f0205c023654fada4684963450f788e : SUCCESS in 1h 35m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3aff7bbed85040eb8858d0a830fd7e4f : SUCCESS in 19m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b4797796c39a48629b9063eee7eed271 : SUCCESS in 1h 02m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/85da6fbc460c4f6795dfc4eae0a9ce63 : SUCCESS in 52m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/85296d58457c4dccb7c434ce57e090f7 : SUCCESS in 1h 09m 15s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d65f231389e740fab7b2d7e99fe407b3 : SUCCESS in 55m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/796a7a3d05f24d16b24e76ccf1993d49 : SUCCESS in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67325fc992b6088567f38e25593fe5c2782dd6fd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 17:09:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41dbd6de5d5fb5141bec18fbcd3fec5ac4b0d7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:10:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e6e2b8b3f34e12811968434d4f8a17b2ad328d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:08:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a622bf7f8047458a931cc69feb567620\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ac798e194e04f6180d7f0bdef1bc4b8 : SUCCESS in 2h 51m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ffec3393c6346e9a8f03792bf8ebbff : SUCCESS in 1h 59m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b3f691ae024409b9b7e127669b41984 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18aae831d7834b2f8b84403b9b69b688 : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f93a00becb60b6920c071da32e706a475af266b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb4df87be161414b9521145354126482\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/297256592484423caefeb1c7a1d33527 : SUCCESS in 16m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/810db5b7a61342a69c716e27ddd13f3c : SUCCESS in 30m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d1434f9bdde24ca4a37e5460c1d68c5a : SUCCESS in 1h 03m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f92af95ea4734e20a1d9e315f1fe8d91 : SUCCESS in 1h 23m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14d0cd68a20a4650bd4faff2c16a0269 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f873ec48dc5741fa989335de3772e261 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63d4fafc0a5944d5bdbb19685e71d13c : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a70d9cbb9e74aaaa27fdc55fab0ec14 : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ef3f44545f764c2cb8beac32da1484a0 : SUCCESS in 26m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1002ae88c78141cd9c186bcc36ad4c06 : SUCCESS in 32m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3186e040900b4ff8bcffe862515c86c4 : SUCCESS in 1h 08m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/882e5019cdde4196a6448a971d807ee6 : SUCCESS in 1h 38m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82ae5a5e72854f1880fbbd8919dd7c42 : SUCCESS in 19m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f25f727400174bdda9306e55084d26c5 : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6186fa578add4a7b983ed70cf0fae121 : SUCCESS in 59m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86e2c1a9ac96435994a0692f443ac95f : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cc55ec851dac2b347355941ac04f33a8a5b9e8d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:50:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f404f88eca2a985f38a61359863cea967d5073b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:51:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ade0f119a254b7ba5c16526bab3d799\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d316f5febf42f19e75af9a13fbb0bf : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e942154060cf9865639de6b30413667a23c4e68","revisions":{"5e942154060cf9865639de6b30413667a23c4e68":{"kind":"REWORK","_number":1,"created":"2025-01-07 15:14:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/938563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/938563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/938563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/938563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/938563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/938563/1"}}},"commit":{"parents":[{"commit":"86d6fac7afbb67216726e0f1fe30dd13e1273de7","subject":"magnum: Add helm binary for magnum-cluster-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86d6fac7afbb67216726e0f1fe30dd13e1273de7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-02 09:47:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 15:14:19.000000000","tz":0},"subject":"neutron-dhcp-agent: change dnsmasq.log rights","message":"neutron-dhcp-agent: change dnsmasq.log rights\n\nFixes /var/log/kolla/neutron/dnsmasq.log rights so\nfluentd can read logs from that file.\n\nCloses-Bug: #1693712\nChange-Id: I57e36ef07e8840eead2efde77e0ab9dfd4842931\n(cherry picked from commit 0f5e39954292a51e7a967d4015a6f024d6ba42f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e942154060cf9865639de6b30413667a23c4e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e942154060cf9865639de6b30413667a23c4e68"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
