)]}'
{"id":"openstack%2Fkolla~938564","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I57e36ef07e8840eead2efde77e0ab9dfd4842931","project":"openstack/kolla","branch":"stable/2023.2","hashtags":[],"change_id":"I57e36ef07e8840eead2efde77e0ab9dfd4842931","subject":"neutron-dhcp-agent: change dnsmasq.log rights","status":"MERGED","created":"2025-01-07 15:14:35.000000000","updated":"2025-01-07 19:35:29.000000000","submitted":"2025-01-07 19:34:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938564","cherry_pick_of_change":938562,"cherry_pick_of_patch_set":2,"meta_rev_id":"cb792c084d1cf79f9b5523672ebca779e590a0e6","_number":938564,"virtual_id_number":938564,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 19:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-07 15:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-07 17:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 17:15:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b12efb28b0a97570c38b8623fd81d6561eabcd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"60d1323155dc038af81fef9fcb41df19f51bfd8a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 15:14:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acc70784247df7d989abf3f3594edc99c28910a2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-07 17:09:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2431756e00c9f3bded6ae50c187a5f28abc711f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:15:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa9c25055bf42969a476107b71fe85a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87d7cf9e0021422dbcd271a558cde10e : SUCCESS in 25m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f4268339be254900888ff722f207366f : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bb9f148c92464031beeff1151e62bd29 : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42f2b365be804aee80304bd580dc2316 : SUCCESS in 41m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8c72b94e721643f1a905fcdefa6b8608 : SUCCESS in 1h 07m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/401551c29546436985bad4d41de4ed91 : SUCCESS in 1h 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dcd252856f4d0f85e05b5c17d3e0d3 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f66f86337cb548c0ab0fcfaf752236e9 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9e1689bcdca437aa7a8f20b17a6be74 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d781546c441433f81e7044140a9370b : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8092e10a271544f3833b1a6c59e5542b : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2183c5ce37114f329c80021748d02fd0 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314bc9ccf1c1487d92dd04beecf98bd1 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ae05409d3c1a44379f9250540928b5bc : SUCCESS in 31m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/42bf378f5a5b4fff9a187c1f9a547b91 : SUCCESS in 37m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc0f4f0147e047448e71c56076445fd9 : SUCCESS in 1h 15m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2766c43431df4a67b34adf21a551db54 : SUCCESS in 1h 52m 51s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2faf134585ed459e885a22d5d3e0fbf6 : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e3d2bafb403444aebe27af19b07c84b3 : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3fc5d22e213545f2bfa435516f062d00 : SUCCESS in 31m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1b122068edd84b1993b5c7688c9fb371 : SUCCESS in 1h 05m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2a0cc06e787e40efb94f9015e78a6a07 : SUCCESS in 1h 03m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ddf5033420d84904835201543d96f515 : SUCCESS in 1h 20m 13s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f2b918b8833049a68c81cd37dd7a4123 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c43fce6e785c4630ab885b3feabe4d17 : SUCCESS in 1h 08m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05a8fb7151676227b708f34a22c29f4f29896f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:15:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5f612d8aa8b16bff350e22ab4df95705363fc84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 18:27:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da85c482e424d0797b730dd62539595\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dfe35095fff947359c67a645196b84d1 : TIMED_OUT in 3h 05m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f89dcefcd684402a32ad9822f6f5b4e : SUCCESS in 2h 02m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1408ec0051e74849a84fb2ced2b86a4b : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98e00c2db4684495ac64d87af255df4e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18bd4cf15be748ffaac7a4dbaa6f3768 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c46a3e580d4406ca871314dadead632 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1084c9bc57d003634105ecf901664baba8c3d08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 19:34:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4628019f07934f1a9e7dbc285177791d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f763eab3bca647419a4f0d03f7a0badb : SUCCESS in 17m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0a2aa5f5c15f44ef8455a07e20566f7b : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1692c17a6f8a49899c65218afbcf7f14 : SUCCESS in 59m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5131f8a180294fc282466a79c8cb5ab2 : SUCCESS in 1h 37m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32a0a981b2df44abb2199314216c8aaa : SUCCESS in 2m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d97f7cc11b6c4fbd93f04b254ca345f8 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123bdb6ddb474671894b9338effd71dc : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4004b5172c9547bb8fb4bec133bd2753 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c139349de40450dbe30c4f2982b5079 : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e604c41a1dab4519a388bc11aef7bee3 : SUCCESS in 25m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff749278817a45d1b207dc2886897523 : SUCCESS in 34m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/41ec676ecc714c0f9920773954fa36c9 : SUCCESS in 38m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ad819c8d773a409f862ef96bcbfd5754 : SUCCESS in 1h 38m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6a2f7490233847c989779a4c1f14a306 : SUCCESS in 17m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/74093dcaf665409a888b97058e411ade : SUCCESS in 1h 05m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f7377b76a7d4433599aed978dc290e1e : SUCCESS in 59m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e96d63e1e03045dd909363193051b4ab : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fcdb58cde5f2ef19feddec9354dc5542044f2d44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 19:34:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb792c084d1cf79f9b5523672ebca779e590a0e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 19:35:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4baea5557d6c4bc197b85ba5f12abc57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc42d0c077a46f1b335c5ef3a88c9da : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b880e8f9962692544c3774deb69292946165c58","revisions":{"1b880e8f9962692544c3774deb69292946165c58":{"kind":"REWORK","_number":1,"created":"2025-01-07 15:14:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/938564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/938564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/938564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/938564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/938564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/938564/1"}}},"commit":{"parents":[{"commit":"01a9e7b992672ca7432bb3fb6d076cb243c9ae05","subject":"CI: Stop using LABEL for ephemeral0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01a9e7b992672ca7432bb3fb6d076cb243c9ae05"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-02 09:47:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 15:14:35.000000000","tz":0},"subject":"neutron-dhcp-agent: change dnsmasq.log rights","message":"neutron-dhcp-agent: change dnsmasq.log rights\n\nFixes /var/log/kolla/neutron/dnsmasq.log rights so\nfluentd can read logs from that file.\n\nCloses-Bug: #1693712\nChange-Id: I57e36ef07e8840eead2efde77e0ab9dfd4842931\n(cherry picked from commit 0f5e39954292a51e7a967d4015a6f024d6ba42f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b880e8f9962692544c3774deb69292946165c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b880e8f9962692544c3774deb69292946165c58"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
