)]}'
{"id":"openstack%2Fkolla~938825","triplet_id":"openstack%2Fkolla~master~Icb4df3d03c79557a6dd73ea2580c93bc4a989765","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-23 17:09:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icb4df3d03c79557a6dd73ea2580c93bc4a989765","subject":"fluentd: Add to group root","status":"ABANDONED","created":"2025-01-09 16:46:26.000000000","updated":"2025-01-23 17:09:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"353e23ad96958e93a2eaaacf656537c4bc6be6ec","_number":938825,"virtual_id_number":938825,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-09 21:38:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-01-10 07:25:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 17:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 07:25:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"68cbfa5be4d0c886976ed92e9fd2573639b75670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-09 16:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f369e987aabff64278fc0ccd30d189c57ee8b317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 17:25:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7801c2e172e4342bf4007c3027e69c9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a188086ff08e4bf393019bb5962d0c10 : FAILURE in 26m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9ad1e97ab20a4c94a55cff8ddab40533 : FAILURE in 20m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9150894897f44dfb86ddd0f953d924cd : FAILURE in 22m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/97dc5437eaf844d697f883f842f9b56d : FAILURE in 33m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5cb547be5e874704b7067dfb75eaad79 : FAILURE in 18m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f26f9b0d519f4fe097e38f16c74178f1 : FAILURE in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92902e4dc4df487e93a8aa98bbc4f56e : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abdece9d442f4fcb89278da5ef5f8e2d : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5059dcfc70b478897847f9e6ae232b3 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba6ef18482754d74bcba00054919ab8a : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a8b080c0c34f768d1b486a1ae81ca7 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/655109eae86e4b3a83acd5363b9b1b1b : FAILURE in 26m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/42596010cb3a49628a6ee49c8d20ca91 : FAILURE in 22m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9d9a1f97e6e14a4d92f54af806cef1dc : FAILURE in 19m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7175ae90aed74628b632a2fd1cdddf51 : FAILURE in 19m 18s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2c68903f742a4fa98bd43a8d0902be31 : FAILURE in 22m 47s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8eaea7deef9a460280697d6015c55fbf : FAILURE in 22m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/48ae8d955f4340aca44a15549a36b97e : FAILURE in 30m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b64c92c0ed044a5e849cdc470c04bba4 : FAILURE in 18m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d52978dee7cb46b9a26898cf8059d82f : FAILURE in 23m 58s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ae6f5b12adf24f35a9758a0576bfcdfe : FAILURE in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb02d4c101436787454e8192eaf8aea252adc9c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 19:49:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378da18d236e4c91bc24d6fac07373c3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/08ccef9bb2e34473be66a6faef511b01 : FAILURE in 2h 58m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/98a66e9c80e048ce9a71101c5c651c71 : FAILURE in 2h 38m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/164bde5158ed4e9ca76632d71ca1b791 : FAILURE in 1h 08m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a6d55525c5b42df8e17a8d3c7d0df46 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fe8db9d995e49a88507dc0c0dd54a52 : SUCCESS in 5m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6750b5e3074147a1a85a378c9c470c36 : FAILURE in 2h 53m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4871d6e64d6044ce92168766d364895e : TIMED_OUT in 3h 01m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"140a807ad12ea6318f2c3fb28853b6b2c4f6a6ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-09 19:51:50.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"91d7cc5ba4746d417fb51800f7fcd1878fb1ef15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 21:38:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a1b57eeadd4d5cae36aba5a45c3681\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f035346ba23f4a1fb5762b0333f9c2d4 : SUCCESS in 21m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/943fc0ed5dd54b1ab61317e02e244ebb : SUCCESS in 1h 04m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0318bddd4d1a432aa7a75a686bc8a769 : SUCCESS in 14m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4f4e2cb2ef374e808bd55aa29ba8f366 : SUCCESS in 31m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15294e3c1107498db270363a2546b402 : SUCCESS in 54m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1f54a8c46f0b4484b7d348b7f8b952a9 : SUCCESS in 1h 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6e7e13cc9e4a4fad9ff301241db524 : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d676fec22064fa3b19176646f8fe442 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af11fa3e335b47f1a1c5e5908f9acb6b : SUCCESS in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/339f134edfcd48b484f710324bebe428 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93781f5915140fab1dcfab6bf0453d8 : SUCCESS in 4m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8287594c1664ce58b34a94aa19dd106 : SUCCESS in 28m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/22fc6822fa2449129e4bebf2fd8dada1 : SUCCESS in 36m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4388409dc93047bf822f2a6602bee0e2 : SUCCESS in 1h 07m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d7c18f2e49be4e67a9c85405ef698175 : SUCCESS in 1h 33m 19s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dd137b6863364f8abc272c42273309d4 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b961a30fefcf49ca86426834918fc264 : SUCCESS in 27m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c0543ae8858d49dbb0fb8287d62f1d9d : SUCCESS in 37m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/45d1d23060f24ed28ddf0ac59a9ec73f : SUCCESS in 1h 09m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0cc41f80e60d47e2b109e1cb7732355d : SUCCESS in 1h 28m 36s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ab3fd764ee3b4b8f881bf77d38a8409a : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19faa0e57919fa3ad3e0453d13edd4e4579c012e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 22:59:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f03ef422e404162a9ea7b897f26bd16\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/74928e9dcc3447f6939a319a05190edd : TIMED_OUT in 3h 02m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4d3e17cac964dfdbd63e32c812a649c : FAILURE in 2h 32m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb59b41b3d864c6db2a8564111d2cab8 : SUCCESS in 1h 38m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d44a34b28f9472286364be7190e9d24 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cedef5af8b24b3caf59ba41b6df68f6 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aa7162b9b1f444dfb8ce61e6edd561b7 : FAILURE in 1h 50m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0145aff7657f4c8aa05a9f2e10fb789b : FAILURE in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4be8887bcd423c6672f4fa3698cc19bda164b364","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 07:25:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"353e23ad96958e93a2eaaacf656537c4bc6be6ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-23 17:09:17.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2f0d59fac49a03b4957784c16372d17e663fe3b","revisions":{"abec92344684e83af525a79917e003c51607e19a":{"kind":"REWORK","_number":1,"created":"2025-01-09 16:46:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/938825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/938825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/938825/1"}}},"commit":{"parents":[{"commit":"c781c66d2b4fe43d169794bb6e34d6a4d147579c","subject":"Merge \"neutron-dhcp-agent: change dnsmasq.log rights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c781c66d2b4fe43d169794bb6e34d6a4d147579c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-09 16:46:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-09 16:46:24.000000000","tz":60},"subject":"fluentd: Add to group root","message":"fluentd: Add to group root\n\nSince openvswitch is running as root, it\u0027s saving it\u0027s logs\nalso as root - and fluentd can\u0027t read them.\n\nChange-Id: Icb4df3d03c79557a6dd73ea2580c93bc4a989765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abec92344684e83af525a79917e003c51607e19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abec92344684e83af525a79917e003c51607e19a"}]},"branch":"refs/heads/master"},"e2f0d59fac49a03b4957784c16372d17e663fe3b":{"kind":"REWORK","_number":2,"created":"2025-01-09 19:51:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/938825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/938825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/938825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/938825/2"}}},"commit":{"parents":[{"commit":"c781c66d2b4fe43d169794bb6e34d6a4d147579c","subject":"Merge \"neutron-dhcp-agent: change dnsmasq.log rights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c781c66d2b4fe43d169794bb6e34d6a4d147579c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-09 16:46:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-09 19:51:50.000000000","tz":0},"subject":"fluentd: Add to group root","message":"fluentd: Add to group root\n\nSince openvswitch is running as root, it\u0027s saving it\u0027s logs\nalso as root - and fluentd can\u0027t read them.\n\nChange-Id: Icb4df3d03c79557a6dd73ea2580c93bc4a989765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2f0d59fac49a03b4957784c16372d17e663fe3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2f0d59fac49a03b4957784c16372d17e663fe3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
