)]}'
{"id":"openstack%2Fkolla~939055","triplet_id":"openstack%2Fkolla~master~Iedfb2df027c0cefc45668a76d45ce62193aba9ef","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-13 18:40:42.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-13 17:03:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-13 15:37:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-01-13 15:47:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iedfb2df027c0cefc45668a76d45ce62193aba9ef","subject":"openvswitch: Fix logs dir permission","status":"MERGED","created":"2025-01-13 08:54:33.000000000","updated":"2025-01-13 18:42:40.000000000","submitted":"2025-01-13 18:40:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939055","meta_rev_id":"e0f8a705d438343078db036009d4da4827534b17","_number":939055,"virtual_id_number":939055,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 18:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-01-13 17:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-13 15:37:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-13 15:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-01-13 17:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-13 15:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-01-13 17:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-13 15:21:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 11:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 13:26:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-13 13:30:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-01-13 14:32:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b1340d6081dc23284de8435a34c85945de41389b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 08:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2d8976a6e27142e78f479d2dc7b4215b0ade153","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 09:12:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7f7aa06c415703c5dcb90f33c3295020751517c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 11:03:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e7061962c449bd97202a88e7087c08\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8e9a504d5db8493ab3ea90cf30ffc735 : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7f0d3b4c2b2a44b49c593cf2855c9300 : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7a6c587bd4b4d24890a00b93b0da64f : SUCCESS in 32m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e628a78af185441a9a93acacc26a80ae : SUCCESS in 36m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/93861dda948b4edb80eecd2f38129219 : SUCCESS in 58m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/986dd33573bb4696958f242f3fdc8be7 : SUCCESS in 1h 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03877a1d8e44102bbf54b2a540899c7 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/025b491886b34368b08bdfd9fc8ca0c2 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd6b5c8a95a8428fbc12dfa96187c95e : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4d445f53ac1490dbf8b33b3f65bbeab : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a71d1fe001c42b5b9b6bd29928cc328 : SUCCESS in 4m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5b0fbc4fca7b43759c7ad93269356504 : SUCCESS in 40m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6649f38fc9174d79abbf38daa64836f8 : SUCCESS in 36m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e76628596c17493da311fff562a2ae7e : SUCCESS in 1h 02m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70458487bd9f472693b9259d271ac21e : SUCCESS in 1h 40m 42s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b05c816cc56f41a7abaa20acd5dc806c : SUCCESS in 1h 22m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0c40ea21dec344f4b8a381e743c136e9 : SUCCESS in 1h 39m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1019244fd89141bb9229a01d5b8e449f : SUCCESS in 36m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f5b4656f85514dc5b4dc1e53249d4bb7 : SUCCESS in 51m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/67f4fb1c9343468e90f087f91532a260 : SUCCESS in 1h 27m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/74dbd473af4345d8b91cd7a1e0f6e476 : SUCCESS in 1h 30m 33s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/97e4b491ed134dd0ac7d21ad197b89e1 : FAILURE in 1h 36m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/65f81716e5db4be0a775805d1846f2b3 : SUCCESS in 1h 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e05446aa050f4e3b0a8a2839f76a15f37bbac1c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 11:23:01.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6203a3cc75f9a59dbdcd806ef986d3735fed5a3a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 13:08:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"b6073b3b895b98f6e51f41388ca95942e3baeee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 13:10:33.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8cc44cdc0b4d077ceeaa3521a702bcf30326f5cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-13 13:26:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7a4cc7cf35fa6af3531fab45897d0a731c0aa3de","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-13 14:32:15.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b1935d5dcee178cda9870a71f7043b0e342e854","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-13 14:47:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b0b3b728dff03a9fb50a0062fabf34f1266d538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 14:59:56.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ea8fb8171a643c4a11bf83b0395e70e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dce8a2adeaa848e18a21b4f472715e6c : SUCCESS in 26m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/99edc40e980f4e11a33e36e8c0ebd791 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/234e4cd2f3ea44ada848a88695199d9c : SUCCESS in 22m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/16a5a2832b544742b1a16627bd251b5b : SUCCESS in 31m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/61510f2c1c17489884dc2b97494a4ca8 : SUCCESS in 1h 10m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25af01cebf3740c38c2f329bfaff59e9 : SUCCESS in 1h 41m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a07ff990204d928982d8807497e8fb : FAILURE in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/804db026195946c9b4ffc26f9393b310 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b822aab08b64a038ce9ba0420f207ca : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f66148df5f4728a2ffa904bac39888 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b0e2f38d3b420f9de5e82ed3f78887 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d103df427da94d0d9bc3958a4ad65176 : SUCCESS in 28m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/42904a4e348744bca2ece000e57aecdd : SUCCESS in 40m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/62c106da43924c88a97a0378f87f4098 : SUCCESS in 1h 09m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e74a97a521a140e4a7b41df25e8e6204 : SUCCESS in 1h 35m 22s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4f955db6a115461ca1e26985137eeedf : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5524cb27222e486db7e9e5b14dc3db1f : SUCCESS in 1h 23m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a4c24875c5b4d148d52b1f8d1b27c7d : SUCCESS in 26m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d1cf26980d6e43eea862d1c31363bd3b : SUCCESS in 35m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1218782110a749d98c2fdcd0c483caae : SUCCESS in 1h 04m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee82881cb3a4428daa53298e0bf197d2 : SUCCESS in 1h 36m 17s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1b06a5fd04c541a99dc5dc91a9cda777 : FAILURE in 1h 42m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/623a9f23a58f4e6ba92df0104bbfb9f3 : SUCCESS in 1h 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6d91deddec2907b35a191d170c7e8f2d06681c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 15:20:39.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cdb162f150bf013385c532a242aded0b87ec8164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 15:21:16.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"04a499cba6e1a5c2163fc50a8f352e57d20fb7c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-13 15:37:57.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"08fc69a61ef2d9def744fa914fe5c2ce17e764bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-13 15:47:59.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1806ece63432364d8fcc827138da057d18d7197a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 17:01:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c434fac89e4eadb754fc131ff5a12d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cb8d119b7e7243e7bc45822f081f8f73 : SUCCESS in 27m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/306bebbe63de499fa3a30ccc42d32c48 : SUCCESS in 1h 03m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/54dfaa1adc0a493a9e1f58d9ec81d3c2 : SUCCESS in 20m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b5abdbbc0b674b9e9ad0bc15a1b1e4e9 : SUCCESS in 26m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/afd9aeea890844a69bfa9ad7f11e60a8 : SUCCESS in 1h 08m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9ef9183b0a8b4d629b7215631ad473a8 : SUCCESS in 1h 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acef7b488b3d4368b23965061ee38d8b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb1a67e490e44b0fbff4d9e05c06af5b : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88055412d31f4e3eb036f9554ba46318 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71ca3edc4b04454ca70b93c9ec4304ac : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9682d12eeef8461f8dfbc74c69f9a9e4 : SUCCESS in 5m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1b0b1f4630bb427ca9cf1c6f185b7b36 : SUCCESS in 26m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/02fb17ef59b74b39859fbd1fa36e6511 : SUCCESS in 37m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cf1d07e75464464d89ffc58bf14ee97f : SUCCESS in 1h 01m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b30c54bb6bd94cdf9e5f064c5c6d3548 : SUCCESS in 1h 34m 05s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d901b210c2fb403cbebadee651acd096 : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9c50a6594a544d26aa121951ee0d33e6 : SUCCESS in 1h 16m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5838aa7a80b64da88671dbc0bbf73cfd : SUCCESS in 26m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c6ef04d4b73441788fee78e55cedd3d4 : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2d8beb7138924ef190d50fa4859dc7c0 : SUCCESS in 1h 04m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6fac9f4bfeba4b65adb39508241c7a90 : SUCCESS in 1h 26m 46s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/89ac3973b7224b6fad19a7f42b43339e : FAILURE in 51m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/27bb1f9ee0934784ac5e69ea9aa8c002 : SUCCESS in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c8d6e09691f7d297a306ebf8aef23ede1de1090a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 17:03:24.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aa01838b4690f2eeab65894678e1ea8d5362c09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 17:03:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1a9522da76107c72ed5b9ff4384141d77cbfcff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:40:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26cc0453c1874fd095b876713382d49a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a70458ffa1b84b6aba38e5640d5355ab : SUCCESS in 13m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8a6d42ef829a43d183b02e517731db20 : SUCCESS in 36m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3905231e842d411abc392f2ad28c981e : SUCCESS in 1h 09m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/02c6e15320ff4bfd9e856bee94380ff7 : SUCCESS in 1h 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a865c9595be4f8eb49fa6a8c9da74d1 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd0226d562a4152b8de722da99c291c : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e87ebf20975e4bfebec9d0a4981e106d : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967aaa10cbee4283a88e7980a158769d : SUCCESS in 5m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5669c4f310904af298c8fc25757a5a1f : SUCCESS in 34m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9c0321b976974332956a588ea58a3d54 : SUCCESS in 35m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4c967b595912423ca767975229b2b2ea : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5a516217e9047808fb0895c302e4c00 : SUCCESS in 1h 25m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3ab504565154e89be827d1d813e964b : SUCCESS in 28m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6cbd41c48804f20b9b89e64502ffdb5 : SUCCESS in 25m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1fa63f6347b547aab6082359cf2b1abf : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a29efe41ec5840629107d7215435b402 : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":6},{"id":"3eaabb82e4c5681d81f7363cb5feb94b6743d15e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:40:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8e8f8f9f22b675a20f36c8c2a12b2217ad0e4cbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:41:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3000277bdb744f4b971bb9d672bae1e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa64d01c72c54aef9189537a6769207a : TIMED_OUT in 3h 01m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ef64d5efde0244b9a9e3430b265c51a4 : FAILURE in 1h 30m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bee81c766d3c450298aaaa6a32b2782f : TIMED_OUT in 3h 03m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03ef1d1899454e6994882b4d90560d9a : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cde91cdc697e4cafb08ae4a7c00dd704 : SUCCESS in 5m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f0c46510e13744819bafc0fe950ae883 : FAILURE in 2h 37m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fd5ac6ed741b4d7ca5987fbf21185b11 : TIMED_OUT in 3h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e0f8a705d438343078db036009d4da4827534b17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:42:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0100822a278d4ba8aa8b0542ce7c8c34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b61ce7a3c86443f85eb7f88fbd39b46 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a9c94abf530f3ed36b9b3f10c89549f04af26df4","revisions":{"a576bd33d966d01450613f5dd30454e537480f8b":{"kind":"REWORK","_number":1,"created":"2025-01-13 08:54:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/1"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:31.000000000","tz":60},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a576bd33d966d01450613f5dd30454e537480f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a576bd33d966d01450613f5dd30454e537480f8b"}]},"branch":"refs/heads/master"},"83da63d2d50cef9961d015a320aab6e70a98dd0c":{"kind":"REWORK","_number":2,"created":"2025-01-13 09:12:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/2"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 09:12:19.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83da63d2d50cef9961d015a320aab6e70a98dd0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83da63d2d50cef9961d015a320aab6e70a98dd0c"}]},"branch":"refs/heads/master"},"b41078157904a29f98664da4897c4ee35924c5fc":{"kind":"REWORK","_number":3,"created":"2025-01-13 11:23:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/3"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 11:23:00.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b41078157904a29f98664da4897c4ee35924c5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b41078157904a29f98664da4897c4ee35924c5fc"}]},"branch":"refs/heads/master"},"912fa174bbc5f9003e3bce401b78eaa53c73632a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-01-13 13:10:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/4"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 13:10:32.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nCloses-Bug: #1795609\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912fa174bbc5f9003e3bce401b78eaa53c73632a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912fa174bbc5f9003e3bce401b78eaa53c73632a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"87feacf16fbf54dffc0884ae4d76f5076e0fea28":{"kind":"REWORK","_number":5,"created":"2025-01-13 15:20:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/5"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 15:20:38.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nCloses-Bug: #1795609\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87feacf16fbf54dffc0884ae4d76f5076e0fea28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87feacf16fbf54dffc0884ae4d76f5076e0fea28"}]},"branch":"refs/heads/master"},"a9c94abf530f3ed36b9b3f10c89549f04af26df4":{"kind":"REWORK","_number":6,"created":"2025-01-13 15:21:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/939055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/939055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/939055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/939055/6"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 15:21:16.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nCloses-Bug: #1795609\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9c94abf530f3ed36b9b3f10c89549f04af26df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9c94abf530f3ed36b9b3f10c89549f04af26df4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
