)]}'
{"id":"openstack%2Fkolla~939088","triplet_id":"openstack%2Fkolla~master~I3f45178c07a9cd054e572ff80d0e19fb014d69d3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-13 18:26:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f45178c07a9cd054e572ff80d0e19fb014d69d3","subject":"ironic: Fix dnsmasq.log permissions","status":"MERGED","created":"2025-01-13 13:24:32.000000000","updated":"2025-01-13 18:27:11.000000000","submitted":"2025-01-13 18:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939088","meta_rev_id":"7ef9cd6e41774adc4658ff6663616a693b2dbe89","_number":939088,"virtual_id_number":939088,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 18:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-01-13 16:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-01-13 13:33:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-01-13 15:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-13 16:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-13 16:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 13:33:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-13 15:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 15:48:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-01-13 16:45:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"87ec35d9227a1ec156b29c441a79523ad5f07785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-13 13:24:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9567a4b6ecfedd8c21189f216a38763ee74127c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-13 13:33:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2177e00bb24d746d75d1854c90f08aa877f61ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:21:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f51302fc23ef4287a3f9fc7703be3b15\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/91260eba6d274226aee15134162ae80d : SUCCESS in 23m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e2c1b9637ec64644b48bcb9d25b79278 : SUCCESS in 38m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4bf231c851d54d198cdcd37d3079965b : SUCCESS in 24m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/01297165e40d43c38dfe83df13504c4e : SUCCESS in 31m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4545998f07bc420788e772a81578ab4c : SUCCESS in 1h 01m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/81f79b39868d4047b2df510041647429 : SUCCESS in 1h 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8378463b9e9b4e7b84c47e9080dfab60 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ff8ea12ef44a2faf6c77296246697f : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5187a9ae5d6431db07dbb814c3ee10d : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b40a20126f3e4fb194381c0ed910b8a1 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37cc170b35f45aa88d09d3ebb77c6a2 : SUCCESS in 4m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/216a7f85f3c141ee99dbd6c82be11d25 : SUCCESS in 20m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/65799b9736dd4620922c45acb5ea49e3 : SUCCESS in 40m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e1b92e6c344449c78f5e503ce8ae4065 : SUCCESS in 1h 04m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b767e18dd3c04aa397768dee1dcf3ed5 : SUCCESS in 1h 35m 19s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1af5e774baff4739858ad994e830e5cc : SUCCESS in 1h 15m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e1eb95cc56747e2a3e7079352ec69e7 : SUCCESS in 22m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/91d3adf416a44a16b39acdbba5201cac : SUCCESS in 30m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cfd8aac8532d40fb8b1f0d88b19d7c80 : SUCCESS in 37m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9493a9d9fd6a4832a24f560e0b2d6531 : SUCCESS in 1h 39m 08s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9dc0bc7e745a48f29b5aae63383bf9f2 : SUCCESS in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc555b5ef6a2f9fab8908d31e1f409f270f35f83","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-13 15:48:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a7067e2bdb3145d7234ebeb412f6960be6808b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 16:31:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8457e29056514c569a57f93842b93760\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5eed610dd66f49b088f593b2245c58d0 : TIMED_OUT in 3h 01m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f389da9d271f4e1b93f573f38ed14066 : FAILURE in 2h 40m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f98502f815c425c9226203fdfb6d93b : SUCCESS in 1h 55m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccab7a7b5173470ea4cc37111e377bad : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12529315210b453c96fd8a1b377ae284 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb70c8c784f8458b94a476b77137da80 : FAILURE in 2h 55m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/62215f05569b4a11907b2c59a050ec6e : TIMED_OUT in 3h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1326fe531b92d7d107da08c5f044b96df1e25dcd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-13 16:45:46.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1e003bb52895b6ce8015d65ad3e7a22195ad46c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 16:46:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68b9cb42c0afdfde1887e2039cfffd0ab4d17c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:26:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea75b9dce1f4e70b2284660786bc15a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4b21b3effe04ae387ac6f79e9e7ff0a : SUCCESS in 24m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cdd5d9dce443487892fa533a74755c37 : SUCCESS in 35m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e44ac6b5bf8d46da9e183fc9d121172a : SUCCESS in 1h 06m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e5b31fca86ce4d909ba3fab619622ae5 : SUCCESS in 1h 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c091c411ca424114ab30d630738d80a7 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/662c124ce0df49ee8bfa9d299199b114 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ad4659210ae40a989120e5e5aba5bf9 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6879ffd0f34449d89e30bf98189796c9 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c862e1ede9034683879c06909b71dfb2 : SUCCESS in 30m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d32e664b451a4f7f8857923afc370284 : SUCCESS in 32m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/465a2a5bc198465eb17dc1b9490ad9cf : SUCCESS in 1h 26m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f5c077ddba114a1d870fcb3d7d221019 : SUCCESS in 1h 27m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86a6bbdf31094a29965094ea09e3e74f : SUCCESS in 25m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3e738001eaad4f76b7280467b038feb8 : SUCCESS in 31m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c51b632236244fd388b8973c6f5f8714 : SUCCESS in 1h 12m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7a2ce04b0908401eb30921f87e3cc1e2 : SUCCESS in 1h 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5de4d55eb82386f5bb1c5d90218ad4d6da141bf8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:26:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ef9cd6e41774adc4658ff6663616a693b2dbe89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:27:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c33334898994404b907d40e56e573b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76f1951e3ed4cfb916cc209a28dfd9c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac2bbcc9f19f039cc2414fe090029f37cf781176","revisions":{"ac2bbcc9f19f039cc2414fe090029f37cf781176":{"kind":"REWORK","_number":1,"created":"2025-01-13 13:24:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/939088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/939088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/939088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/939088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/939088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/939088/1"}}},"commit":{"parents":[{"commit":"744ba3b98e685d6bec1cd590adc5e8727d8a9663","subject":"Merge \"[letsencrypt] Rename script for updating certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/744ba3b98e685d6bec1cd590adc5e8727d8a9663"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 13:21:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 13:24:30.000000000","tz":60},"subject":"ironic: Fix dnsmasq.log permissions","message":"ironic: Fix dnsmasq.log permissions\n\nCloses-Bug: #2093881\nChange-Id: I3f45178c07a9cd054e572ff80d0e19fb014d69d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac2bbcc9f19f039cc2414fe090029f37cf781176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac2bbcc9f19f039cc2414fe090029f37cf781176"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
